)]}'
{"id":"openstack%2Ftaskflow~1005378","triplet_id":"openstack%2Ftaskflow~master~I3c164e3e667039ddb440cc229e942e83756af7f4","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-14 17:57:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c164e3e667039ddb440cc229e942e83756af7f4","subject":"Use oslo.utils to parse redis version","status":"MERGED","created":"2026-09-12 09:15:56.000000000","updated":"2026-09-14 17:57:10.000000000","submitted":"2026-09-14 17:57:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005378","meta_rev_id":"08abfc17e7e6b9d392943aea925b042cb1c42dda","_number":1005378,"virtual_id_number":1005378,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 17:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-14 15:36:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-14 15:36:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-12 10:13:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 15:36:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"63e432c7ce1f03b86f37e04fc498ab4d81dda7f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-12 09:15:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7358f7e8daacc4308d0045fa830b034458cb347b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 10:13:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2285d401d974698995304c5a0ba7ff4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11d1cb94840b43f69df7dd1c62879e05 : SUCCESS in 56m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9b15cfa3bf043a0a230e31e284741d5 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f8b7a026e944231af6ca959888d1a0f : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34629c2dbc314494bd8e669943d71455 : SUCCESS in 8m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1134b09797d477984f2991923f51eec : SUCCESS in 10m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e0a02026333b406080d66eb5883e5e27 : SUCCESS in 11m 59s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/bcfa4671ad9b4d2eb440b8dec2bd15c4 : SUCCESS in 13m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3f705d34f6495dac82b93d0cdfa017 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61be1cb9786c4406ae3ecb6a9953a31b : SUCCESS in 4m 42s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/6230c8f1a3db40a789f9403079946384 : SUCCESS in 7m 23s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/033b1efd27844c2196d4fe60476228db : SUCCESS in 5m 47s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/94c1c7751d3f4f65985fe5b85457e9f7 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d68563467b6291fb1489eb9488ae9c8750993288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-12 17:28:26.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fcc5287420cda59e56d43dee6123b7bd77180ad5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-12 17:31:33.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"63c6e21be3aaa9674f7dbfc1ed17ef032dc3ed6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-12 17:32:27.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"ccefb53dbb24524bfeb550c4b35a5d3142735375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 18:21:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40f0745d8b9744eaa46ab16f92bdd8ee\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16baaea16ae94382a483d256f226df6c : SUCCESS in 48m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc8534404e5843e5a2407e0c359160cc : FAILURE in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee1f2a73df834f0fbfdb6e567263a161 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cb1f650d01e4585b923f83a7912b18e : FAILURE in 9m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a11b3094cb946a39ee08b8c7d248fad : FAILURE in 9m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/93ac719494bc4ebdac3b90fa12d54092 : FAILURE in 13m 29s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/8768e24bdb0b41fc92b6af492d72f97b : FAILURE in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a322d65f2a23443792b129779a1978b3 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a679c63c1e884c11ba06dd15b07bf965 : SUCCESS in 4m 29s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/9243b52d34b7464fbdaa0ec419cdb013 : FAILURE in 10m 21s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/97d3192e0eab49e99705e13e34c6c94e : FAILURE in 6m 20s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/bdc72eed9957492cb93c63461df0881d : FAILURE in 8m 24s","accounts_in_message":[],"_revision_number":4},{"id":"877e718bcf7adebdd3dd07e2963605deda9cdcd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-13 18:21:36.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"d89d120a6e56bb9ea2d466edb8788c37dcde577f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-13 18:24:49.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"cb5f58b08eac9137162d30658f6428a4aae934bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 19:31:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6bf14158bdf476f87648053c01315c6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/079721575ad4437ab5d879587e4b5991 : SUCCESS in 1h 04m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/055c36a69ef9405e82e0bb8d27b32fcb : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77eabfcacc76489e8ac76635026f81b9 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee80678144f4431b8c6b0582377da0a3 : SUCCESS in 10m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6725be7baf0a4dedb026910cdbc6c9b3 : SUCCESS in 10m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3929b67a517e4c9d9b8ad9532d4d8b8e : SUCCESS in 12m 18s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/57d99b2a2cff4dfa93afd3c0e0a789be : SUCCESS in 13m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9c324183d9f4ee98201285cda57628d : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d49ddd8674ef4f5bb42fe2f26209d1a6 : SUCCESS in 4m 18s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/9316efd0d3704a99a4ab7ffbd8c5f4df : FAILURE in 17m 31s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/17a85e9ad1cd46cd99714a2c8726f135 : SUCCESS in 7m 09s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/d81c72cb4b10411f9fdae361ef2b0803 : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":6},{"id":"957a7ae1a364e9d38bf1dcaebdd6df05f27b67ae","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 10:24:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5ee5acd9ce617d6ebbdff44ea501dbe024b6c790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:39:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33e98eaa29674dff9ef09f9327fad672\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3836c1c80c544602835dc8aadf54943f : SUCCESS in 1h 13m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e5d28158bf84e09b1cef89acdd460ec : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/126b35e958654720af6db812f01710c6 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68579687da5943bc802c589142bfa54c : SUCCESS in 9m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a27acd5de1d42f6ad34800019ed3e01 : SUCCESS in 9m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fd2cff93dd744a1ebdb6c843aec6375f : SUCCESS in 12m 03s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/e1e52a18f03342328abb0073e1aa7f63 : SUCCESS in 13m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d95b937bc7d74bfea450f2f6a65cba49 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5afbae8f2734217a9c7213832b276d4 : SUCCESS in 4m 37s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/d76d361add604c2da33f24071fd8c851 : SUCCESS in 6m 43s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/85b620c6b46740a0b0317ebed3566339 : SUCCESS in 8m 45s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/796a8c81ed06487fad639e8dd77d11b4 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":6},{"id":"6a683d8cfcf6c7de103eda15c8e6372ad437a7b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 15:05:54.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"b962ebba4192b82c15e765ad07575f38b1c4cb52","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 15:36:13.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"7826b7d227d1f89347dc2c2469faf004559df197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 16:39:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f03cd8a7b64c4582dfc1b6d74fa5e7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd310cbc0fcf44739917268421f41fe3 : SUCCESS in 1h 29m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a864dcf7f89f46e69eda48628a8b1fcf : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71101a5ed9be45c08336b3bde431ed35 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4fc82a6e4f0b4f828560d6da50cafe0f : SUCCESS in 11m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/772882cdd5a44575af5f12a833e466e3 : SUCCESS in 10m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26ea304fdd844158b942da66c27d1259 : SUCCESS in 10m 47s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c30357d377b84f75b4d33746d5a2f83f : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e6137b6b41468e9f829eefc65c6f46 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c16174474564047a042174d6d27f83d : SUCCESS in 5m 02s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/df2d666d3f1648fdbe7a98202778df0d : SUCCESS in 11m 28s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/732e0834567e454da33b7ff0ce33389f : SUCCESS in 6m 43s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/e359a350169b4ba9a7692e1f57f25dd9 : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":7},{"id":"1c809d1fa96acfcd7560f13d4c77e84478319cd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 16:40:33.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"72dee73e3154cc79845286db73fd7d634aae6a9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:57:10.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05d135a976804eb78c84491a692713d5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/289bda4557c448f9a9a1b170cb4829f8 : SUCCESS in 42m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67e8f13ef7284de48ef498b466eebc77 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f573a8438c794fed81368011778ce230 : SUCCESS in 8m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/328fa6ee937b4928ab622c5634fe7b72 : SUCCESS in 9m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b46bdc6fa4164bfc9b6cd7d20f832f28 : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a53d0bb0f548caa085fd4deb790626 : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60f9d3ecdd9e46919d27b6b6e4c542ae : SUCCESS in 4m 39s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/a646167208e84a2d832261800cf072fb : SUCCESS in 5m 41s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/47ea4882e913497c8610a5adbe484b17 : SUCCESS in 6m 50s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/e56c8b85f01545839a77faea87d71cfd : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":7},{"id":"08abfc17e7e6b9d392943aea925b042cb1c42dda","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:57:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8322d2c290e52fd7703ad252bf28da2007735450","revisions":{"6b3c85eef86a2809b349fbc03da79e6bda0a245e":{"kind":"REWORK","_number":1,"created":"2026-09-12 09:15:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/1005378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/78/1005378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/1"}}},"commit":{"parents":[{"commit":"c34aeff03d9ad208856ab388a49bb6969538e1e0","subject":"Merge \"Remove remaining ETCD_VERSION\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/c34aeff03d9ad208856ab388a49bb6969538e1e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 09:15:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 09:15:35.000000000","tz":540},"subject":"Use oslo.utils to parse redis version","message":"Use oslo.utils to parse redis version\n\n... to reduce duplicated similar logics.\n\nChange-Id: I3c164e3e667039ddb440cc229e942e83756af7f4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/6b3c85eef86a2809b349fbc03da79e6bda0a245e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/6b3c85eef86a2809b349fbc03da79e6bda0a245e"}]},"branch":"refs/heads/master"},"5b8de7ac4435352f3cc524de22cbb0ef8d6d2af5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-12 17:28:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/1005378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/78/1005378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/2"}}},"commit":{"parents":[{"commit":"4d726599d33cc3de67b53caa40a9171b85108154","subject":"Add unit tests for redis_utils module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/4d726599d33cc3de67b53caa40a9171b85108154"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 09:15:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:28:13.000000000","tz":540},"subject":"Use oslo.utils to parse redis version","message":"Use oslo.utils to parse redis version\n\n... to reduce duplicated similar logics.\n\nChange-Id: I3c164e3e667039ddb440cc229e942e83756af7f4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/5b8de7ac4435352f3cc524de22cbb0ef8d6d2af5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/5b8de7ac4435352f3cc524de22cbb0ef8d6d2af5"}]},"branch":"refs/heads/master"},"6a3c17d30740398c971ae78a1f0cd6a385d09d89":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-12 17:31:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/1005378/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/78/1005378/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/3"}}},"commit":{"parents":[{"commit":"dba8e51cbcce888dfc94a21e2d2805d8c678f518","subject":"Add unit tests for redis_utils module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/dba8e51cbcce888dfc94a21e2d2805d8c678f518"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 09:15:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:31:25.000000000","tz":540},"subject":"Use oslo.utils to parse redis version","message":"Use oslo.utils to parse redis version\n\n... to reduce duplicated similar logics.\n\nChange-Id: I3c164e3e667039ddb440cc229e942e83756af7f4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/6a3c17d30740398c971ae78a1f0cd6a385d09d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/6a3c17d30740398c971ae78a1f0cd6a385d09d89"}]},"branch":"refs/heads/master"},"851b223e40b8ce30de86374b5ce26cbd6b5c1d28":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-12 17:32:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/1005378/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/78/1005378/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/4"}}},"commit":{"parents":[{"commit":"5284af50ede1375d08e0f07fed53c1d456e094d1","subject":"Add unit tests for redis_utils module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/5284af50ede1375d08e0f07fed53c1d456e094d1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 09:15:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:32:17.000000000","tz":540},"subject":"Use oslo.utils to parse redis version","message":"Use oslo.utils to parse redis version\n\n... to reduce duplicated similar logics.\n\nChange-Id: I3c164e3e667039ddb440cc229e942e83756af7f4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/851b223e40b8ce30de86374b5ce26cbd6b5c1d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/851b223e40b8ce30de86374b5ce26cbd6b5c1d28"}]},"branch":"refs/heads/master"},"fca8d1f64387cc404ef506522e9bfb4599064aaf":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-13 18:21:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/1005378/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/78/1005378/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/5"}}},"commit":{"parents":[{"commit":"bcbaf84fc274105da137127648c28982f9ca6a9f","subject":"Add unit tests for redis_utils module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/bcbaf84fc274105da137127648c28982f9ca6a9f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 09:15:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 18:21:36.000000000","tz":0},"subject":"Use oslo.utils to parse redis version","message":"Use oslo.utils to parse redis version\n\n... to reduce duplicated similar logics.\n\nChange-Id: I3c164e3e667039ddb440cc229e942e83756af7f4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/fca8d1f64387cc404ef506522e9bfb4599064aaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/fca8d1f64387cc404ef506522e9bfb4599064aaf"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"5284af50ede1375d08e0f07fed53c1d456e094d1","ours":"851b223e40b8ce30de86374b5ce26cbd6b5c1d28","theirs":"bcbaf84fc274105da137127648c28982f9ca6a9f","merge_strategy":"recursive","contains_conflicts":false}},"dd5e7fbf624dae1ddd60031e683f133200cd7465":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-09-13 18:24:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/1005378/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/78/1005378/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/6"}}},"commit":{"parents":[{"commit":"b11cbba0b00028665722158158a4f16010665731","subject":"Add unit tests for redis_utils module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/b11cbba0b00028665722158158a4f16010665731"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 09:15:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 18:24:49.000000000","tz":0},"subject":"Use oslo.utils to parse redis version","message":"Use oslo.utils to parse redis version\n\n... to reduce duplicated similar logics.\n\nChange-Id: I3c164e3e667039ddb440cc229e942e83756af7f4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/dd5e7fbf624dae1ddd60031e683f133200cd7465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/dd5e7fbf624dae1ddd60031e683f133200cd7465"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"bcbaf84fc274105da137127648c28982f9ca6a9f","ours":"fca8d1f64387cc404ef506522e9bfb4599064aaf","theirs":"b11cbba0b00028665722158158a4f16010665731","merge_strategy":"recursive","contains_conflicts":false}},"8322d2c290e52fd7703ad252bf28da2007735450":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-09-14 15:05:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/1005378/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/78/1005378/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/78/1005378/7"}}},"commit":{"parents":[{"commit":"2b64257cd615a26159514840b11f78bdbdbd0a1a","subject":"Merge \"Use enum directly to convert raw values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/2b64257cd615a26159514840b11f78bdbdbd0a1a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 09:15:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 15:05:54.000000000","tz":0},"subject":"Use oslo.utils to parse redis version","message":"Use oslo.utils to parse redis version\n\n... to reduce duplicated similar logics.\n\nChange-Id: I3c164e3e667039ddb440cc229e942e83756af7f4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/8322d2c290e52fd7703ad252bf28da2007735450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/8322d2c290e52fd7703ad252bf28da2007735450"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"b11cbba0b00028665722158158a4f16010665731","ours":"dd5e7fbf624dae1ddd60031e683f133200cd7465","theirs":"2b64257cd615a26159514840b11f78bdbdbd0a1a","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
