)]}'
{"id":"openstack%2Ftaskflow~1005385","triplet_id":"openstack%2Ftaskflow~master~Ib669944df34619c3c55ccfccda98bf6136e41de8","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-14 11:18:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib669944df34619c3c55ccfccda98bf6136e41de8","subject":"Create a directory to gather tests for utils modules","status":"MERGED","created":"2026-09-12 17:28:26.000000000","updated":"2026-09-14 11:19:28.000000000","submitted":"2026-09-14 11:18:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005385","meta_rev_id":"e9dc5834adc95d462428d06631f17cf65b93c0b2","_number":1005385,"virtual_id_number":1005385,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 11:18:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-14 10:06:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-14 10:06:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-12 18:45:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 10:06:21.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"96eac1c1c1809639851f4c44811497393580f617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-12 17:28:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81ce1f8dc4f848e6fd52180143bae375701c4e8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-12 17:31:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2307c4f019cebff7ece9935f6a748873ed18099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 18:45:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/109ad740d58a4bb09bc5c24ce485c9de\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6714eaacd2414223bdfe74a6eff0d54d : SUCCESS in 1h 07m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/deabe96be91d4651a87f78d0d6070e00 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/150206ce9d67452fa099f316bb4f238c : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2ea6e0e0c1b4cc993ec15f58dca37c2 : SUCCESS in 11m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b76b548074394e3a8f2bfe98502f8a34 : SUCCESS in 11m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4a417bffdbdd4c0eb3389bf3d9bc2dbe : SUCCESS in 12m 16s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/049e183103c24959b09a3fa443750202 : SUCCESS in 15m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b3292c5ba10409eb711cb42ab95725d : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e66afef13714e42a47f92f3a6127ab5 : SUCCESS in 4m 15s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/943a902292a445c3a9bf08053db6f3c9 : SUCCESS in 7m 27s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/9532d27a57814c7880f5bad7811afa53 : SUCCESS in 6m 13s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/c2a0bebce9ab4efabc60c26aadfa48c0 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"51ea16b29ae0f4abe124f22d191ce0b875dbd431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-13 18:24:39.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"cb017950a30fb42666a0c2f78b2be7162cdc199b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 19:53:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/741d686ec578489a85382eff677ad89f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/698333cb4b7e42b18fc64abfb4c6f894 : SUCCESS in 1h 26m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89d37e009ee842aa85b0dd4f2fccfa91 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ce8f09aa1c4bbf9f1bafad8fc353d7 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aab9d07ddf9a42e29e0c0d04a570656a : SUCCESS in 9m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26a649f7ec2a4bfbbadb061f2d911fed : SUCCESS in 9m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9af8d94f8c5547808903e71ab5ff9e7a : SUCCESS in 12m 37s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/58fb22ddfdd34db6ba6895419d5dc7ab : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91328e8e1c104bac8157c3ef6c0cef8b : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57838a46f68d48258de1599334402bca : SUCCESS in 4m 30s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/baf287deb13746c18aaf2447453cd2c9 : SUCCESS in 7m 35s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/878a3d84dc7440caaa477f64eb8feac0 : SUCCESS in 6m 49s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/f9f6d48171434854aa4f4894e0b132d1 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":3},{"id":"486936987714987c2ddc5f6d7b24963a02ceb81b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 10:06:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7c1a34c500dded9ee75cf39ed327a5e0197ff24a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 10:07:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"30457dc73e5c88ae9ad017e7e9e0c59e9d2ef9ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:18:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdddde9ab3054e668a3cdc936fb0f2e0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/858185501ec3420e8962b1bd21d683cd : SUCCESS in 1h 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be26c7bb036487291fa751f9e598938 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec19ed10766d41898e01c6ebba8e236d : SUCCESS in 9m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2568838abb5b4f1c982b2c0e66f5cad5 : SUCCESS in 9m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fbdb741ca53a41be926d3033d05e9620 : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7720336a08154c5bb918f495e76b895b : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/926071f411ad4d7cb80a795634cd70e8 : SUCCESS in 4m 06s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/71f4fa09c0ca474ba6bde373ec2a4b9f : SUCCESS in 6m 42s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/578f6d703cdb4137a0de77fe84cad31d : SUCCESS in 7m 08s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/ab4cb6735bd946f0a5de3bcd37e73aa7 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":3},{"id":"f966e3e5c77fe15fcce4f66a9fc2630f2e58f9d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:18:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e9dc5834adc95d462428d06631f17cf65b93c0b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:19:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e39eecbc196b4b7488e1f536e93e93dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d675642ac2c4c8a850d910d3218f48c : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"78f7ada7af907cab88f96fa259cb12c81b7ace79","revisions":{"24d5d048707b47f2f2cb9f18fd26083a0af10a0b":{"kind":"REWORK","_number":1,"created":"2026-09-12 17:28:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/1005385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/85/1005385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/85/1005385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/85/1005385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/85/1005385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/85/1005385/1"}}},"commit":{"parents":[{"commit":"c34aeff03d9ad208856ab388a49bb6969538e1e0","subject":"Merge \"Remove remaining ETCD_VERSION\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/c34aeff03d9ad208856ab388a49bb6969538e1e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:06:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:28:13.000000000","tz":540},"subject":"Create a directory to gather tests for utils modules","message":"Create a directory to gather tests for utils modules\n\n... to prepare improving test coverage of these modules.\n\nChange-Id: Ib669944df34619c3c55ccfccda98bf6136e41de8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/24d5d048707b47f2f2cb9f18fd26083a0af10a0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/24d5d048707b47f2f2cb9f18fd26083a0af10a0b"}]},"branch":"refs/heads/master"},"7dbedf0aa707d58ee27cd887b2b977ebda1272c6":{"kind":"REWORK","_number":2,"created":"2026-09-12 17:31:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/1005385/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/85/1005385/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/85/1005385/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/85/1005385/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/85/1005385/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/85/1005385/2"}}},"commit":{"parents":[{"commit":"c34aeff03d9ad208856ab388a49bb6969538e1e0","subject":"Merge \"Remove remaining ETCD_VERSION\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/c34aeff03d9ad208856ab388a49bb6969538e1e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:06:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:30:20.000000000","tz":540},"subject":"Create a directory to gather tests for utils modules","message":"Create a directory to gather tests for utils modules\n\n... to prepare improving test coverage of these modules.\n\nChange-Id: Ib669944df34619c3c55ccfccda98bf6136e41de8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/7dbedf0aa707d58ee27cd887b2b977ebda1272c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/7dbedf0aa707d58ee27cd887b2b977ebda1272c6"}]},"branch":"refs/heads/master"},"78f7ada7af907cab88f96fa259cb12c81b7ace79":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-13 18:24:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/1005385/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/85/1005385/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/85/1005385/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/85/1005385/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/85/1005385/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/85/1005385/3"}}},"commit":{"parents":[{"commit":"ced453a5351660d44d20f31cd30fa929c254ad8d","subject":"Merge \"Remove old note about test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/ced453a5351660d44d20f31cd30fa929c254ad8d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:06:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 18:24:39.000000000","tz":0},"subject":"Create a directory to gather tests for utils modules","message":"Create a directory to gather tests for utils modules\n\n... to prepare improving test coverage of these modules.\n\nChange-Id: Ib669944df34619c3c55ccfccda98bf6136e41de8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/78f7ada7af907cab88f96fa259cb12c81b7ace79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/78f7ada7af907cab88f96fa259cb12c81b7ace79"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"c34aeff03d9ad208856ab388a49bb6969538e1e0","ours":"7dbedf0aa707d58ee27cd887b2b977ebda1272c6","theirs":"ced453a5351660d44d20f31cd30fa929c254ad8d","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
