)]}'
{"id":"openstack%2Ftaskflow~1005386","triplet_id":"openstack%2Ftaskflow~master~I0ca3c5be977c61080857c0978e3fa056054566f5","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-14 11:52:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0ca3c5be977c61080857c0978e3fa056054566f5","subject":"Add unit tests for redis_utils module","status":"MERGED","created":"2026-09-12 17:28:26.000000000","updated":"2026-09-14 11:53:23.000000000","submitted":"2026-09-14 11:52:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005386","meta_rev_id":"14667d4d6cdb2e2a90db1f2e1070b702ab8e0a33","_number":1005386,"virtual_id_number":1005386,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 11:52:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-14 10:06:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-14 10:06:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-12 18:19:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 10:06:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"8d4c3aaeb437906538643d6c69e60ce58dc3ee7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-12 17:28:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f40883b1f7f46b853c951c49e55db00a4b2fa2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-12 17:31:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"680f731e5b3865f2e04ceaaafe1257e4fe174b2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-12 17:32:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8fa855ef7e9b9896c2b9401f5da1e574afbf51e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 18:19:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2c13ab785e04a39af9ee25c14c05be3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c189e550a50422db2cf61aaaec9f8da : SUCCESS in 46m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/951306cb40df48ae9f360fad43388f82 : FAILURE in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eb3d6648b2148e18dc450a556521904 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8521f860fe3240869966166fb7ec5940 : FAILURE in 9m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36d6aba22762436e880e2a43f53bd8bb : FAILURE in 10m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5e13a0a5c5f8437493b4bb45fd895e22 : FAILURE in 12m 17s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0cd7b03c094b499792753289152fa84a : FAILURE in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8366d40b54d4ea6898b5d29e5baf581 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7341098487cd42d0b897f043f5c197f1 : SUCCESS in 5m 16s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/b197b550665c4a9a897f13e365cc4b54 : FAILURE in 7m 10s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/5bac99e9dbaf44f39de58238ab0f5303 : FAILURE in 9m 42s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/346828101caf442eb89e897a74d2a9ba : FAILURE in 6m 05s","accounts_in_message":[],"_revision_number":3},{"id":"1b52badbab739fdcd08a4fb2e74b0d4c12404e3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-13 03:46:34.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ac0812fa394cd586b58571cc44158e02db1c1a3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 05:32:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eea5c4a2e4e14ef293b67e67f27fc229\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9551cb04d69c49629bc67a44aa6199fc : SUCCESS in 1h 38m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/155923c8c468465e875dc0950b38ad1b : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a147d78e0f4981a8dc21818542146a : FAILURE in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27ec13a09d5a4e1aabc72dc1bd8cb942 : SUCCESS in 8m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/beaa471026ec467bb7d80b6c81452b83 : SUCCESS in 9m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e44a83ae708f4395928a87bb0a772f01 : SUCCESS in 12m 33s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/d8a0261dcf8f4c76a5c4bb0eea6191f4 : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0f0c07d16a4bb4aca79abe55194c8d : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c20bd776e9ce4d7ab63dfca2e6b484dc : SUCCESS in 3m 32s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/f524c3a1f1314a88b31689fd2c8b05c6 : SUCCESS in 6m 24s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/cbbec84ee6694d7ea3435c7769f58339 : SUCCESS in 6m 30s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/7454a462aff14b9dab67720398cfdf03 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":4},{"id":"ac11d1e9b66bc5e94f49b13ba007834deaa91712","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-13 18:21:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"d7c52b3bd62ea5fc393a86166cff98660f39ea64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-13 18:24:44.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"4140bcbd4fd951086bf1db1f06de5a11b1514ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 20:02:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b871eb183440d4aafdb489af230ef9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8049d8535d4b42379a929cc497cd40ed : SUCCESS in 1h 32m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d8d27368aaf40ad87766529475955ca : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/019ad4d20a58440ebdabcba8b84bc6a0 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d79de716ea47483185a9631fe911ac2e : SUCCESS in 8m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f19f1881d235487b833534334b29da47 : SUCCESS in 12m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02e220cc7ae54f6cb5d3a2303a8f2031 : SUCCESS in 10m 13s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/788ca18a3bf543e6a1177308b9702a6c : SUCCESS in 16m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d12bb66e71ad4615b249712956ba8a91 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fa57d0c820f408dbeeb701f407ff438 : SUCCESS in 3m 51s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/a00b5be752e048de962c360bdded59a1 : SUCCESS in 6m 32s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/16517814e6b44c23a3f433ec1ecda8d9 : SUCCESS in 6m 23s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/8eee6a7f4573405283006891ecd2fa17 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":6},{"id":"506f4fb785068cf9225e81856c459d687c425e0e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 10:06:41.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2d91d18a26619b8dfa97386c52b3520579231ef1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 10:07:29.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b68b23af276a536db72b887085383677fae7f9a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:52:25.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed9e8bb60974210a55a63b245c902e4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44fb0767603d49e2834690e9c3feac42 : SUCCESS in 1h 39m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fce1f1e940640718684cfacd1ffcb9f : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cae5bcb66f184fed8c9e171f4eb19379 : SUCCESS in 9m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1d8226fbe8f430aaf014866f04e1b22 : SUCCESS in 10m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5c2bfd9acce74ca18092b6b1c72877de : SUCCESS in 12m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47911a157be34ca5bcfce3115e89cb4a : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c76680db896e49efb41d811d54cc311f : SUCCESS in 4m 37s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/b321d65bfb0d4393b7722c7ac8be8999 : SUCCESS in 7m 07s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/fdcd5f4eaf8f430a85d0bd9086f25e0a : SUCCESS in 6m 48s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/dc743efa16344656ae74ae96f1d60237 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":6},{"id":"ab3f07a053fe455ce9b8ea4d78d4233809d6304d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:52:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"14667d4d6cdb2e2a90db1f2e1070b702ab8e0a33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:53:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce4ee1a7cd1b4c208ba4f035f4457da8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac0ba919ae384dd781e9fdb5fd416b4c : SUCCESS in 47s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b11cbba0b00028665722158158a4f16010665731","revisions":{"4d726599d33cc3de67b53caa40a9171b85108154":{"kind":"REWORK","_number":1,"created":"2026-09-12 17:28:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/1005386/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/86/1005386/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/1"}}},"commit":{"parents":[{"commit":"24d5d048707b47f2f2cb9f18fd26083a0af10a0b","subject":"Create a directory to gather tests for utils modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/24d5d048707b47f2f2cb9f18fd26083a0af10a0b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:27:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:28:13.000000000","tz":540},"subject":"Add unit tests for redis_utils module","message":"Add unit tests for redis_utils module\n\nChange-Id: I0ca3c5be977c61080857c0978e3fa056054566f5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/4d726599d33cc3de67b53caa40a9171b85108154"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/4d726599d33cc3de67b53caa40a9171b85108154"}]},"branch":"refs/heads/master"},"dba8e51cbcce888dfc94a21e2d2805d8c678f518":{"kind":"REWORK","_number":2,"created":"2026-09-12 17:31:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/1005386/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/86/1005386/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/2"}}},"commit":{"parents":[{"commit":"7dbedf0aa707d58ee27cd887b2b977ebda1272c6","subject":"Create a directory to gather tests for utils modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/7dbedf0aa707d58ee27cd887b2b977ebda1272c6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:27:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:31:22.000000000","tz":540},"subject":"Add unit tests for redis_utils module","message":"Add unit tests for redis_utils module\n\nChange-Id: I0ca3c5be977c61080857c0978e3fa056054566f5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/dba8e51cbcce888dfc94a21e2d2805d8c678f518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/dba8e51cbcce888dfc94a21e2d2805d8c678f518"}]},"branch":"refs/heads/master"},"5284af50ede1375d08e0f07fed53c1d456e094d1":{"kind":"REWORK","_number":3,"created":"2026-09-12 17:32:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/1005386/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/86/1005386/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/3"}}},"commit":{"parents":[{"commit":"7dbedf0aa707d58ee27cd887b2b977ebda1272c6","subject":"Create a directory to gather tests for utils modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/7dbedf0aa707d58ee27cd887b2b977ebda1272c6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:27:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:32:07.000000000","tz":540},"subject":"Add unit tests for redis_utils module","message":"Add unit tests for redis_utils module\n\nChange-Id: I0ca3c5be977c61080857c0978e3fa056054566f5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/5284af50ede1375d08e0f07fed53c1d456e094d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/5284af50ede1375d08e0f07fed53c1d456e094d1"}]},"branch":"refs/heads/master"},"58dc7f0bef84c83c157fa090434f4db802fa2735":{"kind":"REWORK","_number":4,"created":"2026-09-13 03:46:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/1005386/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/86/1005386/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/4"}}},"commit":{"parents":[{"commit":"7dbedf0aa707d58ee27cd887b2b977ebda1272c6","subject":"Create a directory to gather tests for utils modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/7dbedf0aa707d58ee27cd887b2b977ebda1272c6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:27:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 03:46:16.000000000","tz":540},"subject":"Add unit tests for redis_utils module","message":"Add unit tests for redis_utils module\n\nChange-Id: I0ca3c5be977c61080857c0978e3fa056054566f5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/58dc7f0bef84c83c157fa090434f4db802fa2735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/58dc7f0bef84c83c157fa090434f4db802fa2735"}]},"branch":"refs/heads/master"},"bcbaf84fc274105da137127648c28982f9ca6a9f":{"kind":"REWORK","_number":5,"created":"2026-09-13 18:21:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/1005386/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/86/1005386/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/5"}}},"commit":{"parents":[{"commit":"7dbedf0aa707d58ee27cd887b2b977ebda1272c6","subject":"Create a directory to gather tests for utils modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/7dbedf0aa707d58ee27cd887b2b977ebda1272c6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:27:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 18:21:21.000000000","tz":540},"subject":"Add unit tests for redis_utils module","message":"Add unit tests for redis_utils module\n\nChange-Id: I0ca3c5be977c61080857c0978e3fa056054566f5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/bcbaf84fc274105da137127648c28982f9ca6a9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/bcbaf84fc274105da137127648c28982f9ca6a9f"}]},"branch":"refs/heads/master"},"b11cbba0b00028665722158158a4f16010665731":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-09-13 18:24:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/1005386/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/86/1005386/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/86/1005386/6"}}},"commit":{"parents":[{"commit":"78f7ada7af907cab88f96fa259cb12c81b7ace79","subject":"Create a directory to gather tests for utils modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/78f7ada7af907cab88f96fa259cb12c81b7ace79"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-12 17:27:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 18:24:44.000000000","tz":0},"subject":"Add unit tests for redis_utils module","message":"Add unit tests for redis_utils module\n\nChange-Id: I0ca3c5be977c61080857c0978e3fa056054566f5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/b11cbba0b00028665722158158a4f16010665731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/b11cbba0b00028665722158158a4f16010665731"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"7dbedf0aa707d58ee27cd887b2b977ebda1272c6","ours":"bcbaf84fc274105da137127648c28982f9ca6a9f","theirs":"78f7ada7af907cab88f96fa259cb12c81b7ace79","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
