)]}'
{"id":"openstack%2Ftaskflow~1005446","triplet_id":"openstack%2Ftaskflow~master~I642c5da62533cac63245db9eb66aac9ad96bd933","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-14 10:24:36.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-14 13:50:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I642c5da62533cac63245db9eb66aac9ad96bd933","subject":"Use enum directly to convert raw values","status":"MERGED","created":"2026-09-13 18:28:21.000000000","updated":"2026-09-14 13:52:00.000000000","submitted":"2026-09-14 13:50:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005446","meta_rev_id":"e6298e0e0705f5d3607b03b148e3f5cf0ad3c65b","_number":1005446,"virtual_id_number":1005446,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 13:50:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-14 10:24:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-14 10:24:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-13 19:13:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 10:13:35.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"f691fae93e6fde5215c67780ff49e8d18db79395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-13 18:28:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0ee41d6548402b30cedba8c71438d27434d73e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 19:13:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55422a22489249029dedb46cf55471cf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b5bb04b265340e5ae26a116226fb9bb : SUCCESS in 43m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e026e862696441dabb7c45925dee1090 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30bfad6af30349c49d0d6b0b29cf8b21 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a62ec9aeaa441f094a5ec6d657d74f5 : SUCCESS in 8m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c9882bee92c64352b82084a7965733e9 : SUCCESS in 10m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ffb6f3030f1f4e81baabab81fafe7720 : SUCCESS in 11m 09s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3c0d0a05e4fe472888d381a9af720ad4 : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f82da0f2621645618929a5fc483f1a30 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5b46b0261bd44d8a1796bd7a3018aa3 : SUCCESS in 3m 23s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/aae99c7ed68c4058ba2f4835c26236c5 : SUCCESS in 7m 19s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/dd1c3e56c47f42be8f910f2e81a0ebce : SUCCESS in 5m 07s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/6ba607ee291b49619c00cc31b23232b3 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"baaa6812466bc383e58cef0e43e8ef5973fd501c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 10:13:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51c2c0446213316929f915b7e5f21fa18354f146","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 10:23:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5351c661d476fd9d6070ca4157e362f7cb1ac5f5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 10:24:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df308b08f52ccde419e0e82eac8bd0101d3c4c70","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 10:24:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a6d710f6400ab41f71876a0f1e9a008089009de5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 12:13:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a5cabb8876a452d884ab82d7e85334d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13e4e04c1af342a2a8b254032c626693 : SUCCESS in 1h 34m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fb4eab1d9934be8964cc85e030cb953 : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d313f6d11ca74e699e2793135180bb85 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ad090beeba94d9798c3c1af2c3cf46d : SUCCESS in 9m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb2401d2b2c54397bbbdc83b62f9ba8b : SUCCESS in 10m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/22de990b466a4783b5a26c55cdbefbbc : SUCCESS in 14m 38s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c278e0376ef640c0aa33367cf27fc2a2 : SUCCESS in 15m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b351b017a42e431cbcac9652153f9ea4 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe92543b97064d909ee4094a849b4db6 : SUCCESS in 4m 48s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/8de0f918ea684ce5af51878ddcf90db3 : SUCCESS in 6m 59s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/988f331219584ce1be12069e84653ea7 : SUCCESS in 6m 29s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/05c0a0c792864ac1b1a495e7f200d3f7 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":2},{"id":"22833b3ea6769e1c05979e17d7cb67e5776dfb86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 12:14:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a1ef5e1c9fffc397982233200efcac018721cc76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 13:50:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4680e2328e449d1bb7f55cdc3a4c0cc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a4a41a520ed4896a7181444ac777692 : SUCCESS in 1h 31m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc805034aa7a4eaca3ffe091f064b2ab : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e97bae05b8994f5e8e544956daa1d017 : SUCCESS in 9m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1b08d17199740a58568facd592453e9 : SUCCESS in 9m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f9bacf3988b3421a8889160b5ba7f581 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64939f4b64ab4449a38804bd2d56b80c : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ca27acc37d44317aa2e16b9d58a5383 : SUCCESS in 4m 32s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/743a78b554ea4a178fe9a73f18bf98b2 : SUCCESS in 7m 06s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/4254c8e2f84d4ef3a17cd8b82f4d25bb : SUCCESS in 6m 25s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/9de93f874a7a4757ae4abacc901d46db : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":2},{"id":"ad2bd4ced72403614dc239794f498b76c50b90da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 13:50:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e6298e0e0705f5d3607b03b148e3f5cf0ad3c65b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 13:52:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70f63e4273444da4bc369f0f4cb9eab6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42c2eca55d04aa691bd09d97110dd12 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83d1edf593ca9e0fb51acb1d7c327ca9351a3d14","revisions":{"59fdade9b27d5d031ab3e6b01b78ebe309d06ecf":{"kind":"REWORK","_number":1,"created":"2026-09-13 18:28:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/1005446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/46/1005446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/46/1005446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/46/1005446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/46/1005446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/46/1005446/1"}}},"commit":{"parents":[{"commit":"b11cbba0b00028665722158158a4f16010665731","subject":"Add unit tests for redis_utils module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/b11cbba0b00028665722158158a4f16010665731"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 18:28:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 18:28:01.000000000","tz":540},"subject":"Use enum directly to convert raw values","message":"Use enum directly to convert raw values\n\nChange-Id: I642c5da62533cac63245db9eb66aac9ad96bd933\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/59fdade9b27d5d031ab3e6b01b78ebe309d06ecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/59fdade9b27d5d031ab3e6b01b78ebe309d06ecf"}]},"branch":"refs/heads/master"},"83d1edf593ca9e0fb51acb1d7c327ca9351a3d14":{"kind":"REWORK","_number":2,"created":"2026-09-14 10:23:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/1005446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/46/1005446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/46/1005446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/46/1005446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/46/1005446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/46/1005446/2"}}},"commit":{"parents":[{"commit":"b11cbba0b00028665722158158a4f16010665731","subject":"Add unit tests for redis_utils module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/b11cbba0b00028665722158158a4f16010665731"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 18:28:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 10:23:30.000000000","tz":540},"subject":"Use enum directly to convert raw values","message":"Use enum directly to convert raw values\n\nChange-Id: I642c5da62533cac63245db9eb66aac9ad96bd933\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/83d1edf593ca9e0fb51acb1d7c327ca9351a3d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/83d1edf593ca9e0fb51acb1d7c327ca9351a3d14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
