)]}'
{"id":"openstack%2Ftaskflow~1005592","triplet_id":"openstack%2Ftaskflow~master~Ic52d24bb441d33eee257ff0a3291eb7372319d36","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-14 17:57:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic52d24bb441d33eee257ff0a3291eb7372319d36","subject":"Remove logic for Redis \u003c 2.6","status":"MERGED","created":"2026-09-14 15:14:58.000000000","updated":"2026-09-14 17:59:08.000000000","submitted":"2026-09-14 17:57:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005592","meta_rev_id":"af80eeb8f9d7b46ca2937acada3b82d5189c3378","_number":1005592,"virtual_id_number":1005592,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 17:57:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-14 15:36:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-14 15:36:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 15:36:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-14 16:01:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2dcedb971311927e3b39e65087e2bb937ed546d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 15:14:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9cde9a709f282f673e2217545ce1f09fc26af7c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 15:36:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1432a7f4a434db31e11d5da0c91073039aacf90a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 16:01:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/478725a4f75347639201b5d68de0152b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91797f9877754a9ab409172dbfefec30 : SUCCESS in 44m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3ad15dea7054d65b924aceaeb1748dd : SUCCESS in 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cf2b0fb48714432a4b1cd408d7aa284 : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fcad8c7e86a84f4bbe94ec99c5cb7cca : SUCCESS in 12m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e5cb7652cfc40e8b630aba214466c46 : SUCCESS in 11m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2bb791a1f8ec43e1a0ea10de73537aa0 : SUCCESS in 12m 47s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/fd4f6ab277414864acf94d1aca74ec74 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bca9109f3164d6b8ccc1faf41c4421c : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f507ed1e77d54e50b48b43bdc3c46e3b : SUCCESS in 5m 35s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/de53fae651a942fb85080641012efd0a : SUCCESS in 8m 53s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/4a494bf12e7e423a9f8592be5745b446 : SUCCESS in 6m 43s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/fb52f77144eb494ab56442d283b6b154 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e9a988f7a260a1012ad350518997b7a7b185fe03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 16:40:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e66ce38e6bcc1b00c1cf3dce03ddae8b9fb63efa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:57:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07b740e6bb9842ffa22897f17f0f5200\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffae2a15d4f34d478c789a048f41936f : SUCCESS in 57m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4acf7019a900451baeae1a153b664f9a : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c0e2b48764f454d9fe7dd1c612d6f38 : SUCCESS in 9m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd5f0100121f4670958611b36c38db0d : SUCCESS in 9m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e1887590fbb54b51aa1aa12578d075c9 : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bba60664ba34e4f80de868fd0745363 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66bbede1016845a18e3dd2776fe7a3bf : SUCCESS in 3m 54s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/fad21df4e4a34d838a878b00c6bed96c : SUCCESS in 6m 50s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/2a777b2647c7430ca12ee7564ef8bb33 : SUCCESS in 6m 15s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/07b31d109b0e43b39b95f7d28e9c982b : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"448aecda50f8d3132079bb346a8c8c0239e07bd0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:57:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"af80eeb8f9d7b46ca2937acada3b82d5189c3378","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:59:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2064e97107e547909f2c2d358c2679f7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3443e44df94f498795a611ae2b8c5183 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74cdd19643b367a8a28051aebd288eba4f6bb290","revisions":{"74cdd19643b367a8a28051aebd288eba4f6bb290":{"kind":"REWORK","_number":1,"created":"2026-09-14 15:14:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/1005592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/92/1005592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/92/1005592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/92/1005592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/92/1005592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/92/1005592/1"}}},"commit":{"parents":[{"commit":"8322d2c290e52fd7703ad252bf28da2007735450","subject":"Use oslo.utils to parse redis version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/8322d2c290e52fd7703ad252bf28da2007735450"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 15:08:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 15:14:33.000000000","tz":540},"subject":"Remove logic for Redis \u003c 2.6","message":"Remove logic for Redis \u003c 2.6\n\nCurrent redis jobboard driver has required Redis \u003e\u003d 2.6 . So the logic\nis not actually used within taskflow.\n\nRedis 2.6 was relased in 2012, so no user may attempt to use such\nold version these days.\n\nChange-Id: Ic52d24bb441d33eee257ff0a3291eb7372319d36\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/74cdd19643b367a8a28051aebd288eba4f6bb290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/74cdd19643b367a8a28051aebd288eba4f6bb290"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
