)]}'
{"id":"openstack%2Ftaskflow~1005593","triplet_id":"openstack%2Ftaskflow~master~Ib0418c71054e5d89cb78bd973a89c4d45ec17884","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-14 17:57:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib0418c71054e5d89cb78bd973a89c4d45ec17884","subject":"Fix a typo in error message","status":"MERGED","created":"2026-09-14 15:17:33.000000000","updated":"2026-09-14 17:58:14.000000000","submitted":"2026-09-14 17:57:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005593","meta_rev_id":"4d7a539c88d3e2cbf369a3eb8bd6058004ae96ed","_number":1005593,"virtual_id_number":1005593,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 17:57:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-14 15:35:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-14 15:35:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 15:35:52.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-14 16:29:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4d113fb20d68f58f1017bdee395d2c1478ead8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 15:17:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be5077d8a7b8fce1e572e4d4f29e558a644d4e54","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 15:35:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"251fc19c3821f873532bf36c970b216da457e4f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 16:29:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3792a0387da84b46b6c4931cd25956d8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0226acad5b4f4c739f5be9eaa8dbfd89 : SUCCESS in 1h 10m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d906efd094e4a5a800d339b15cbd778 : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a93eee6191f948199a2040b021d8f9de : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32236158aa3f4d929e029d083ffe1b22 : SUCCESS in 8m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1fcf6f6782124a098d3aa9bd4a6dadc6 : SUCCESS in 8m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/876a8f6b728c452986bfd45243fe6d4a : SUCCESS in 14m 28s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0e38dfd001814ac89dad089f6aceff53 : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eca4b0daaf724ca9ba1cda49c917e954 : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f69b5f3522c246db80e699675862ccbe : SUCCESS in 4m 43s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/585dc04571a9485086b2991f59a83edf : SUCCESS in 7m 26s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/6fd4e6e9093c49ce9aa05ec5d2627aec : SUCCESS in 5m 24s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/bc55f7a7937444fba7d45fda4650e210 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1901e88fee026a05d2888bfd3300ddd13b6cf85a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 16:30:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c3fd2fecf3c07c2d37c8f278844262670cb269a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:57:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19859ec1f31a4279a526c70c70619f05\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d81cd82d1bb34c1c88c786ba5a2cf675 : SUCCESS in 1h 25m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb81a8edd93f40218f140f66eb6734fa : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8dc99c2fc23b4c64b027309ff5abc30c : SUCCESS in 8m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0e6d04f936842e98ac487703902019f : SUCCESS in 8m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c8949f9139dc4f87a86640376d123548 : SUCCESS in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f633b278f14244a1805ecdbcc2d4513a : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd3c569a26584055b9e5d37615190a64 : SUCCESS in 4m 51s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/f2be35a23afb4fbbb7541cbac1a6d223 : SUCCESS in 6m 19s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/3d5315416b8d496a883d944bc5d20def : SUCCESS in 6m 41s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/2cdb905970f1449fbb9435546ba4f4d1 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2466d71150bd84a1e32afbc0ae854b7cabbb7932","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:57:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4d7a539c88d3e2cbf369a3eb8bd6058004ae96ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:58:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5700452e9426493492ca155cf1f7e718\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/856b228c2fcd48f5a8342159b58e781b : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"456b6013328df21095e3054a66abbf623ba1a857","revisions":{"456b6013328df21095e3054a66abbf623ba1a857":{"kind":"REWORK","_number":1,"created":"2026-09-14 15:17:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/1005593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/93/1005593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/93/1005593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/93/1005593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/93/1005593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/93/1005593/1"}}},"commit":{"parents":[{"commit":"2b64257cd615a26159514840b11f78bdbdbd0a1a","subject":"Merge \"Use enum directly to convert raw values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/2b64257cd615a26159514840b11f78bdbdbd0a1a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 15:16:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 15:17:24.000000000","tz":540},"subject":"Fix a typo in error message","message":"Fix a typo in error message\n\nTrivial-Fix\n\nChange-Id: Ib0418c71054e5d89cb78bd973a89c4d45ec17884\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/456b6013328df21095e3054a66abbf623ba1a857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/456b6013328df21095e3054a66abbf623ba1a857"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
