)]}'
{"id":"openstack%2Ftaskflow~1005772","triplet_id":"openstack%2Ftaskflow~master~Ia4cc9ecdc3c9fa2257373a260d1f5169a825c508","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-15 22:32:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia4cc9ecdc3c9fa2257373a260d1f5169a825c508","subject":"redis: Correct exception raised for incompatible redis version","status":"MERGED","created":"2026-09-15 17:41:53.000000000","updated":"2026-09-15 22:33:45.000000000","submitted":"2026-09-15 22:32:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005772","meta_rev_id":"9ba08bdafbe71e8abc851de512574acbe0bab1e0","_number":1005772,"virtual_id_number":1005772,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 22:32:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-15 20:57:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-15 20:57:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 19:43:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-15 20:57:43.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"14b693aafdceb9be8b857ea5945106e862dab5af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-15 17:41:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06b1f4db8c6820e89e1f3678b1305e7da887978e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-15 17:49:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bb225683da9163281dec260019033939cca449cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 19:43:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/089cfec2935a4d27aacb5f0f0f2dcde1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4618a329b004677a02a26678f86ba6c : SUCCESS in 1h 46m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de7db82d0fae46e8a037a86e1a8a8569 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d7077d1b4784391a5193e48aba2cd21 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e850612b7c74d8ca5c961033ace7a17 : SUCCESS in 9m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46ea53e7d04d4b86b48dea7b8365c055 : SUCCESS in 10m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/617959ad83284765b1809d6df54dbb38 : SUCCESS in 13m 35s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/8a26158c4a1446d499e30edc01e790f0 : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6394de9deee04fb4a8364ef40d0196da : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cef067bb164c47c4b8380f589139b663 : SUCCESS in 5m 35s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/aae273d12be94368b7e2914409b319db : SUCCESS in 5m 22s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/ace2dedc9f2f4283b36ee83afc366e67 : SUCCESS in 6m 41s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/55f94d94d47c4fb6b1dba3015da1bdca : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"a2fb5458356c50251030dbaad7a0fc15a6b4469e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 20:57:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e148f9041a3184408064c2c65a0ca0afd3e8b7dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 20:58:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f38af00a3b0d1b1a6597f4883712f44699840879","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:32:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b62a7f1c924599a5e3c9a87b376825\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/227c0decb6484486aff05e8e2f06c4e1 : SUCCESS in 1h 31m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9d0ad13b0b84d2ab24b57d06ef81758 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/887a31da1afb4fbaa03e7de3269d43df : SUCCESS in 10m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52b991b0cb614fee9606403199809e87 : SUCCESS in 10m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c13dad6b8d7947818a1b654db4f27e5a : SUCCESS in 12m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d40a0380f5b45368f21aebe0e40fd8d : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de7acdd527d34d049804414658adf09a : SUCCESS in 5m 46s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/9b70b2cb197a456babc580a204c25de4 : SUCCESS in 7m 24s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/b5407ccb2f2b4784a2246db73c791e9c : SUCCESS in 6m 19s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/acf44a89895f48acb9846118b795f2a6 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"5fa2e9170fa8282fe2f01572132e9d7844a56557","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:32:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9ba08bdafbe71e8abc851de512574acbe0bab1e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:33:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19369f9cad44497eada7b5b35bb89221\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/077233aee618452a91f6636f108898ea : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef3fa566ad664af397ee89360a556ee4 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"672c8a213d568d77e782ef28e0b83dca74e58b74","revisions":{"678dc4d14d0038d63b45e36e7df2072601a8f27f":{"kind":"REWORK","_number":1,"created":"2026-09-15 17:41:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/1005772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/72/1005772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/72/1005772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/72/1005772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/72/1005772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/72/1005772/1"}}},"commit":{"parents":[{"commit":"a140b5485e994b794dcab8bba596f4c3acf6b643","subject":"Merge \"Use close method from base class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/a140b5485e994b794dcab8bba596f4c3acf6b643"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-15 17:38:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-15 17:41:43.000000000","tz":540},"subject":"redis: Correct exception raised when redis version is not compatible","message":"redis: Correct exception raised when redis version is not compatible\n\nWe have a specific exception to notify version incompatibility. Using\nJobFailure is confusing because it looks transient or caused by\na requested job.\n\nChange-Id: Ia4cc9ecdc3c9fa2257373a260d1f5169a825c508\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/678dc4d14d0038d63b45e36e7df2072601a8f27f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/678dc4d14d0038d63b45e36e7df2072601a8f27f"}]},"branch":"refs/heads/master"},"672c8a213d568d77e782ef28e0b83dca74e58b74":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-15 17:49:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/1005772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/72/1005772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/72/1005772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/72/1005772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/72/1005772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/72/1005772/2"}}},"commit":{"parents":[{"commit":"a140b5485e994b794dcab8bba596f4c3acf6b643","subject":"Merge \"Use close method from base class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/a140b5485e994b794dcab8bba596f4c3acf6b643"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-15 17:38:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-15 17:49:35.000000000","tz":0},"subject":"redis: Correct exception raised for incompatible redis version","message":"redis: Correct exception raised for incompatible redis version\n\nWe have a specific exception to notify version incompatibility. Using\nJobFailure is confusing because it looks transient or caused by\na requested job.\n\nChange-Id: Ia4cc9ecdc3c9fa2257373a260d1f5169a825c508\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/672c8a213d568d77e782ef28e0b83dca74e58b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/672c8a213d568d77e782ef28e0b83dca74e58b74"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
