)]}'
{"id":"openstack%2Ftaskflow~1005775","triplet_id":"openstack%2Ftaskflow~master~I486459439b67657aa164bf3f4621bba3a00bb246","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-15 22:32:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I486459439b67657aa164bf3f4621bba3a00bb246","subject":"Remove unused async_utils module","status":"MERGED","created":"2026-09-15 17:49:16.000000000","updated":"2026-09-15 22:34:46.000000000","submitted":"2026-09-15 22:32:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005775","meta_rev_id":"e21f5696e04a87a07ed8eebee9c861b2b48f2dac","_number":1005775,"virtual_id_number":1005775,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 22:32:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-15 20:58:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-15 20:58:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 19:00:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-15 20:58:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"b6a560048d094e0e5ce478eb20eb9c55dbaec385","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-15 17:49:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8979982c086f69909a892bee4b4cfb968454a67c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 19:00:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a09e87471d34545baa2d250757c720b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7717ba888ee044c88682b855c5d89c9d : SUCCESS in 1h 06m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/791b595a82a6437fa98e0e59e39c5464 : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a62ffedae214a629385d8c60cebaff7 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04c10c03b6464ede9465c949bd7e20aa : SUCCESS in 9m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce4e7175223d4c2c801578b8d59b7692 : SUCCESS in 11m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d56e1d5e55b140f69e3a8120118ef209 : SUCCESS in 12m 30s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0a8f944cf01549c98e61aa0ba0538bba : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ebe88ea47c43869b6fac0e839236b3 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6c3d8df33474092a1d40c44984f664d : SUCCESS in 5m 05s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/9aed245b2492405b96c33a89db7f68ed : SUCCESS in 6m 15s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/20fa49da621f4d588ac160f25a0d081b : SUCCESS in 6m 58s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/2ee8429b5233439698322492881a2c2f : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b34411317b922cf6c3d52a9e4dd786bfc5e0a5f9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 20:58:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cf8acd064f9831926a8af5eab64f3f7b3e3319e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 20:59:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d10b891650781e5eb6df70c04e99f05f1e33b87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:32:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02afddee5512439886a13c5853a30cbf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b20bb6bded894df8b3d27c69fb0a22eb : SUCCESS in 1h 28m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/651e342eb1d848ba8def328719f7ee0d : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0581c1b2870b40e592c4d5fb9a59df4d : SUCCESS in 13m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4b04b3170d5491ba86d9e9f95fba15c : SUCCESS in 8m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1477c30755f440ddab02b80f4bc2fd56 : SUCCESS in 12m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a35b8f74ba2245f9ae2de3f22809b6f8 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b378f6834c064efe91ee88eb00e66623 : SUCCESS in 2m 43s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/843c48d336494f7686bda395fb9fe3a2 : SUCCESS in 8m 54s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/7094da84ada144b3924189a00dfbacf6 : SUCCESS in 8m 22s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/1826feffd64a4f4cbd134c4618f34763 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7363e5b68533599aa500d5127cfc17fca564ebe7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:32:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e21f5696e04a87a07ed8eebee9c861b2b48f2dac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:34:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a93558a699e43728e7354182c534f1f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fcc1d3b9d714b5aaa2beb6a96a24b5d : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fa24aee614b408c95e2327390da1529 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"89974c8743fbe8c932a9c257905dc9d9e1abe854","revisions":{"89974c8743fbe8c932a9c257905dc9d9e1abe854":{"kind":"REWORK","_number":1,"created":"2026-09-15 17:49:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/1005775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/75/1005775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/75/1005775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/75/1005775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/75/1005775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/75/1005775/1"}}},"commit":{"parents":[{"commit":"a140b5485e994b794dcab8bba596f4c3acf6b643","subject":"Merge \"Use close method from base class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/a140b5485e994b794dcab8bba596f4c3acf6b643"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-15 17:47:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-15 17:49:06.000000000","tz":540},"subject":"Remove unused async_utils module","message":"Remove unused async_utils module\n\nThe module has been unused for 10+ yeas, since 1.25.0 release [1].\n\nThe module is directly removed because it has been stated that these\nutils module are mainly for internal use only and can be removed\nwithout notice.\n\n[1] ae9c701f9073941fbe063d2b7854ff6eed5b5fc0\n\nChange-Id: I486459439b67657aa164bf3f4621bba3a00bb246\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/89974c8743fbe8c932a9c257905dc9d9e1abe854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/89974c8743fbe8c932a9c257905dc9d9e1abe854"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
