)]}'
{"id":"openstack%2Ftaskflow~816798","triplet_id":"openstack%2Ftaskflow~master~I9078086cd4c3dbfa954422dfa3d5b19fefd7264f","project":"openstack/taskflow","branch":"master","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2021-11-15 12:59:11.000000000","reason":"Hervé Beraud replied on the change"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-01-10 22:56:19.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-01-10 21:36:13.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I9078086cd4c3dbfa954422dfa3d5b19fefd7264f","subject":"Handle invalid redis entries in RedisJobBoard","status":"MERGED","created":"2021-11-05 09:24:39.000000000","updated":"2022-01-10 22:57:59.000000000","submitted":"2022-01-10 22:56:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"816798","meta_rev_id":"3d79d7927566ef042b18d769bb31a1de75a792e7","_number":816798,"virtual_id_number":816798,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-10 22:56:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-01-10 12:58:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2022-01-10 21:36:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2022-01-10 21:36:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-05 10:58:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-08 10:03:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-11-15 12:59:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-01-07 17:41:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b8170d021fbec5d380757c60df4b978faf4b6a10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-05 09:24:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1261f16cfe49e814ee638ac6d8b063f7702b801c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 10:58:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec61110fb6e94b1db7f29fa93e14cf6c : SUCCESS in 1h 28m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd01b83f7721483eb86a8dde136a0a62 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fa886dea898439f9ebb255d9800243d : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0944cbf9ce545f9888d573e92d4981c : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c9a70e5542044ad89631562b9aec71a : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85cc8865b777485fa05e33f7e1a2af8d : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f105b0153fc44ef7ab1c689b4ff1cf2f : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"99208f7f7e36ce7f82c4262efd27c3c9b40f7474","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-08 10:03:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f17c8dfa94c0aa82fc04cd63205b32e67607c50f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-08 13:05:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ee4be26e708694645462db785ebf2ce943eb0fd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-15 12:59:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aafab66a788d58dc81486bd27bdc0966a02ed660","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-07 17:41:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0432fc5921c1e46988a30227feb31816263ad2cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-10 09:35:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c1385f8b13064823ca8def1e80460185f5f2e22e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-10 09:36:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15a387b7021b5b2dd9b7a378d7d0b83da75a7523","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-10 10:30:18.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"251dafd45e75b9646d23af568f3b9f361b8fb1ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 11:05:08.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e22577da3e240f2a53187531f695b80 : SUCCESS in 1h 26m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85d31f1ef4634a7f91ebea19f650d335 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/790b3582b08346f2b16bbfed2173fe67 : FAILURE in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9574c5094ff4dda985b992013c74c02 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/793d81197d03479b9677371daddeb9a0 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9261f74974f64bf5a0993144528f936d : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b6a732799f4096a5dad544ac4776fd : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"74bf79cd494001af8c9916d70cbe6aa5429083ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-10 11:40:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4091109115ae716d8b355290922bbcecd05126c0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-10 12:58:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6289627f34b269cc84f259f467cea2a0233d52ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 13:20:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d035a1ff95042a8b3e7f862af1607d9 : SUCCESS in 1h 34m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e12838c58a34b1eada5c66361a5f1fb : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffaa2cd8ff05433daf019da14c1550ab : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15a8fc7916ae4495a8220fd91f600f98 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d23e215e6c01427b9eb0c42f045044ef : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e17a9d41898f400cbd35a595cee170ad : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e375e23392cf4f5f85cd118c1ab8dd70 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":3},{"id":"4ce8c22bae68e51941ddcbc48d84d2686bd1c84f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-10 21:36:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"520e7784e26321d7844d0354b9a2a6dbcf324401","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:37:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"11a3dee4b9572284f89285aedd8fcf9a2e68bee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 22:56:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0d697419a194d3c9406b610c5fad8d0 : SUCCESS in 1h 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74ab01381e784b679379d3f06a55741b : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5c8a9ad30514708b5005fabc5d05c89 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd325a26789849c18b89d4644c407dc7 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2682fc1ad40a4e2db2703445a88f5380 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":3},{"id":"1952acffcb0b231b4d76ac3ea46ca6f6302c11be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 22:56:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3d79d7927566ef042b18d769bb31a1de75a792e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 22:57:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da0de213251d4479b1fe55ac7be2857e : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ba2cbd72f9deee5fedf15864dac41471f4402c82","revisions":{"0009940209852821797615fc6d025faebd83aa67":{"kind":"REWORK","_number":1,"created":"2021-11-05 09:24:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/98/816798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/98/816798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/98/816798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/98/816798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/98/816798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/98/816798/1"}}},"commit":{"parents":[{"commit":"6d9330a1b0afaa1fdb8c995c009b4ac90db94c00","subject":"Merge \"Use custom JSONType columns\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/6d9330a1b0afaa1fdb8c995c009b4ac90db94c00"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-05 08:23:35.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-05 09:23:29.000000000","tz":60},"subject":"Handle invalid redis entries in RedisJobBoard","message":"Handle invalid redis entries in RedisJobBoard\n\nWhen catching that a redis buffer is not formatted correctly,\n_fetch_jobs now deletes the data from the hashmap, so the same error\nwill not occur during the next call to the function.\n\nThe commit also adds JobFailure in the list of caught exceptions (might\nbe raised by _loads) in _fetch_jobs\n\nCloses-Bug: #1949950\n\nChange-Id: I9078086cd4c3dbfa954422dfa3d5b19fefd7264f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/0009940209852821797615fc6d025faebd83aa67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/0009940209852821797615fc6d025faebd83aa67"}]},"branch":"refs/heads/master"},"3df4c396d996179203230c860ab80fdd86f06358":{"kind":"REWORK","_number":2,"created":"2022-01-10 09:35:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/98/816798/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/98/816798/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/98/816798/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/98/816798/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/98/816798/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/98/816798/2"}}},"commit":{"parents":[{"commit":"6d9330a1b0afaa1fdb8c995c009b4ac90db94c00","subject":"Merge \"Use custom JSONType columns\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/6d9330a1b0afaa1fdb8c995c009b4ac90db94c00"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-05 08:23:35.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-10 09:35:32.000000000","tz":60},"subject":"Handle invalid redis entries in RedisJobBoard","message":"Handle invalid redis entries in RedisJobBoard\n\nWhen catching that a redis buffer is not formatted correctly,\n_fetch_jobs now deletes the data from the hashmap, so the same error\nwill not occur during the next call to the function.\n\nThe commit also adds JobFailure in the list of caught exceptions (might\nbe raised by _loads) in _fetch_jobs\n\nCloses-Bug: #1949950\n\nChange-Id: I9078086cd4c3dbfa954422dfa3d5b19fefd7264f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/3df4c396d996179203230c860ab80fdd86f06358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/3df4c396d996179203230c860ab80fdd86f06358"}]},"branch":"refs/heads/master"},"ba2cbd72f9deee5fedf15864dac41471f4402c82":{"kind":"REWORK","_number":3,"created":"2022-01-10 11:40:37.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/98/816798/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/98/816798/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/98/816798/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/98/816798/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/98/816798/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/98/816798/3"}}},"commit":{"parents":[{"commit":"6d9330a1b0afaa1fdb8c995c009b4ac90db94c00","subject":"Merge \"Use custom JSONType columns\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/6d9330a1b0afaa1fdb8c995c009b4ac90db94c00"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-05 08:23:35.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-10 11:40:28.000000000","tz":60},"subject":"Handle invalid redis entries in RedisJobBoard","message":"Handle invalid redis entries in RedisJobBoard\n\nWhen catching that a redis buffer is not formatted correctly,\n_fetch_jobs now deletes the data from the hashmap, so the same error\nwill not occur during the next call to the function.\n\nThe commit also adds JobFailure in the list of caught exceptions (might\nbe raised by _loads) in _fetch_jobs\n\nCloses-Bug: #1949950\n\nChange-Id: I9078086cd4c3dbfa954422dfa3d5b19fefd7264f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/ba2cbd72f9deee5fedf15864dac41471f4402c82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/ba2cbd72f9deee5fedf15864dac41471f4402c82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
