)]}'
{"id":"openstack%2Ftaskflow~975760","triplet_id":"openstack%2Ftaskflow~stable%2F2025.2~Iced8a1fd02ef5766f4017bb1b6c6d48b4c061b5c","project":"openstack/taskflow","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2026-09-02 11:54:59.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-02 08:21:20.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Iced8a1fd02ef5766f4017bb1b6c6d48b4c061b5c","subject":"Avoid iterating over the same atoms in a graph","status":"MERGED","created":"2026-02-05 07:15:37.000000000","updated":"2026-09-02 11:56:01.000000000","submitted":"2026-09-02 11:54:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975760","cherry_pick_of_change":974999,"cherry_pick_of_patch_set":4,"meta_rev_id":"2f7d27472f36423c6c0aa1ba7ead7570d05a6cdc","_number":975760,"virtual_id_number":975760,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-02 11:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-02 08:21:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-02 08:21:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-05 07:15:37.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2026-02-05 09:07:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-14 01:30:00.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-04-03 07:01:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-08-31 09:11:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"0f06302e0799e52b989b0f42a78a047d910b1007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-02-05 07:15:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"64c0db0d3f04f0566cf0226e2491944babacd1ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 09:07:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbe227a3caab418bb933c06a91572636\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56dbaa234024472eb9c523c2e4064ecf : SUCCESS in 1h 44m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19a162f28c0e4f4fb4059953096db985 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f49c9e5cdc60417cbc1695d945a44185 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2785ffd19c7345a4ba3ec1cb5566cdf6 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72860e496c964b42a1382c1d3ba912b2 : SUCCESS in 5m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5c7178ea70f4a39abe747713544b899 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b89f73ca72f744baaadb6af388970413 : SUCCESS in 2m 58s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/5051c8b55f314f699a0211b85b590e24 : SUCCESS in 4m 30s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/d514aa5d362647e5ae227f8328254048 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ec712f751513cf080ad9737cb5aedc8a78afb008","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-14 01:30:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5091a5e4be54cf2eaa412e9d902706c6a18eb61c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-14 01:30:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e4eebd63ddc8e6ab17493a4ef82a84d3bca7380","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-14 03:18:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30d3c655913043d7ae70c7a0d3937dac\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d85fed45bac4529a11a9a0b1d46c524 : SUCCESS in 1h 39m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/972e240bf8d143879fb7cf1d77ff0b5c : SUCCESS in 2m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2bf0e38e3a84c83b04477500d4e40bf : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf2d36b48b354be7a69faf9d39787cc7 : FAILURE in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68beb594efc4485fa5a55c761b42f160 : SUCCESS in 5m 15s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/50b3966d8b7b4bcea92b9b37a50aa3c2 : FAILURE in 4m 54s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/9e4f5d46f8704ce486b5f15cad2ef3fd : FAILURE in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0743e58c7fdf5d204c426c0913e4dba130f8db2b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-02-18 15:00:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1397c3f55bdd119e218b6f283d72cf979b5fd42c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 16:51:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6766848ac7774c6ca1765f41f322463d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4e313e8c82840ddb4b6cc3c6282ea9f : SUCCESS in 1h 46m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce4b3afc4e1642619857f16dedf4920b : FAILURE in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60fce1bb03924cab9d970c5f6eecdf69 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/453cfac2d7ae448ab143d0e050cef717 : FAILURE in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8686e769617746a19f533bfa0617869f : FAILURE in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3196a9c0cec94c01b1aa781bb812b516 : FAILURE in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dec83379c154989ac5a0ed4903303f2 : SUCCESS in 5m 51s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/95bc01f2e31a40919e9e5fe1419aaea1 : FAILURE in 4m 30s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/6d034ef273924b9eb0345581da6c9c5a : FAILURE in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e67145e3193d3ff33c68697d8de87d27d0b19696","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-03 07:01:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0d9a37a80b9a39f0b4c3122a6af1027d9d658dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 08:49:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b0c84dc506e4c63b8dc6f8da8d9df40\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe2886147a184f87a9608dd5895d001f : SUCCESS in 1h 41m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3c7cba9439242d8b1f1a43ccf34d84a : FAILURE in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49acfd71f584abf9ab9cf3be2c4b145 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36c3053870864283a9e1be0929d18c4d : FAILURE in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba34d9b263814994bea814001a07a749 : FAILURE in 5m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/41c4d10ff3044764a36bf7146f34fb34 : FAILURE in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d981b0b828c548ae8c4ee367a115dd7e : SUCCESS in 4m 38s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/a7133427e429469cb1fb6d60d4b8c304 : FAILURE in 2m 46s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/7d629c8658d94676a47f179bc44ca5e9 : FAILURE in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"88083b996b0685b122701d99e9898552ac7a0a34","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-04-17 14:05:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be7ed97b814e8ea5b32059a7cac896d2341e90f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 16:10:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89d7352043124cb2b8b1062527dece66\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/493f635b365a46d394f09d679117bceb : SUCCESS in 1h 57m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/841251142c7245f8a8e835c65818764d : FAILURE in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe3147c67c0a493ebcda008f8a6712fe : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca4469cf3f894ce8a60ad74b99b815d8 : FAILURE in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05deb3962c694abdb7506b10125ffdb7 : FAILURE in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/85fe79e5bc5944229d2d23fe7006d7a6 : FAILURE in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/308b61f235754344bbeed64707db82a4 : SUCCESS in 4m 57s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/1768a1c487b84edaaf03746b5c6ff5cf : FAILURE in 4m 39s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/d1986e137bf04512a7f12dd87e595d78 : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"236dc4959fd309000f58b28cda177a6f01fade88","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-31 09:11:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e363667a3fdab3c82fd7222b8fb34c64bc0495d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 11:07:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/024835b9829540c6a1db53aeac1bb865\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68a71df4b01149ab9c965d74db2dde0f : SUCCESS in 1h 49m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e3759fae3cc4c06ae7d10c8344176d7 : FAILURE in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/762c71c4fa3b4472be448a61f3db18f4 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b470092f3a88450cbf34aac6486430c1 : FAILURE in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5b8a77c1aa148b99598fd798feb6fce : FAILURE in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d25fdfee56a6460e8c7bf000ede7927b : FAILURE in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f598fff9b08749a2991ad76a16f5c867 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67aa1acfa784498d90831d90b0fa0742 : SUCCESS in 5m 12s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/2befde7628ba49868a702f31d1de6e56 : FAILURE in 6m 45s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/2fddb1f2ee2747d6983e8ddd4dd74566 : FAILURE in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a51c6791a7046a6e69a62fed2a4c2a80acec1298","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"real_author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-31 12:43:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_16312\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}],"_revision_number":2},{"id":"5200e72820530d7819d2a7f32e2aedaf46e70e53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 13:38:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc3d2f62eb414ef39b40f540279a6b4b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b4d13ec2cc64df6aeaec9d9fcaca433 : FAILURE in 46m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a33660d649d4d388035c0326cdfea29 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d353b8825bf040b6aa9b1a11f94ec02a : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bddbb4ac38c8481dabf745d506516ebc : SUCCESS in 6m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf377a447ce848b69703700dbecead19 : SUCCESS in 6m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b93961d31534dc6b826897edb6a0a11 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae29283d70f844f99d839b0dc0f6f742 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59b0f45d6c0d400688aed6cfb844b47d : SUCCESS in 4m 14s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/04c9fdc3bf5141abafa02c5d7593257c : SUCCESS in 7m 25s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/3653bceb21bb45b58ce3b17e1699b3fe : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":2},{"id":"5308cd96aaa1bdef567f43ed2b8849015f6a1391","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-02 08:21:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7942651a40dc7090303a74be67715cc7c8bcabd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 10:03:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3980eb3cded44118ac46c9b3c2627a2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4fbcad93e264946bed3a6bfa5d745fd : SUCCESS in 1h 39m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22f9f625cc3749af9982bd7c1726ff35 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73e3f4d9a0124f868c20d65d1ff2125e : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff1d2c53589b4e0ab3a4812dc772f6bd : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/205f267207b7456fb5e403b84762cb86 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76f9def9cf9240fdb77c1984e747667b : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19f5b27e90d547f8b2609ebe87266787 : SUCCESS in 3m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42afbc9c5dca4828bfb5cdd3f51e9c00 : SUCCESS in 4m 39s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/8ea328e576894b4a98452ab687b88aef : SUCCESS in 5m 28s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/eb0603f51bce4f30b87b65063729048c : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"8576d66b4a030c4fc14b65b8373ce8eaaa89fa9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 10:03:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ea499a5ca87cc74a89bc42381a5910833a4361c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 11:54:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25f93974a63a41af97bed02e75af035f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b52137839cd7432e9bc140ff74643b65 : SUCCESS in 1h 45m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/552d7b3993b54fb8ab3e6c5b7febe0d4 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29cbfd850e144ad8bb0274dcec61dc3f : SUCCESS in 6m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/519f0cc8d9404b0f9b4b353033e5ada3 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89347687d1164b0ba59a88c39376a343 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1333ef1a1e384d12b8f76cd812ee91dd : SUCCESS in 4m 49s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/2130d3a7d38c430bb287b0ec6301e2c7 : SUCCESS in 7m 53s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/41f15f7241c54b69926b848cf90fe3ed : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"da473ae8c2e75de1a4ff831004d304d5e8f6fb7a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 11:54:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2f7d27472f36423c6c0aa1ba7ead7570d05a6cdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 11:56:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48d8544f030e42308ecdab869a2bc471\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a27d613b945448fac3fae2ed233c922 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ffe5977e404439570e0630e09cb6fcb03b71f92c","revisions":{"c375a6ed89ba7f1bbe7ad6408121f2f7f9a948ac":{"kind":"REWORK","_number":1,"created":"2026-02-05 07:15:37.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/60/975760/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/60/975760/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/60/975760/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/60/975760/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/60/975760/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/60/975760/1"}}},"commit":{"parents":[{"commit":"9b186ac8b827532af2419d1560a5644ce0da31ad","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/9b186ac8b827532af2419d1560a5644ce0da31ad"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-28 12:31:33.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2026-02-05 07:15:37.000000000","tz":0},"subject":"Avoid iterating over the same atoms in a graph","message":"Avoid iterating over the same atoms in a graph\n\nWhen iterating over a \"diamond\" graph, some atoms may be processed\nmultiple times. Depending on the size of the graph, it has a huge impact\non the complexity of the algorithm.\nThe patch ensures that each node is processed only once.\n\nCloses-Bug: #2139228\nCloses-Bug: #2086453\nChange-Id: Iced8a1fd02ef5766f4017bb1b6c6d48b4c061b5c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 6fc3a9b0f1868e3a6092007bb0fd8e79df6f5d95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/c375a6ed89ba7f1bbe7ad6408121f2f7f9a948ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/c375a6ed89ba7f1bbe7ad6408121f2f7f9a948ac"}]},"branch":"refs/heads/stable/2025.2"},"ffe5977e404439570e0630e09cb6fcb03b71f92c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-31 12:43:44.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"real_uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/975760/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/60/975760/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/60/975760/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/60/975760/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/60/975760/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/60/975760/2"}}},"commit":{"parents":[{"commit":"fdbbdbca4d9b285422cec955196656f1257c7f0b","subject":"stable-only: Pin setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/fdbbdbca4d9b285422cec955196656f1257c7f0b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-28 12:31:33.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2026-08-31 12:43:44.000000000","tz":0},"subject":"Avoid iterating over the same atoms in a graph","message":"Avoid iterating over the same atoms in a graph\n\nWhen iterating over a \"diamond\" graph, some atoms may be processed\nmultiple times. Depending on the size of the graph, it has a huge impact\non the complexity of the algorithm.\nThe patch ensures that each node is processed only once.\n\nCloses-Bug: #2139228\nCloses-Bug: #2086453\nChange-Id: Iced8a1fd02ef5766f4017bb1b6c6d48b4c061b5c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 6fc3a9b0f1868e3a6092007bb0fd8e79df6f5d95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/ffe5977e404439570e0630e09cb6fcb03b71f92c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/ffe5977e404439570e0630e09cb6fcb03b71f92c"}]},"branch":"refs/heads/stable/2025.2","description":"Rebase","conflicts":{"base":"9b186ac8b827532af2419d1560a5644ce0da31ad","ours":"c375a6ed89ba7f1bbe7ad6408121f2f7f9a948ac","theirs":"fdbbdbca4d9b285422cec955196656f1257c7f0b","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
