)]}'
{"id":"openstack%2Ftaskflow~996934","triplet_id":"openstack%2Ftaskflow~master~I3f238ad98044e164234f1a34b36ec0d333362a4a","project":"openstack/taskflow","branch":"master","topic":"tox-use-extras","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-23 23:35:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f238ad98044e164234f1a34b36ec0d333362a4a","subject":"Remove duplicated optional dependencies from test requirements","status":"MERGED","created":"2026-07-12 10:51:16.000000000","updated":"2026-08-23 23:37:23.000000000","submitted":"2026-08-23 23:35:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996934-tox-use-extras","meta_rev_id":"397603974ce0467a860cb2b55a3ec555f06c4e2c","_number":996934,"virtual_id_number":996934,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-23 23:35:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-23 22:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-23 22:11:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-12 11:33:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-23 22:11:19.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"e9c7e34b3293ccf47e2f6d89ee7cde607b0d2991","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 10:51:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"228652ed6de1fb7a8c881e12e0f1ea4d6253995b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 11:33:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2b66b0b3bb54e2dbb87050227de6b05\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c5353c7de7624c9cb313575e50698229 : SUCCESS in 2m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3c42306fb814ff7a632539e6b586b80 : SUCCESS in 40m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd682fd4dc7e4bdfafcbd11af01a98a2 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc893f19f46548888f6770e7a54cdc61 : SUCCESS in 2m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/889485ef2e2f4dd1a52d31a1a98fe6b5 : SUCCESS in 9m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c4d192bb7694a8da93d7ce2b4a223b3 : SUCCESS in 10m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/45746399fcdc4346b20a8a53d632e563 : SUCCESS in 11m 49s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ca4ba36ec404449583260cc7e9f808d4 : FAILURE in 12m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f19ade7306404d87f80c0fa000bbe3 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5544827d077f4d188df81438b0307546 : SUCCESS in 5m 15s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/6419b3491c34459a85eac577d7a495f9 : SUCCESS in 5m 12s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/9d232f1f84a744b6a3c98c349bf4185b : SUCCESS in 6m 48s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/bcc41a90df9c449e89ad68e30facd33e : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b125624277269adc21a3afea3cfc2f59db5b041c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 13:32:58.000000000","message":"Topic set to tox-user-extras","accounts_in_message":[],"_revision_number":1},{"id":"3348579f9f7ce645d91e190c45b82d0c1563f004","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 13:37:27.000000000","message":"Topic tox-user-extras removed","accounts_in_message":[],"_revision_number":1},{"id":"66951a123d484ddca39723037a79d8571f6af55b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 13:37:33.000000000","message":"Topic set to tox-use-extras","accounts_in_message":[],"_revision_number":1},{"id":"05f86838e82e4e96f118fcf9b0428396a1949dce","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-23 22:11:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db96e71f6146457e849e05bac739ed011e191543","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-23 22:11:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"972d31001e0566ea5b56cce61b86b479dcc851b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-23 23:35:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba1e7e8e82a64522ac6ac223c4cf570a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ef3ba0df5f13467da54608ab08d8c740 : SUCCESS in 1m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6493755d43584ee982a822f218293ecb : SUCCESS in 1h 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5fc18cca0ba48198c3bd4668e423b88 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b4f721707ff42329f8f8070b9db7c2f : SUCCESS in 8m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f5ddb598a9f4553b95683dddeb7c26d : SUCCESS in 9m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d8544047aa0e4309a8c815170aa82244 : SUCCESS in 13m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22aba36461a54ad09bfd8c1ab68f4fcf : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b59bac2c6019406491da91034e606602 : SUCCESS in 4m 10s\n- taskflow-functional-zookeeper https://zuul.opendev.org/t/openstack/build/e902a6fe3c3a4d34bbe7c53f016699ab : SUCCESS in 6m 42s\n- taskflow-functional-redis https://zuul.opendev.org/t/openstack/build/be3c1bf360434ffda861ff369b8e33db : SUCCESS in 6m 25s\n- taskflow-functional-etcd https://zuul.opendev.org/t/openstack/build/a7609a0ed4ab47ec9f3b78da90ac4ce7 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ee700e08713233171a320527d1c37592873c578d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-23 23:35:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"397603974ce0467a860cb2b55a3ec555f06c4e2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-23 23:37:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18c1b545c0e640ff8c81cc5c7266e526\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2192b414cb1b47239062b82c8d307c16 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80395caeea88453db62f0d6651458131 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2be4fa1a7742cecd5a864c931aadd8f99cba839c","revisions":{"2be4fa1a7742cecd5a864c931aadd8f99cba839c":{"kind":"REWORK","_number":1,"created":"2026-07-12 10:51:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/996934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/taskflow","ref":"refs/changes/34/996934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/34/996934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/34/996934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/taskflow refs/changes/34/996934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/taskflow refs/changes/34/996934/1"}}},"commit":{"parents":[{"commit":"ed199a15a8024e944bb627b45a0bb8310f60d24b","subject":"zuul: Use openstack-python3-next-jobs template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/ed199a15a8024e944bb627b45a0bb8310f60d24b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 10:50:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 10:50:54.000000000","tz":540},"subject":"Remove duplicated optional dependencies from test requirements","message":"Remove duplicated optional dependencies from test requirements\n\n... and use extras to define these, instead.\n\nChange-Id: I3f238ad98044e164234f1a34b36ec0d333362a4a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/2be4fa1a7742cecd5a864c931aadd8f99cba839c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/taskflow/commit/2be4fa1a7742cecd5a864c931aadd8f99cba839c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
