)]}'
{"id":"openstack%2Ftelemetry-specs~845485","triplet_id":"openstack%2Ftelemetry-specs~master~Ia098c1a0acbfa66ec190b0774742dcd03c948ea0","project":"openstack/telemetry-specs","branch":"master","topic":"prometheus-spec","attention_set":{},"removed_from_attention_set":{"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2022-12-12 13:34:02.000000000","reason":"Change was abandoned"},"5241":{"account":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"last_update":"2022-11-23 09:48:41.000000000","reason":"\u003cGERRIT_ACCOUNT_5241\u003e replied on the change","reason_account":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"}},"30893":{"account":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"last_update":"2022-12-12 13:34:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia098c1a0acbfa66ec190b0774742dcd03c948ea0","subject":"Prometheus exporter spec","status":"ABANDONED","created":"2022-06-10 21:36:00.000000000","updated":"2022-12-12 13:34:02.000000000","total_comment_count":21,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"68f31d93eada0f5608f38e4ced017c916e6038d3","_number":845485,"virtual_id_number":845485,"owner":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-27 12:37:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-06-28 09:45:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-10 21:45:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-13 09:41:40.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2022-06-13 16:08:17.000000000","updated_by":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"reviewer":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"state":"REVIEWER"}],"messages":[{"id":"cd4128a9026f63c506ff750d8edf5611b282f326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2022-06-10 21:36:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f632c8c214aeb2ce6e7b9ba5c42632b43b919be4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 21:45:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/320a002575b941979753f7d19201d2c0 : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a52859785f74498ac4d14561d58204939dc48195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2022-06-10 21:49:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e2ac2e403c49ff81e556f0b21f9cbe4e2828a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 21:53:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ca41a71db2f4fa98b9b8488cf109483 : FAILURE in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"1665867ec407db7fac38a6ad59f906a94dc4ec69","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-06-13 09:41:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9982c084d218ec2381a9079d5bd0a604206d764f","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2022-06-13 11:18:46.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"52adec8591767fe72dfee7e006ea64ddc269a25b","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2022-06-13 16:08:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d889108505acfbadc7532e011121f2e438249456","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2022-06-13 16:08:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6b7cc37a935e54a90573a3523d37ac095e06aac0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2022-06-27 11:23:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0f15c533f4570760099c52eac56b217b39c68665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 11:32:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4aed8e726dd4bc79726368c0be1a4dc : FAILURE in 3m 17s","accounts_in_message":[],"_revision_number":3},{"id":"a3241f94e81829866d4c740ec3417450edee92c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-06-27 12:25:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5759dfe4b030c887296a9c9eb28925bdad13d6dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 12:37:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74acaa2b23b4d238933e485c1c85afa : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":4},{"id":"6ac1fb5184d087ccf8a3db95084d3d72ba4a8511","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-06-28 09:45:19.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"52dc8506af0220d5070b27b3d92ed672f76e785b","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2022-10-05 14:54:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8e47e0a200f487f3965f4a7b4a26cc9590409a65","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-10-05 15:05:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"849a99175b568198a1c5a29118ba863c4ddd1cbf","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2022-11-23 09:48:41.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"68f31d93eada0f5608f38e4ced017c916e6038d3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2022-12-12 13:34:02.000000000","message":"Abandoned\n\nAbandoning for the reason noted in my last comment.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6d633d447b4cb64ffdbfb258ba940bd7cc3f17bc","revisions":{"875e45fb8d32fa6276fb82da59a4580e4c959a81":{"kind":"REWORK","_number":1,"created":"2022-06-10 21:36:00.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/85/845485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-specs","ref":"refs/changes/85/845485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/1"}}},"commit":{"parents":[{"commit":"039cfd48541220867ed7390b22e18a68f5ca938e","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-specs/commit/039cfd48541220867ed7390b22e18a68f5ca938e"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2022-06-10 21:35:17.000000000","tz":120},"committer":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2022-06-10 21:35:55.000000000","tz":120},"subject":"Prometheus exporter spec","message":"Prometheus exporter spec\n\nChange-Id: Ia098c1a0acbfa66ec190b0774742dcd03c948ea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-specs/commit/875e45fb8d32fa6276fb82da59a4580e4c959a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-specs/commit/875e45fb8d32fa6276fb82da59a4580e4c959a81"}]},"branch":"refs/heads/master"},"ed5f0fc276bbb1a132d5b152e0ff96fe4d8d00a0":{"kind":"REWORK","_number":2,"created":"2022-06-10 21:49:26.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/85/845485/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-specs","ref":"refs/changes/85/845485/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/2"}}},"commit":{"parents":[{"commit":"039cfd48541220867ed7390b22e18a68f5ca938e","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-specs/commit/039cfd48541220867ed7390b22e18a68f5ca938e"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2022-06-10 21:35:17.000000000","tz":120},"committer":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2022-06-10 21:49:03.000000000","tz":120},"subject":"Prometheus exporter spec","message":"Prometheus exporter spec\n\nChange-Id: Ia098c1a0acbfa66ec190b0774742dcd03c948ea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-specs/commit/ed5f0fc276bbb1a132d5b152e0ff96fe4d8d00a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-specs/commit/ed5f0fc276bbb1a132d5b152e0ff96fe4d8d00a0"}]},"branch":"refs/heads/master"},"2d0432f2ed0af10265ec5fb462eb22c4a1958f41":{"kind":"REWORK","_number":3,"created":"2022-06-27 11:23:33.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/85/845485/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-specs","ref":"refs/changes/85/845485/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/3"}}},"commit":{"parents":[{"commit":"039cfd48541220867ed7390b22e18a68f5ca938e","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-specs/commit/039cfd48541220867ed7390b22e18a68f5ca938e"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2022-06-10 21:35:17.000000000","tz":120},"committer":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2022-06-27 11:22:01.000000000","tz":120},"subject":"Prometheus exporter spec","message":"Prometheus exporter spec\n\nChange-Id: Ia098c1a0acbfa66ec190b0774742dcd03c948ea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-specs/commit/2d0432f2ed0af10265ec5fb462eb22c4a1958f41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-specs/commit/2d0432f2ed0af10265ec5fb462eb22c4a1958f41"}]},"branch":"refs/heads/master"},"6d633d447b4cb64ffdbfb258ba940bd7cc3f17bc":{"kind":"REWORK","_number":4,"created":"2022-06-27 12:25:51.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/85/845485/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-specs","ref":"refs/changes/85/845485/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-specs refs/changes/85/845485/4"}}},"commit":{"parents":[{"commit":"039cfd48541220867ed7390b22e18a68f5ca938e","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-specs/commit/039cfd48541220867ed7390b22e18a68f5ca938e"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2022-06-10 21:35:17.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2022-06-27 12:25:04.000000000","tz":120},"subject":"Prometheus exporter spec","message":"Prometheus exporter spec\n\nChange-Id: Ia098c1a0acbfa66ec190b0774742dcd03c948ea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-specs/commit/6d633d447b4cb64ffdbfb258ba940bd7cc3f17bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-specs/commit/6d633d447b4cb64ffdbfb258ba940bd7cc3f17bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
