)]}'
{"id":"openstack%2Ftelemetry-tempest-plugin~1007451","triplet_id":"openstack%2Ftelemetry-tempest-plugin~master~Ibdd7bed3e8375fc53c69e46cfede974e75951ebb","project":"openstack/telemetry-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ibdd7bed3e8375fc53c69e46cfede974e75951ebb","subject":"Generalize mTLS options","status":"ABANDONED","created":"2026-09-25 10:12:15.000000000","updated":"2026-10-05 10:23:33.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c426b7175ea7d8a808693355945553c7c852f418","_number":1007451,"virtual_id_number":1007451,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-25 11:20:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2026-09-25 10:12:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 11:20:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23d87d18dcc30fa51411945616def1a50ef7468a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-09-25 10:12:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db5eeba6c87b9804e66ca2e5abdfa5fdcf0c77ff","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-09-25 10:12:45.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"24397895c0735f4c1e63f191cd0a0df5c63cb7b6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-09-25 10:12:49.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"1e51eeceea2268a38e63370b3702762f182a3a43","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-09-25 10:12:54.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"011c6b899301c404152ad6179ee33adea3b65917","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-09-25 10:14:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9aaa0adaabcfe990006c45310ceed7a872645b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:20:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23dfb8f7aa6e4baa804adc8c95833142\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b47b79fd3042454cbb7e56d13eee1a1d : SUCCESS in 3m 02s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/36ddf7593d1c49c2bc7062aa2769b1e6 : SUCCESS in 57m 21s\n- telemetry-dsvm-integration-2025-1 https://zuul.opendev.org/t/openstack/build/cbf0976332e64c3ebfa3152788d67cac : SUCCESS in 59m 57s\n- telemetry-dsvm-integration-2025-2 https://zuul.opendev.org/t/openstack/build/435d039c02d14a5abe13d2080c23a527 : SUCCESS in 51m 38s\n- telemetry-dsvm-integration-2026-1 https://zuul.opendev.org/t/openstack/build/b41b8e8aca6144f198dd5ed91ec783c9 : SUCCESS in 33m 47s\n- telemetry-dsvm-integration-2026-2 https://zuul.opendev.org/t/openstack/build/28d16874f86046239759232a82c8ab4b : SUCCESS in 1h 03m 52s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/0ed2945879ad42539c80f9ce08cc1d69 : SUCCESS in 54m 20s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/630b5794455a42f7aaea71564804bd02 : SUCCESS in 58m 44s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/a5e622c14b0d40778d85f7db47304840 : SUCCESS in 1h 02m 19s (non-voting)\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/7868f3c168d7448680db99a07f6bd356 : SUCCESS in 47m 36s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/afb32e5b034c49a49d0b3b818b95a971 : SUCCESS in 41m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"c426b7175ea7d8a808693355945553c7c852f418","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-10-05 10:23:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a34ec38a110e1be4ebc20b0ef1ee27e9db4d40b","revisions":{"9a34ec38a110e1be4ebc20b0ef1ee27e9db4d40b":{"kind":"REWORK","_number":1,"created":"2026-09-25 10:12:15.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/51/1007451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/51/1007451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/51/1007451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/51/1007451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/51/1007451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/51/1007451/1"}}},"commit":{"parents":[{"commit":"0037426bafafd4b6cf0926f17ae601620643b887","subject":"Add ability to test sg-core with mTLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/0037426bafafd4b6cf0926f17ae601620643b887"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-09-25 10:11:10.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-09-25 10:12:09.000000000","tz":-240},"subject":"Generalize mTLS options","message":"Generalize mTLS options\n\nGeneralize the recently introduced sg-core mTLS config options to be\nuseful for all services.\n\nChange-Id: Ibdd7bed3e8375fc53c69e46cfede974e75951ebb\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/9a34ec38a110e1be4ebc20b0ef1ee27e9db4d40b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/9a34ec38a110e1be4ebc20b0ef1ee27e9db4d40b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
