)]}'
{"id":"openstack%2Ftelemetry-tempest-plugin~865386","triplet_id":"openstack%2Ftelemetry-tempest-plugin~master~Iaa61eebf0e652b60182f893bba009b30d4734f95","project":"openstack/telemetry-tempest-plugin","branch":"master","topic":"autoscaling","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2022-12-19 19:08:20.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2022-11-24 08:14:33.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}},"hashtags":[],"change_id":"Iaa61eebf0e652b60182f893bba009b30d4734f95","subject":"Disable SSL validation in gabbi tests","status":"MERGED","created":"2022-11-23 10:24:57.000000000","updated":"2022-12-19 19:08:20.000000000","submitted":"2022-12-19 19:08:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865386-autoscaling","meta_rev_id":"18378d10b17c6d6fdcb472e81d1042f8a3859136","_number":865386,"virtual_id_number":865386,"owner":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-19 19:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-11-24 08:14:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-12-19 18:28:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-23 12:37:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 16:41:24.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"df73115170374f09317b88ef907e72957751b7ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-23 10:24:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25fac63072e243466f5513b7afe1a52a009181a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-23 11:31:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"155b3c60a98db45c1311fb47d80b5c73fe9554a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 12:37:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/341aff0f4dc84c1b996565f47e5bb99f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a377fc0fed4f338d925e214120f9d6 : SUCCESS in 2m 39s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/cb3719b196474167bc807b27cccae461 : SUCCESS in 50m 16s\n- telemetry-dsvm-integration-wallaby https://zuul.opendev.org/t/openstack/build/262ecd75b0ca4aa1b6a8a92118e27e8c : FAILURE in 23m 27s (non-voting)\n- telemetry-dsvm-integration-train https://zuul.opendev.org/t/openstack/build/40357cbef99a4b8787f493209933b79d : SUCCESS in 46m 21s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/797fa7ee796c4cca8a1367d842554488 : SUCCESS in 51m 23s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/bdc5fc9a37f5494eafd18ba9ab849fa2 : SUCCESS in 55m 12s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/db92639c5a6e4a749a1b59226ceb27e8 : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":2},{"id":"9ca759eca58b2086a4ac5360d54ed6b1968307e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-23 12:41:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4943e34c19d6387d49712270b40402a2704f1658","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 14:22:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36520b41247c4d8288f9d7083df7e379\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84762a6b15c340c49988dea03bda11a5 : SUCCESS in 3m 08s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/14a45ac13440483ea0b8f9ee9c835c2e : SUCCESS in 49m 48s\n- telemetry-dsvm-integration-wallaby https://zuul.opendev.org/t/openstack/build/043a8422cc2c4dc9bf637da861ae8050 : FAILURE in 22m 45s (non-voting)\n- telemetry-dsvm-integration-train https://zuul.opendev.org/t/openstack/build/501c62a4b97643db9d7169e9c6cda13e : SUCCESS in 43m 30s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/1afc57001e884f8398d54f26a3c98ea5 : SUCCESS in 45m 45s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/e7cecbab73b14dae917419b6820489bb : SUCCESS in 1h 40m 22s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/bfdcaf6e5d00450ca64474911a3cd92c : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":3},{"id":"6c047b0aa8c4b8d2e3f600d628e3196b9cc6d4b2","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-23 16:41:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4ad43813beaf248d3a7554f64a2921eb479cafb1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-11-24 08:14:33.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"117a454d93aae2296bcb877f9a226cff6bf91019","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-12-19 15:31:09.000000000","message":"Topic set to autoscaling","accounts_in_message":[],"_revision_number":3},{"id":"950deebe0f2535c957f5d70d2b288d47bf69836a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-12-19 18:28:11.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c3313b09b3f1934ba88786f7a6f237e50b7cfa38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 18:28:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"88be822ebccd19222e75d5baf0adfecdda4a0f45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 19:08:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fefbd30a376743cebd943b90bb988bb5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50e017c69da84eed9eda6693eaf54678 : SUCCESS in 2m 26s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/acdf204fb5b94678a4a72d23d6e0a807 : SUCCESS in 33m 07s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/cf67beeed48745ca86b059cac26edda6 : SUCCESS in 38m 43s","accounts_in_message":[],"_revision_number":3},{"id":"18378d10b17c6d6fdcb472e81d1042f8a3859136","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 19:08:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eff7af3610bccdd7694033360938614d617f72ec","revisions":{"2277e23d326bdd80a76afa1d1e3fe8bf283a59e5":{"kind":"REWORK","_number":1,"created":"2022-11-23 10:24:57.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/86/865386/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/86/865386/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/86/865386/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/86/865386/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/86/865386/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/86/865386/1"}}},"commit":{"parents":[{"commit":"885adad3ffc6d195ae41c438ef30835a76c6cee0","subject":"Merge \"remove unicode from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/885adad3ffc6d195ae41c438ef30835a76c6cee0"}]}],"author":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-11-23 10:21:13.000000000","tz":-300},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-11-23 10:21:13.000000000","tz":-300},"subject":"Disable SSL validation in gabbi tests","message":"Disable SSL validation in gabbi tests\n\nDisable SSL validation to prevent autoscaling tests\nfrom failing with error\n\"certificate verify failed: unable to get local issuer certificate\"\n\nSet parameter \"disable_ssl_certificate_validation\" to True\nwhich was introduced in 959fc69e492ed1d2297cce624c1efbf79e9f1b66\n\nSigned-off-by: Yadnesh Kulkarni \u003cykulkarn@redhat.com\u003e\nChange-Id: Iaa61eebf0e652b60182f893bba009b30d4734f95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/2277e23d326bdd80a76afa1d1e3fe8bf283a59e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/2277e23d326bdd80a76afa1d1e3fe8bf283a59e5"}]},"branch":"refs/heads/master"},"850937ee306710c406f73f6d038d407383a4b420":{"kind":"REWORK","_number":2,"created":"2022-11-23 11:31:44.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/86/865386/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/86/865386/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/86/865386/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/86/865386/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/86/865386/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/86/865386/2"}}},"commit":{"parents":[{"commit":"885adad3ffc6d195ae41c438ef30835a76c6cee0","subject":"Merge \"remove unicode from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/885adad3ffc6d195ae41c438ef30835a76c6cee0"}]}],"author":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-11-23 10:21:13.000000000","tz":-300},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-11-23 11:29:43.000000000","tz":-300},"subject":"Disable SSL validation in gabbi tests","message":"Disable SSL validation in gabbi tests\n\nDisable SSL validation to prevent autoscaling tests\nfrom failing with error\n\"certificate verify failed: unable to get local issuer certificate\"\n\nSet parameter \"disable_ssl_certificate_validation\" to True\nwhich was introduced in 959fc69e492ed1d2297cce624c1efbf79e9f1b66\n\nSigned-off-by: Yadnesh Kulkarni \u003cykulkarn@redhat.com\u003e\nChange-Id: Iaa61eebf0e652b60182f893bba009b30d4734f95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/850937ee306710c406f73f6d038d407383a4b420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/850937ee306710c406f73f6d038d407383a4b420"}]},"branch":"refs/heads/master"},"eff7af3610bccdd7694033360938614d617f72ec":{"kind":"REWORK","_number":3,"created":"2022-11-23 12:41:14.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/86/865386/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/86/865386/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/86/865386/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/86/865386/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/86/865386/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/86/865386/3"}}},"commit":{"parents":[{"commit":"885adad3ffc6d195ae41c438ef30835a76c6cee0","subject":"Merge \"remove unicode from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/885adad3ffc6d195ae41c438ef30835a76c6cee0"}]}],"author":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-11-23 10:21:13.000000000","tz":-300},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-11-23 12:39:05.000000000","tz":-300},"subject":"Disable SSL validation in gabbi tests","message":"Disable SSL validation in gabbi tests\n\nDisable SSL validation to prevent autoscaling tests\nfrom failing with error\n\"certificate verify failed: unable to get local issuer certificate\"\n\nSet parameter \"disable_ssl_certificate_validation\" to True\nwhich was introduced in 959fc69e492ed1d2297cce624c1efbf79e9f1b66\n\nSigned-off-by: Yadnesh Kulkarni \u003cykulkarn@redhat.com\u003e\nChange-Id: Iaa61eebf0e652b60182f893bba009b30d4734f95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/eff7af3610bccdd7694033360938614d617f72ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/eff7af3610bccdd7694033360938614d617f72ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
