)]}'
{"id":"openstack%2Ftelemetry-tempest-plugin~992918","triplet_id":"openstack%2Ftelemetry-tempest-plugin~master~Ie374d10a10bd99df2c34161599c157cebc6ff70b","project":"openstack/telemetry-tempest-plugin","branch":"master","topic":"ceilometer_use_openstack_sdk","attention_set":{"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2026-06-11 15:26:40.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie374d10a10bd99df2c34161599c157cebc6ff70b","subject":"[WIP] Add Ceilometer image metric tests","status":"NEW","created":"2026-06-11 14:42:17.000000000","updated":"2026-08-05 18:25:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"051a9b2c515403f7433bcc80e5032b449b76228a","_number":992918,"virtual_id_number":992918,"owner":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-05 18:25:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-11 15:26:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55f3147654dde09f7f10ef8aeef51700e38f1d4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-06-11 14:42:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aeed7dc1ba8b5bceb14fdf5b70a4661173196d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 15:26:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f07b9b8031d47abb15886dcf0f48243\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de8ce28ae7b4c8fb3873c6aba467d01 : SUCCESS in 2m 03s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4c60f7981d284583a5e21d39f3a1be1c : FAILURE in 37m 23s\n- telemetry-tempest-metric-verification https://zuul.opendev.org/t/openstack/build/7658d307d29d486586804b7c1b03e3a0 : FAILURE in 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73fe1b13ec6f6ec3f381f13a74196feb30815ae7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-03 17:56:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"68792bf1e1519f3d5f443ae52db78b12178e9cb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 19:16:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36b84d44febf45a38bc710849f509392\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a7f0091075147288effc039d62f2d0c : SUCCESS in 2m 26s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/67bcfc6efb9f4eb584ff62608f802240 : FAILURE in 52m 21s\n- telemetry-tempest-metric-verification https://zuul.opendev.org/t/openstack/build/9fa7836707cb49aa9afa3511ee072286 : FAILURE in 1h 09m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4671c7e16029393fd87961b706b80454b886a1ed","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-06 11:28:18.000000000","message":"Topic set to ceilometer_use_openstack_sdk","accounts_in_message":[],"_revision_number":2},{"id":"5084719ad9a5b7d04f8e988c49c30da302507def","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-06 18:57:50.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b9e3fd131da0bbbfa30cb8f38db1b0c8dee7e63f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 20:13:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13b4c4e34eec4ba3b125d266735a947a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b5f995f2cd64b4997c5b1b1a47f0967 : SUCCESS in 2m 22s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2bb5a6bd5aa4439b8791d90ae514b4de : FAILURE in 1h 01m 38s\n- telemetry-tempest-metric-verification https://zuul.opendev.org/t/openstack/build/e3f2400c3ba64a1083147393d2a50640 : FAILURE in 1h 05m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba2f5fe767b434c958e922d395adabdcc09f03ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-14 13:28:09.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"d18653e57b4e6525b6f1cf51090b8bed39077f7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 14:36:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/797d028792ce49e5aecbed3fe2a494c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a7bfa28a474a1e8100a617a58deabe : SUCCESS in 3m 11s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/43922c7c1a784b7cb2f9cadf2cc807c5 : SUCCESS in 54m 53s\n- telemetry-tempest-metric-verification https://zuul.opendev.org/t/openstack/build/6b1e581751ab41cda20fd8141754f0e6 : SUCCESS in 50m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b77c4b7840fa4129222010e399442d02da6ac674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-27 21:03:01.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"8c441a1b812ff2bd393203f43640b8e5a6530696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 22:14:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a330bf40fc034c2f8ee871677c7735fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/831893ec72874e958b55697aced7f923 : SUCCESS in 3m 20s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e5cd08d4d72449c28439ab29fcd0ddcf : SUCCESS in 53m 48s\n- telemetry-tempest-metric-verification https://zuul.opendev.org/t/openstack/build/6f1d1d52331a4d88bb05d24c0837d1db : SUCCESS in 1h 01m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d8bcd7ac14fad7e692425a4ae407ebcb53679932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-08-05 17:14:30.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"051a9b2c515403f7433bcc80e5032b449b76228a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 18:25:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c75dda49b40e4f1888651f8d281543c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/768f40ad6e784fd5b1cf8729e3c7d519 : SUCCESS in 3m 15s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bdb28efd57b14eb3abec83383e83b481 : SUCCESS in 47m 14s\n- telemetry-tempest-metric-verification https://zuul.opendev.org/t/openstack/build/5699835a1e9b4c4eb2a8e0f8638528f4 : SUCCESS in 53m 09s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"856280479372a9cb245476be218cf1a1d1e05b95","revisions":{"9c364b9b02dbb76a03e3d583646aa0ca43a50601":{"kind":"REWORK","_number":1,"created":"2026-06-11 14:42:17.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/18/992918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/18/992918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/1"}}},"commit":{"parents":[{"commit":"26dcf9b1857ff619abe8b5742037a1b69c8c5995","subject":"[WIP] Add Ceilometer compute metric tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/26dcf9b1857ff619abe8b5742037a1b69c8c5995"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-05-12 15:59:25.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-06-11 13:27:41.000000000","tz":-240},"subject":"[WIP] Add Ceilometer image metric tests","message":"[WIP] Add Ceilometer image metric tests\n\nTests Glance image size metrics.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ie374d10a10bd99df2c34161599c157cebc6ff70b\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/9c364b9b02dbb76a03e3d583646aa0ca43a50601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/9c364b9b02dbb76a03e3d583646aa0ca43a50601"}]},"branch":"refs/heads/master"},"a976a26584c8b430cf1f9bee6430fc8f984f9757":{"kind":"REWORK","_number":2,"created":"2026-07-03 17:56:19.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/18/992918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/18/992918/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/2"}}},"commit":{"parents":[{"commit":"7081c185990db9a0d8bc9fe3c31abdf4bfbf43d5","subject":"[WIP] Add Ceilometer compute metric tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/7081c185990db9a0d8bc9fe3c31abdf4bfbf43d5"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-05-12 15:59:25.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-03 17:54:51.000000000","tz":-240},"subject":"[WIP] Add Ceilometer image metric tests","message":"[WIP] Add Ceilometer image metric tests\n\nTests Glance image size metrics.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ie374d10a10bd99df2c34161599c157cebc6ff70b\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/a976a26584c8b430cf1f9bee6430fc8f984f9757"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/a976a26584c8b430cf1f9bee6430fc8f984f9757"}]},"branch":"refs/heads/master"},"352d7f604308bbfd84b5fff4fa923c48a6256b14":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-07-06 18:57:50.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/18/992918/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/18/992918/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/3"}}},"commit":{"parents":[{"commit":"51530c35f2b3171e1b0d76210bb0fae36006fc3b","subject":"[WIP] Add Ceilometer compute metric tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/51530c35f2b3171e1b0d76210bb0fae36006fc3b"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-05-12 15:59:25.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-06 18:57:35.000000000","tz":-240},"subject":"[WIP] Add Ceilometer image metric tests","message":"[WIP] Add Ceilometer image metric tests\n\nTests Glance image size metrics.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ie374d10a10bd99df2c34161599c157cebc6ff70b\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/352d7f604308bbfd84b5fff4fa923c48a6256b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/352d7f604308bbfd84b5fff4fa923c48a6256b14"}]},"branch":"refs/heads/master"},"0d665b16148c867748eb7838619a431c93de8449":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-07-14 13:28:09.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/18/992918/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/18/992918/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/4"}}},"commit":{"parents":[{"commit":"fb4270a3d030636ae05ea0f1fc10fd9ba2c7d7a9","subject":"[WIP] Add Ceilometer compute metric tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/fb4270a3d030636ae05ea0f1fc10fd9ba2c7d7a9"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-05-12 15:59:25.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-14 13:24:00.000000000","tz":-240},"subject":"[WIP] Add Ceilometer image metric tests","message":"[WIP] Add Ceilometer image metric tests\n\nTests Glance image size metrics.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ie374d10a10bd99df2c34161599c157cebc6ff70b\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/0d665b16148c867748eb7838619a431c93de8449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/0d665b16148c867748eb7838619a431c93de8449"}]},"branch":"refs/heads/master"},"169ad35d701265b2c6a6606a9ec9b8f2bcab645e":{"kind":"NO_CHANGE","_number":5,"created":"2026-07-27 21:03:01.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/18/992918/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/18/992918/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/5"}}},"commit":{"parents":[{"commit":"ce46f965d122a73335afb4a98f5794ebebef8cc3","subject":"[WIP] Add Ceilometer compute metric tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/ce46f965d122a73335afb4a98f5794ebebef8cc3"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-05-12 15:59:25.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-27 21:02:10.000000000","tz":-240},"subject":"[WIP] Add Ceilometer image metric tests","message":"[WIP] Add Ceilometer image metric tests\n\nTests Glance image size metrics.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ie374d10a10bd99df2c34161599c157cebc6ff70b\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/169ad35d701265b2c6a6606a9ec9b8f2bcab645e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/169ad35d701265b2c6a6606a9ec9b8f2bcab645e"}]},"branch":"refs/heads/master"},"856280479372a9cb245476be218cf1a1d1e05b95":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-08-05 17:14:30.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/18/992918/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/18/992918/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/18/992918/6"}}},"commit":{"parents":[{"commit":"b1b0d49370e2b0ab2d4dcf0626b2628610388682","subject":"[WIP] Add Ceilometer compute metric tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/b1b0d49370e2b0ab2d4dcf0626b2628610388682"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-05-12 15:59:25.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-08-05 17:13:39.000000000","tz":-240},"subject":"[WIP] Add Ceilometer image metric tests","message":"[WIP] Add Ceilometer image metric tests\n\nTests Glance image size metrics.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ie374d10a10bd99df2c34161599c157cebc6ff70b\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/856280479372a9cb245476be218cf1a1d1e05b95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/856280479372a9cb245476be218cf1a1d1e05b95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
