)]}'
{"id":"openstack%2Ftelemetry-tempest-plugin~995995","triplet_id":"openstack%2Ftelemetry-tempest-plugin~master~Ia5253ee865b14cc1087eae28226a9919aa5667c0","project":"openstack/telemetry-tempest-plugin","branch":"master","topic":"ceilometer_use_openstack_sdk","attention_set":{"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2026-07-03 19:28:36.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia5253ee865b14cc1087eae28226a9919aa5667c0","subject":"[WIP] Add Ceilometer VPN metric tests","status":"NEW","created":"2026-07-03 17:56:19.000000000","updated":"2026-08-05 18:18:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b15f9314d63e9ebfa78276a4d5d8be5a45f5b730","_number":995995,"virtual_id_number":995995,"owner":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-05 18:18:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-03 19:28:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a19783f6743f32b98031ef76c2f3cc6375c7552","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-03 17:56:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab09a6d66e5aedf5dc209ffe66398d5063269191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 19:28:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8613e699d5f84e9a98b974110cee56ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac409496843141b4b543edf5944767b0 : SUCCESS in 4m 54s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/51d030aaefe6499cb592084f6356c594 : FAILURE in 1h 15m 22s\n- telemetry-tempest-metric-verification https://zuul.opendev.org/t/openstack/build/5cf52ced1969421cacd8de9c4e464fdc : FAILURE in 1h 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aab57109ceecd16c41cfbae5d33c69a84b832ed3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-06 11:24:10.000000000","message":"Topic set to ceilometer_use_openstack_sdk","accounts_in_message":[],"_revision_number":1},{"id":"ea6a7e586ca5d2efb226dc24f53db63b4ba7f93d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-06 18:57:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d11dba5a7b14715a4cd23c624f0f893e4b1feb33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 20:22:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48fd155ee2bf4be28588f81b6dc02146\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/336b167dd005450a9735012b1bf6db1c : SUCCESS in 5m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8771b8091fa042be96a8cfc481dbba11 : FAILURE in 1h 09m 59s\n- telemetry-tempest-metric-verification https://zuul.opendev.org/t/openstack/build/1a32623e37d04f41be9e0c53a5712f70 : FAILURE in 1h 01m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"766df19eb3c58b078acdfe798e5c4e1bfb8faa1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-14 13:28:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"2177dc1ecd5d4d15050b5cb355d685c4819bc465","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 14:41:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6b4e7949e6546969241162af20f5670\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3a4aa690b0c4ada987dd09edb55a296 : SUCCESS in 3m 25s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/59b3a31c6b214b45a82a1ab7d9090794 : FAILURE in 59m 58s\n- telemetry-tempest-metric-verification https://zuul.opendev.org/t/openstack/build/275749be32fa4d4c8222c9536c13af0a : FAILURE in 39m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"51a8d66a8fc2c9672d63e6629f84b39e102668c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-27 21:03:01.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"a33686d1d16e22e0ce9d809ad18d6159608ba4ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 22:32:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b733ae1bc37c4928925d8362108126f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a210fdaa524768b6987550569b23a2 : SUCCESS in 3m 07s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/626f377ef1c14b3098a67d7c901fe4d2 : FAILURE in 42m 11s\n- telemetry-tempest-metric-verification https://zuul.opendev.org/t/openstack/build/8542c00c06dc46a8b362658e7cee32db : SUCCESS in 1h 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"83dc614a4df8551e7360027af1e1e2d569a53d9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-08-05 17:14:30.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"b15f9314d63e9ebfa78276a4d5d8be5a45f5b730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 18:18:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96b39d8794a048928f60aa2a83e8ff8a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c8f6f2e6719403fadf9a41ea19b1c85 : SUCCESS in 3m 28s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/60f42b84a60e43cc8f67fd8ddc5c5152 : FAILURE in 39m 02s\n- telemetry-tempest-metric-verification https://zuul.opendev.org/t/openstack/build/1b97c427bd334a5f837136869a8b0135 : SUCCESS in 45m 07s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8e19ff70c0d7b48037b3df8ce082f44c03734847","revisions":{"85caec2ac3a78400321825ef60c4ed0776892bc9":{"kind":"REWORK","_number":1,"created":"2026-07-03 17:56:19.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/95/995995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/95/995995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/1"}}},"commit":{"parents":[{"commit":"bd9c25547190850a523afb033739ab8f51c9dbf1","subject":"[WIP] Add Ceilometer firewall metric tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/bd9c25547190850a523afb033739ab8f51c9dbf1"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-05-12 16:06:49.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-03 17:54:51.000000000","tz":-240},"subject":"[WIP] Add Ceilometer VPN metric tests","message":"[WIP] Add Ceilometer VPN metric tests\n\nTests Neutron VPNaaS VPN service and IPSec connection metrics.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ia5253ee865b14cc1087eae28226a9919aa5667c0\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/85caec2ac3a78400321825ef60c4ed0776892bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/85caec2ac3a78400321825ef60c4ed0776892bc9"}]},"branch":"refs/heads/master"},"f910f32d68cb8dcaf0603f591b350b4e035ab5ae":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-06 18:57:50.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/95/995995/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/95/995995/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/2"}}},"commit":{"parents":[{"commit":"863cf261a3aba4bd3503033a28f3f9fcb4d4497b","subject":"[WIP] Add Ceilometer firewall metric tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/863cf261a3aba4bd3503033a28f3f9fcb4d4497b"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-05-12 16:06:49.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-06 18:57:36.000000000","tz":-240},"subject":"[WIP] Add Ceilometer VPN metric tests","message":"[WIP] Add Ceilometer VPN metric tests\n\nTests Neutron VPNaaS VPN service and IPSec connection metrics.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ia5253ee865b14cc1087eae28226a9919aa5667c0\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/f910f32d68cb8dcaf0603f591b350b4e035ab5ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/f910f32d68cb8dcaf0603f591b350b4e035ab5ae"}]},"branch":"refs/heads/master"},"41ede86e8feec10485b4a2dd97098eb994dbc6ba":{"kind":"REWORK","_number":3,"created":"2026-07-14 13:28:09.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/95/995995/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/95/995995/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/3"}}},"commit":{"parents":[{"commit":"736e0964cb2ccf901df00321387a1c46427d4bfb","subject":"[WIP] Add Ceilometer firewall metric tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/736e0964cb2ccf901df00321387a1c46427d4bfb"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-05-12 16:06:49.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-14 13:24:01.000000000","tz":-240},"subject":"[WIP] Add Ceilometer VPN metric tests","message":"[WIP] Add Ceilometer VPN metric tests\n\nTests Neutron VPNaaS VPN service and IPSec connection metrics.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ia5253ee865b14cc1087eae28226a9919aa5667c0\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/41ede86e8feec10485b4a2dd97098eb994dbc6ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/41ede86e8feec10485b4a2dd97098eb994dbc6ba"}]},"branch":"refs/heads/master"},"0294c1ad65480d8f7fe600a1c0300b3a5c6d8715":{"kind":"NO_CHANGE","_number":4,"created":"2026-07-27 21:03:01.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/95/995995/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/95/995995/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/4"}}},"commit":{"parents":[{"commit":"4fd212bbe53b37483a7b211ca4aaa6406f1bcc9c","subject":"[WIP] Add Ceilometer firewall metric tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/4fd212bbe53b37483a7b211ca4aaa6406f1bcc9c"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-05-12 16:06:49.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-27 21:02:10.000000000","tz":-240},"subject":"[WIP] Add Ceilometer VPN metric tests","message":"[WIP] Add Ceilometer VPN metric tests\n\nTests Neutron VPNaaS VPN service and IPSec connection metrics.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ia5253ee865b14cc1087eae28226a9919aa5667c0\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/0294c1ad65480d8f7fe600a1c0300b3a5c6d8715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/0294c1ad65480d8f7fe600a1c0300b3a5c6d8715"}]},"branch":"refs/heads/master"},"8e19ff70c0d7b48037b3df8ce082f44c03734847":{"kind":"REWORK","_number":5,"created":"2026-08-05 17:14:30.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/95/995995/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/telemetry-tempest-plugin","ref":"refs/changes/95/995995/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/telemetry-tempest-plugin refs/changes/95/995995/5"}}},"commit":{"parents":[{"commit":"5486c6c581dfb81ef39f142f925f550ebac26ce1","subject":"[WIP] Add Ceilometer firewall metric tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/5486c6c581dfb81ef39f142f925f550ebac26ce1"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-05-12 16:06:49.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-08-05 17:13:39.000000000","tz":-240},"subject":"[WIP] Add Ceilometer VPN metric tests","message":"[WIP] Add Ceilometer VPN metric tests\n\nTests Neutron VPNaaS VPN service and IPSec connection metrics.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ia5253ee865b14cc1087eae28226a9919aa5667c0\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/8e19ff70c0d7b48037b3df8ce082f44c03734847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/telemetry-tempest-plugin/commit/8e19ff70c0d7b48037b3df8ce082f44c03734847"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
