)]}'
{"id":"openstack%2Ftempest-lib~302892","triplet_id":"openstack%2Ftempest-lib~master~Ic3ed2c3d806ab8a4d62eaf1d567c585a1369fab3","project":"openstack/tempest-lib","branch":"master","topic":"openstack/requirements","hashtags":[],"change_id":"Ic3ed2c3d806ab8a4d62eaf1d567c585a1369fab3","subject":"Updated from global requirements","status":"MERGED","created":"2016-04-07 15:29:52.000000000","updated":"2016-04-08 09:31:08.000000000","submitted":"2016-04-08 09:31:07.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d926fffe07fb2806be7a67b3f0848cfc894e64be","_number":302892,"virtual_id_number":302892,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":16272,"name":"Clint Adams","email":"clint@gcfm.net","username":"clint"},{"value":0,"_account_id":20993,"name":"Cody Eilrich","email":"cody.eilrich@ericsson.com","username":"ceilrich"},{"value":0,"_account_id":18675,"name":"Sharath Rao","email":"sharath.rao@ericsson.com","username":"sharath.rao"},{"value":2,"date":"2016-04-08 09:31:07.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":20996,"name":"Avinash Raghavendra","email":"avinash.raghavendra@ericsson.com","username":"eragavi"},{"value":0,"_account_id":18265,"name":"Rony Haddad","email":"rony.haddad@ericsson.com","username":"RonyH"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20810,"name":"Kshitiz Arora","email":"kshitiz.arora@ericsson.com","username":"EKSHARO"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"recommended":{"_account_id":18265,"name":"Rony Haddad","email":"rony.haddad@ericsson.com","username":"RonyH"},"all":[{"value":1,"date":"2016-04-08 03:31:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16272,"name":"Clint Adams","email":"clint@gcfm.net","username":"clint"},{"value":1,"date":"2016-04-08 02:19:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20993,"name":"Cody Eilrich","email":"cody.eilrich@ericsson.com","username":"ceilrich"},{"value":1,"date":"2016-04-08 03:43:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18675,"name":"Sharath Rao","email":"sharath.rao@ericsson.com","username":"sharath.rao"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":20996,"name":"Avinash Raghavendra","email":"avinash.raghavendra@ericsson.com","username":"eragavi"},{"value":1,"date":"2016-04-08 03:37:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18265,"name":"Rony Haddad","email":"rony.haddad@ericsson.com","username":"RonyH"},{"value":2,"date":"2016-04-08 08:25:47.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20810,"name":"Kshitiz Arora","email":"kshitiz.arora@ericsson.com","username":"EKSHARO"},{"value":2,"date":"2016-04-08 08:28:56.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":16272,"name":"Clint Adams","email":"clint@gcfm.net","username":"clint"},{"value":0,"_account_id":20993,"name":"Cody Eilrich","email":"cody.eilrich@ericsson.com","username":"ceilrich"},{"value":0,"_account_id":18675,"name":"Sharath Rao","email":"sharath.rao@ericsson.com","username":"sharath.rao"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":20996,"name":"Avinash Raghavendra","email":"avinash.raghavendra@ericsson.com","username":"eragavi"},{"value":0,"_account_id":18265,"name":"Rony Haddad","email":"rony.haddad@ericsson.com","username":"RonyH"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20810,"name":"Kshitiz Arora","email":"kshitiz.arora@ericsson.com","username":"EKSHARO"},{"value":1,"date":"2016-04-08 08:28:56.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16272,"name":"Clint Adams","email":"clint@gcfm.net","username":"clint"},{"_account_id":18265,"name":"Rony Haddad","email":"rony.haddad@ericsson.com","username":"RonyH"},{"_account_id":18675,"name":"Sharath Rao","email":"sharath.rao@ericsson.com","username":"sharath.rao"},{"_account_id":20810,"name":"Kshitiz Arora","email":"kshitiz.arora@ericsson.com","username":"EKSHARO"},{"_account_id":20993,"name":"Cody Eilrich","email":"cody.eilrich@ericsson.com","username":"ceilrich"},{"_account_id":20996,"name":"Avinash Raghavendra","email":"avinash.raghavendra@ericsson.com","username":"eragavi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-04-07 15:30:21.000000000","updated_by":{"_account_id":20996,"name":"Avinash Raghavendra","email":"avinash.raghavendra@ericsson.com","username":"eragavi"},"reviewer":{"_account_id":20996,"name":"Avinash Raghavendra","email":"avinash.raghavendra@ericsson.com","username":"eragavi"},"state":"REVIEWER"},{"updated":"2016-04-07 15:31:41.000000000","updated_by":{"_account_id":20810,"name":"Kshitiz Arora","email":"kshitiz.arora@ericsson.com","username":"EKSHARO"},"reviewer":{"_account_id":20810,"name":"Kshitiz Arora","email":"kshitiz.arora@ericsson.com","username":"EKSHARO"},"state":"REVIEWER"},{"updated":"2016-04-08 02:19:21.000000000","updated_by":{"_account_id":20993,"name":"Cody Eilrich","email":"cody.eilrich@ericsson.com","username":"ceilrich"},"reviewer":{"_account_id":20993,"name":"Cody Eilrich","email":"cody.eilrich@ericsson.com","username":"ceilrich"},"state":"REVIEWER"},{"updated":"2016-04-08 03:31:11.000000000","updated_by":{"_account_id":16272,"name":"Clint Adams","email":"clint@gcfm.net","username":"clint"},"reviewer":{"_account_id":16272,"name":"Clint Adams","email":"clint@gcfm.net","username":"clint"},"state":"REVIEWER"},{"updated":"2016-04-08 03:37:37.000000000","updated_by":{"_account_id":18265,"name":"Rony Haddad","email":"rony.haddad@ericsson.com","username":"RonyH"},"reviewer":{"_account_id":18265,"name":"Rony Haddad","email":"rony.haddad@ericsson.com","username":"RonyH"},"state":"REVIEWER"},{"updated":"2016-04-08 03:43:20.000000000","updated_by":{"_account_id":18675,"name":"Sharath Rao","email":"sharath.rao@ericsson.com","username":"sharath.rao"},"reviewer":{"_account_id":18675,"name":"Sharath Rao","email":"sharath.rao@ericsson.com","username":"sharath.rao"},"state":"REVIEWER"},{"updated":"2016-04-08 08:25:47.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2016-04-08 08:28:56.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2016-04-08 09:31:07.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"7cef223334de0ccd45538c2cbc174b3f93bffed3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2016-04-07 15:29:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10079969dd380ddca32c7d0eafaf809630129841","author":{"_account_id":20996,"name":"Avinash Raghavendra","email":"avinash.raghavendra@ericsson.com","username":"eragavi"},"date":"2016-04-07 15:30:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e4d903cb7a1d0c2bfd94ebe307e79f0c1fdc0b70","author":{"_account_id":20810,"name":"Kshitiz Arora","email":"kshitiz.arora@ericsson.com","username":"EKSHARO"},"date":"2016-04-07 15:31:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"44d67ab787e2aa73a73be6f09ab1b3189faf21d3","author":{"_account_id":20993,"name":"Cody Eilrich","email":"cody.eilrich@ericsson.com","username":"ceilrich"},"date":"2016-04-07 16:33:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"31bf202fa3a65fcea845c32a8dc45c6b7495e408","author":{"_account_id":16272,"name":"Clint Adams","email":"clint@gcfm.net","username":"clint"},"date":"2016-04-07 17:26:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4f2acb7ab461a9324d99af3e1180f2b8af1a92b7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-07 18:04:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-lib-requirements http://logs.openstack.org/92/302892/1/check/gate-tempest-lib-requirements/915edb6/ : SUCCESS in 4m 02s\n- gate-tempest-dsvm-neutron-src-tempest-lib http://logs.openstack.org/92/302892/1/check/gate-tempest-dsvm-neutron-src-tempest-lib/eeec2f2/ : SUCCESS in 54m 09s\n- gate-tempest-lib-docs http://docs-draft.openstack.org/92/302892/1/check/gate-tempest-lib-docs/578bcf9//doc/build/html/ : SUCCESS in 1m 34s\n- gate-tempest-lib-pep8 http://logs.openstack.org/92/302892/1/check/gate-tempest-lib-pep8/cd0f8ae/ : SUCCESS in 2m 01s\n- gate-tempest-lib-python27 http://logs.openstack.org/92/302892/1/check/gate-tempest-lib-python27/310d700/ : SUCCESS in 3m 49s\n- gate-tempest-lib-python34 http://logs.openstack.org/92/302892/1/check/gate-tempest-lib-python34/cc9549c/ : SUCCESS in 3m 14s\n- gate-tempest-lib-releasenotes http://docs-draft.openstack.org/92/302892/1/check/gate-tempest-lib-releasenotes/0abf31b//releasenotes/build/html/ : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1},{"id":"262b15281890ff36e280ddf01db7e0bc54c2a462","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2016-04-08 00:34:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e908877b6d637248055c3f74188dfb14307dd98c","author":{"_account_id":20993,"name":"Cody Eilrich","email":"cody.eilrich@ericsson.com","username":"ceilrich"},"date":"2016-04-08 02:19:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"afbc025c4164b21ddeb2e9a4414fd2e67e7c5a23","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-08 02:31:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-lib-requirements http://logs.openstack.org/92/302892/2/check/gate-tempest-lib-requirements/7f513a5/ : SUCCESS in 5m 16s\n- gate-tempest-dsvm-neutron-src-tempest-lib http://logs.openstack.org/92/302892/2/check/gate-tempest-dsvm-neutron-src-tempest-lib/90176e9/ : SUCCESS in 1h 34m 31s\n- gate-tempest-lib-docs http://docs-draft.openstack.org/92/302892/2/check/gate-tempest-lib-docs/5eabb85//doc/build/html/ : SUCCESS in 2m 22s\n- gate-tempest-lib-pep8 http://logs.openstack.org/92/302892/2/check/gate-tempest-lib-pep8/a2ca0bd/ : SUCCESS in 2m 08s\n- gate-tempest-lib-python27 http://logs.openstack.org/92/302892/2/check/gate-tempest-lib-python27/246b277/ : SUCCESS in 2m 43s\n- gate-tempest-lib-python34 http://logs.openstack.org/92/302892/2/check/gate-tempest-lib-python34/9643221/ : SUCCESS in 2m 57s\n- gate-tempest-lib-releasenotes http://docs-draft.openstack.org/92/302892/2/check/gate-tempest-lib-releasenotes/63d09ed//releasenotes/build/html/ : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":2},{"id":"bb6c6e8e0fcda8f19560b0fbdfe23f47a35ec55c","author":{"_account_id":16272,"name":"Clint Adams","email":"clint@gcfm.net","username":"clint"},"date":"2016-04-08 03:31:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c84a0ef59de9c0fab8f8251d647dc09e0441f61f","author":{"_account_id":18265,"name":"Rony Haddad","email":"rony.haddad@ericsson.com","username":"RonyH"},"date":"2016-04-08 03:37:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7ea9b1fb4c226330a2467def1586d3757592d8e9","author":{"_account_id":18675,"name":"Sharath Rao","email":"sharath.rao@ericsson.com","username":"sharath.rao"},"date":"2016-04-08 03:43:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"25990c96d26a671f7a989d1a0ea0e59eabd0eb58","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2016-04-08 08:25:47.000000000","message":"Patch Set 2: Code-Review+2\n\nI think we should keep requirements to up to date until the end of deprecation. Otherwise, we should drop tempest-lib from global requirements updates.","accounts_in_message":[],"_revision_number":2},{"id":"4bd1c6a589340ba956d24eee53d063c2587c1741","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2016-04-08 08:28:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nYea, till we have this in g-r at least requirement should be up to date.","accounts_in_message":[],"_revision_number":2},{"id":"c2c7fddfeed14b1f52462231066dc95e39ef433c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-08 08:29:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"953d1c50df87f05fd1e8fd3690af78e275c0a3a8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-08 09:31:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tempest-lib-requirements http://logs.openstack.org/92/302892/2/gate/gate-tempest-lib-requirements/e30e533/ : SUCCESS in 1m 40s\n- gate-tempest-dsvm-neutron-src-tempest-lib http://logs.openstack.org/92/302892/2/gate/gate-tempest-dsvm-neutron-src-tempest-lib/9330621/ : SUCCESS in 57m 22s\n- gate-tempest-lib-docs http://docs-draft.openstack.org/92/302892/2/gate/gate-tempest-lib-docs/be87531//doc/build/html/ : SUCCESS in 2m 11s\n- gate-tempest-lib-pep8 http://logs.openstack.org/92/302892/2/gate/gate-tempest-lib-pep8/2cf9d40/ : SUCCESS in 1m 30s\n- gate-tempest-lib-python27 http://logs.openstack.org/92/302892/2/gate/gate-tempest-lib-python27/9b7acbe/ : SUCCESS in 1m 40s\n- gate-tempest-lib-python34 http://logs.openstack.org/92/302892/2/gate/gate-tempest-lib-python34/e519f61/ : SUCCESS in 4m 01s\n- gate-tempest-lib-releasenotes http://docs-draft.openstack.org/92/302892/2/gate/gate-tempest-lib-releasenotes/94fb07e//releasenotes/build/html/ : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4e0424db252e7d705e62c4a95d0cea25198a5e41","date":"2016-04-08 09:31:08.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"000c2b49ee674b875f3b0aae5c61572394c6991f","revisions":{"7be53419b13bbdb9b40b1ffb489fa2371c417d25":{"kind":"REWORK","_number":1,"created":"2016-04-07 15:29:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/302892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest-lib","ref":"refs/changes/92/302892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/92/302892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/92/302892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/92/302892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest-lib refs/changes/92/302892/1"}}},"commit":{"parents":[{"commit":"781aec763e6f46f070c960e5e19a4c0d5be1ecb8","subject":"Add note to README about tempest-lib deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/781aec763e6f46f070c960e5e19a4c0d5be1ecb8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-04-07 15:29:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-04-07 15:29:52.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: Ic3ed2c3d806ab8a4d62eaf1d567c585a1369fab3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/7be53419b13bbdb9b40b1ffb489fa2371c417d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/7be53419b13bbdb9b40b1ffb489fa2371c417d25"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"781aec763e6f46f070c960e5e19a4c0d5be1ecb8","is_merged_in_target_branch":true,"change_id":"I7be8357d67e31d0dc51f453edf02b46dab4f6f63","change_number":285380,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"000c2b49ee674b875f3b0aae5c61572394c6991f":{"kind":"REWORK","_number":2,"created":"2016-04-08 00:34:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/302892/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest-lib","ref":"refs/changes/92/302892/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/92/302892/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/92/302892/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/92/302892/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest-lib refs/changes/92/302892/2"}}},"commit":{"parents":[{"commit":"781aec763e6f46f070c960e5e19a4c0d5be1ecb8","subject":"Add note to README about tempest-lib deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/781aec763e6f46f070c960e5e19a4c0d5be1ecb8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-04-08 00:34:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-04-08 00:34:32.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: Ic3ed2c3d806ab8a4d62eaf1d567c585a1369fab3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/000c2b49ee674b875f3b0aae5c61572394c6991f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/000c2b49ee674b875f3b0aae5c61572394c6991f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"781aec763e6f46f070c960e5e19a4c0d5be1ecb8","is_merged_in_target_branch":true,"change_id":"I7be8357d67e31d0dc51f453edf02b46dab4f6f63","change_number":285380,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
