)]}'
{"id":"openstack%2Ftempest-lib~529489","triplet_id":"openstack%2Ftempest-lib~master~I39109fb0dc5a9b5668ddfcaf4c3a221cf1c5017d","project":"openstack/tempest-lib","branch":"master","topic":"misspelling-fix","hashtags":[],"change_id":"I39109fb0dc5a9b5668ddfcaf4c3a221cf1c5017d","subject":"Fix the misspelling of \"tempest\"","status":"ABANDONED","created":"2017-12-21 06:07:05.000000000","updated":"2018-03-19 09:20:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4fb84c9e92143206041589baed52d62b5318d065","_number":529489,"virtual_id_number":529489,"owner":{"_account_id":24209,"name":"Ding Guoqiang","email":"dgq8211@gmail.com","username":"dingguoqiang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2018-03-12 03:18:24.000000000","_account_id":24209,"name":"Ding Guoqiang","email":"dgq8211@gmail.com","username":"dingguoqiang"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2018-03-12 03:18:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24209,"name":"Ding Guoqiang","email":"dgq8211@gmail.com","username":"dingguoqiang"},{"value":-2,"date":"2018-03-19 08:38:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24209,"name":"Ding Guoqiang","email":"dgq8211@gmail.com","username":"dingguoqiang"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24209,"name":"Ding Guoqiang","email":"dgq8211@gmail.com","username":"dingguoqiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-12 03:18:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-19 08:38:02.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"b5088d0f67459bc1bfbd8e883d63211729250e3d","author":{"_account_id":24209,"name":"Ding Guoqiang","email":"dgq8211@gmail.com","username":"dingguoqiang"},"date":"2017-12-21 06:07:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cb0dc6e527017d8486f06fd8811d9ca64c23fa5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-21 06:48:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/529489/1/check/build-openstack-sphinx-docs/c3289bf/html/ : SUCCESS in 5m 38s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/89/529489/1/check/legacy-tempest-dsvm-neutron-src/ae7c30d/ : FAILURE in 38m 19s\n- openstack-tox-pep8 http://logs.openstack.org/89/529489/1/check/openstack-tox-pep8/9862478/ : SUCCESS in 5m 52s\n- build-openstack-releasenotes http://logs.openstack.org/89/529489/1/check/build-openstack-releasenotes/957442e/ : FAILURE in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6a5966649cb7b26d7e7eab939c083b20e1006724","author":{"_account_id":24209,"name":"Ding Guoqiang","email":"dgq8211@gmail.com","username":"dingguoqiang"},"date":"2018-03-12 03:18:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6780d3fd37e474c4e10920e324517395390e1261","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-12 03:18:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop noop : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"f99c910a65ee508aaed88de5cb13f60e4640de25","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-03-19 08:38:02.000000000","message":"Patch Set 1: Code-Review-2\n\nThanks for patch, tempest-lib is deprecated long back in favor of Tempest/lib. We avoid any updates in tempest-lib. please propose patch if same issue is on tempest also.","accounts_in_message":[],"_revision_number":1},{"id":"d16d2579a70236cd0d34e6451213a1e3c1ae7b63","author":{"_account_id":24209,"name":"Ding Guoqiang","email":"dgq8211@gmail.com","username":"dingguoqiang"},"date":"2018-03-19 09:20:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9954b8bb693f8c7eea1fd0c0f460ea9d550743f0","revisions":{"9954b8bb693f8c7eea1fd0c0f460ea9d550743f0":{"kind":"REWORK","_number":1,"created":"2017-12-21 06:07:05.000000000","uploader":{"_account_id":24209,"name":"Ding Guoqiang","email":"dgq8211@gmail.com","username":"dingguoqiang"},"ref":"refs/changes/89/529489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest-lib","ref":"refs/changes/89/529489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/89/529489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/89/529489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/89/529489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest-lib refs/changes/89/529489/1"}}},"commit":{"parents":[{"commit":"000c2b49ee674b875f3b0aae5c61572394c6991f","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/000c2b49ee674b875f3b0aae5c61572394c6991f"}]}],"author":{"name":"Guoqiang Ding","email":"dingguoqiang@cloudin.cn","date":"2017-12-21 06:06:06.000000000","tz":480},"committer":{"name":"Guoqiang Ding","email":"dingguoqiang@cloudin.cn","date":"2017-12-21 06:06:41.000000000","tz":480},"subject":"Fix the misspelling of \"tempest\"","message":"Fix the misspelling of \"tempest\"\n\nChange-Id: I39109fb0dc5a9b5668ddfcaf4c3a221cf1c5017d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/9954b8bb693f8c7eea1fd0c0f460ea9d550743f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/9954b8bb693f8c7eea1fd0c0f460ea9d550743f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
