)]}'
{"id":"openstack%2Ftempest-lib~788221","triplet_id":"openstack%2Ftempest-lib~master~I303e7f56eb192036c13dfd1f3a3526b77db76599","project":"openstack/tempest-lib","branch":"master","hashtags":[],"change_id":"I303e7f56eb192036c13dfd1f3a3526b77db76599","subject":"setup.cfg: Replace dashes with underscores","status":"ABANDONED","created":"2021-04-27 08:59:55.000000000","updated":"2021-05-24 14:58:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a79d78d60a8a6f9fb8d32d5c8d7d7d509f3aa982","_number":788221,"virtual_id_number":788221,"owner":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-28 07:03:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-2,"date":"2021-05-24 14:56:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 09:00:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-24 14:56:20.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"47da3abe9d1c6279a30485b9f5d6a2252b93b97c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2021-04-27 08:59:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bd243dc4ab025b368f54c6bf338b4cb0ab61049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 09:00:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/24806d0ef2c04d6b9c9d2798c4e7e8be : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"134f59f6b4d7b72dc9da9c1bad5f08ff05cc2dbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2021-04-28 07:03:14.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"779ada77a893ec5481c67724e49eaaa30a5c0c90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 07:03:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/ec4aa21f5c174aa89fa5c313be28e8ab : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"aa48645258729f49a2ecf24da09c229b63643538","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-24 14:56:20.000000000","message":"Patch Set 2: Code-Review-2\n\ntempest-lib does not accept the changes any more, please do it on tempest repo.","accounts_in_message":[],"_revision_number":2},{"id":"a79d78d60a8a6f9fb8d32d5c8d7d7d509f3aa982","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-24 14:58:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0bf963e1e2d2a90485eda124646948d250fce72c","revisions":{"d571685d7af13d0a2fd3b676845de76ca0b27591":{"kind":"REWORK","_number":1,"created":"2021-04-27 08:59:55.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/21/788221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest-lib","ref":"refs/changes/21/788221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/21/788221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/21/788221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/21/788221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest-lib refs/changes/21/788221/1"}}},"commit":{"parents":[{"commit":"023426894a4f72d906ed6f79c55ed7152a732b44","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/023426894a4f72d906ed6f79c55ed7152a732b44"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2021-04-27 08:59:01.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2021-04-27 08:59:01.000000000","tz":480},"subject":"isetup.cfg: Replace dashes with underscores","message":"isetup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: I303e7f56eb192036c13dfd1f3a3526b77db76599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/d571685d7af13d0a2fd3b676845de76ca0b27591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/d571685d7af13d0a2fd3b676845de76ca0b27591"}]},"branch":"refs/heads/master"},"0bf963e1e2d2a90485eda124646948d250fce72c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-28 07:03:14.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/21/788221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest-lib","ref":"refs/changes/21/788221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/21/788221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/21/788221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest-lib refs/changes/21/788221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest-lib refs/changes/21/788221/2"}}},"commit":{"parents":[{"commit":"023426894a4f72d906ed6f79c55ed7152a732b44","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/023426894a4f72d906ed6f79c55ed7152a732b44"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2021-04-27 08:59:01.000000000","tz":480},"committer":{"name":"MaAoyu","email":"maaoyu@inspur.com","date":"2021-04-28 07:02:59.000000000","tz":0},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: I303e7f56eb192036c13dfd1f3a3526b77db76599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/0bf963e1e2d2a90485eda124646948d250fce72c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest-lib/commit/0bf963e1e2d2a90485eda124646948d250fce72c"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
