)]}'
{"id":"openstack%2Ftempest~1003196","triplet_id":"openstack%2Ftempest~master~I71c762ea963d900396c510fef236238c8b163cb3","project":"openstack/tempest","branch":"master","attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-02 00:22:35.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-02 00:22:30.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I71c762ea963d900396c510fef236238c8b163cb3","subject":"Fix typo in error message","status":"NEW","created":"2026-09-01 01:30:19.000000000","updated":"2026-09-02 00:22:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e7c80330449597538f9f67946951c8cba406513b","_number":1003196,"virtual_id_number":1003196,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-01 22:53:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2026-09-01 01:42:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 01:42:18.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-01 03:19:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 00:22:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"98ff3bcb5708ef6abad6d01c99b9268b73230fc0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 01:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9579b78a919e939ff09fcf00f6c5da8cd045c888","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 01:34:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e6eac5b7f8d1d9f95274120f336d8416885faa22","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 01:34:57.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"4d248b1e2e046222f5ea4f4f4b83672fe8a570b0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 01:42:18.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/6a4427b4e8c94a7eaa637e72d081be7f\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/fc04dfdc8242433eafd1a3c32e33e74b : SUCCESS in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab97cb5ce484fbb0f12e2e7990d71f98654c3bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 03:19:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/512077c753434a57ac4d9c636a090e65\n\n- grenade https://zuul.opendev.org/t/openstack/build/4b7b15d2a4f6471f977e6c1289cc1e58 : SUCCESS in 1h 05m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06798923440b4f2d9bbaad1f781c9086 : SUCCESS in 1h 43m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87d5907059214c7580df35d999400e2e : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e1ec9f45764684afee549292c36a6c : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7366a1989f4d4f1ab6bb42887cd4bc1f : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b70313c1e99b4539818b5d65ca228db3 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6411dcf991d846a79963dedfd00bdf15 : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f02e516dabe42d0b96b9b34c00a4a4a : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0be8b436661d4be7b7f51b11ee6302bf : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e7abf9725ae74d1591148b96a4de5f8e : SUCCESS in 5m 48s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/46eb74588d09489ab58aebd0c58f5e43 : SUCCESS in 51m 01s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/d364c51141934bc58b1af2a3a9606c4e : SUCCESS in 36m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0719a75e7ae64f7789dc4c3d412e8328 : SUCCESS in 1h 39m 52s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/c21bb6bce01c4a47a7931a45b6c0be52 : SUCCESS in 53m 21s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/e0cfada13d814b90b7cd636b0db65388 : SUCCESS in 1h 09m 00s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/d3e3f1bb9d0a49ab88ffd1f569767893 : SUCCESS in 1h 34m 51s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d3049ef51a7e4cb7a3bf2ef478dfb734 : SUCCESS in 40m 51s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f130d7a117154f8798c85b7a35f12990 : SUCCESS in 5m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8c5806881294076a654eb3982b77aba : SUCCESS in 45m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd36bf300352435d80c06c136f39f711 : FAILURE in 43m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f34a09b2d5ea4ec184172eb4b1443845 : SUCCESS in 42m 23s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/647c87dbbfe446ae9e096260f0e9771c : SUCCESS in 40m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3543f24afe554183b0ba56b8b0080528 : SUCCESS in 1h 16m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/272645e9cb504998b7b65ea4a6eb06c0 : SUCCESS in 50m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d6777627c0f04d23b2053f6081a991b7 : SUCCESS in 52m 24s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/ff5cf9862536441c9d72ceeed673dc85 : SUCCESS in 33m 12s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/f5d26d6588fd4a35b48e507cb7548491 : SUCCESS in 52m 22s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/4a5edbf741bc4fb6a402ce4c27069cf5 : SUCCESS in 1h 06m 02s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/dfed984375a2465f982b78bfd78ba57a : SUCCESS in 28m 36s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/eab6838848d04fdba3d878c55dad371d : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"70a898c8814c9c9e74830e00ea47d05f13241844","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 20:41:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5642aeb802a9ade1fee3a84bc046548061faeee7","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 20:48:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/524cb1c4b93d43e49004f5f547fc72e5\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/bda302dd711245fc8daac6589a4ab724 : SUCCESS in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8f8c9f5603872d5628b10a84a301e100094505c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 22:53:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2bf388b71444eeeb92ee5d883a122f3\n\n- grenade https://zuul.opendev.org/t/openstack/build/6b7e0fa676c84c30ade3153b50119d0a : SUCCESS in 1h 06m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dee555773b934c299344524b911b9675 : SUCCESS in 1h 36m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61bd1dd654cd45b288dc0c14b9b11234 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f04a7bf62dcc44f484cad559bd1e6d8f : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd47b64d573c42579d56d8abce464d2f : SUCCESS in 35m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de248207a5bb4fcab79db6b54945d65c : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95307e48ac064f92b30b9306cac3851f : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f488ace677643aeb0cbf7179eb7c1b6 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15c9689c0ea44ed18153f22e12dcd1eb : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a0bfb2a4cf2e4f0d9b4d1ed9a398c6e3 : SUCCESS in 8m 36s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f0f367a4146c494492ed432633517085 : SUCCESS in 38m 47s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/8060fd1769754406a008eb91283e661f : SUCCESS in 1h 11m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a03da53705334dc489798c194a2b65b3 : SUCCESS in 2h 08m 12s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/15d13c38e5f04bb0aed1e0f8554a6810 : SUCCESS in 1h 22m 06s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/85bce2ae770944a8919441e3409cab3e : SUCCESS in 1h 43m 37s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/fe926e83c44b4d8fa664bbdcdf89f30d : SUCCESS in 1h 20m 29s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0edd3ef9ba1c4a3298915cb27f65470a : SUCCESS in 1h 10m 40s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/34b9de67083b44ddbd6145a96b4e25f1 : SUCCESS in 13m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1328db2240da452cad9824be60af1570 : SUCCESS in 1h 02m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0c9f95564d3c4e01b23d45f37127bbe4 : SUCCESS in 1h 24m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/24bede8332e2463f8ccee40ac7c7e559 : SUCCESS in 59m 08s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/5045e5a7654a46708f4c3bf018db971e : SUCCESS in 46m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/09ae3027ddeb4b72b491dd7674110921 : SUCCESS in 1h 32m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a08d9f82a9a467aa3d13d299b784c07 : SUCCESS in 39m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a0811d9c99445d1808d93ecb3dac24f : SUCCESS in 1h 07m 58s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/53e9f3bbc0704b61a00ca699da6bf55b : SUCCESS in 1h 18m 14s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/bd32a420a9a94ad384c43143996c608c : SUCCESS in 1h 38m 22s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/6df0c1e09e50403290ee61fc0ef6af9b : SUCCESS in 31m 03s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/91a29dfe66b14563af0ba6d5be447889 : SUCCESS in 58m 02s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ede2610864d94109a97af65432a69039 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"55b702f636042220095a2918cf8d9efc1208da17","revisions":{"bd5e39d76c15a1d9419428ffc957bca1c1007589":{"kind":"REWORK","_number":1,"created":"2026-09-01 01:30:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1003196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/96/1003196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/1003196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/1003196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/1003196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/96/1003196/1"}}},"commit":{"parents":[{"commit":"ea2da20690b04b3cd62de1cc57ed5c1a1b370a2e","subject":"Fix list_live_migrations for the server uuid field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ea2da20690b04b3cd62de1cc57ed5c1a1b370a2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 01:30:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 01:30:19.000000000","tz":0},"subject":"Fix typo in error message","message":"Fix typo in error message\n\nChange-Id: I71c762ea963d900396c510fef236238c8b163cb3\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bd5e39d76c15a1d9419428ffc957bca1c1007589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bd5e39d76c15a1d9419428ffc957bca1c1007589"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"55b702f636042220095a2918cf8d9efc1208da17":{"kind":"REWORK","_number":2,"created":"2026-09-01 01:34:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1003196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/96/1003196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/1003196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/1003196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/1003196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/96/1003196/2"}}},"commit":{"parents":[{"commit":"ea2da20690b04b3cd62de1cc57ed5c1a1b370a2e","subject":"Fix list_live_migrations for the server uuid field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ea2da20690b04b3cd62de1cc57ed5c1a1b370a2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 01:30:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 01:34:50.000000000","tz":0},"subject":"Fix typo in error message","message":"Fix typo in error message\n\nChange-Id: I71c762ea963d900396c510fef236238c8b163cb3\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/55b702f636042220095a2918cf8d9efc1208da17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/55b702f636042220095a2918cf8d9efc1208da17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
