)]}'
{"id":"openstack%2Ftempest~1003527","triplet_id":"openstack%2Ftempest~master~Ie59e431d70e21f9d1bc4fd66d94bba69722d503c","project":"openstack/tempest","branch":"master","topic":"OSPRH-36303","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-09-04 20:25:27.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-09-04 20:25:27.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-09-04 20:25:27.000000000","reason":"Change was submitted"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2026-09-04 20:25:27.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-04 18:40:26.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Ie59e431d70e21f9d1bc4fd66d94bba69722d503c","subject":"Skip test_router_rescheduling test if there is no L3 agent available","status":"MERGED","created":"2026-09-02 13:10:28.000000000","updated":"2026-09-04 20:26:36.000000000","submitted":"2026-09-04 20:25:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003527-OSPRH-36303","meta_rev_id":"e01a91e6b432fa69432b3d41ff4cfb13eaaf836f","_number":1003527,"virtual_id_number":1003527,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-04 20:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2026-09-04 18:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2026-09-04 18:40:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 19:11:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 04:39:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-09-03 04:39:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2026-09-03 04:39:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-09-03 14:11:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"cc96883c4c9ddda0d1d6cbac30af8e2866413ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-09-02 13:10:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55af86e60311c559da83dbcb1e5a8d4c38055831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 19:11:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56515d891d3541eda6499c1196f454e0\n\n- grenade https://zuul.opendev.org/t/openstack/build/61361b28257c4c6e9a0434f6eff2e98c : SUCCESS in 58m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2606b519b9d4e519e901da0aae60896 : SUCCESS in 1h 34m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d636726d3654951b7d1f4c399743449 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9f9a06f672343b1b55966fbce290f41 : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4186e3d4c2b64baf80cfefabd9a2682d : SUCCESS in 34m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53612dc796304d12825adc3054494b0e : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ee60deeed2f4666a234d7e66dd4d390 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd80aea1d59b44dfa8f0e7144fcc0492 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bd8cf0c8e034098af32629a8c4c0400 : SUCCESS in 5m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f49756b49da4a56a8add3caa27f08af : SUCCESS in 9m 35s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a803bc587a5f4e3eb3be253bece205f6 : SUCCESS in 51m 45s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/349fee515b95463facc6660616f72773 : SUCCESS in 1h 01m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/51dfb9cf4bb24fcba21dafdc4d1021bc : SUCCESS in 1h 56m 33s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/57b4ee10b7fc4d22bf75f6d6ef9e030f : SUCCESS in 1h 21m 30s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/e4dab01a629246f4ad297e52a660bb70 : SUCCESS in 1h 09m 05s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/0bf554304e5c472fb96396fbdcb29162 : SUCCESS in 1h 43m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/99052c810be64b3eac5cf6f4d795ff81 : SUCCESS in 40m 19s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/bd9feb14606f4329b8dda933c378a62c : SUCCESS in 13m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0e10a4de5224cb491aab94c644d76ab : SUCCESS in 57m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1e53315a504f46dbb7eac10164975f69 : SUCCESS in 1h 53m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/447a06cec6854befac24356785627075 : SUCCESS in 52m 04s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/9871fb14aef14e45924c4f29c8a1527c : SUCCESS in 1h 45m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e5e39255b762490495125cca4843adcf : SUCCESS in 1h 04m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba8c6a3ff50148c2bd89ca8a8eacd913 : SUCCESS in 59m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ff30c403be834aeebdf85a8ed4248d2d : SUCCESS in 32m 48s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/f25f785943254619b13ac54e6fc0db86 : SUCCESS in 1h 06m 59s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/e266380974ab46989dbaee97ff33c049 : SUCCESS in 1h 36m 33s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/a378ae34b0e94a8b8c71bad1418491f2 : SUCCESS in 1h 21m 37s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f638d92a53b44c0582b6a34d8fd311ec : SUCCESS in 1h 01m 36s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9692018bdecc476fa24986c2e6dc5f37 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"fb0285c674161c7ac95bf3e76e66ba0149397bb6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-03 14:11:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2868a583f0e6597c68291fa47f74d966bec89f8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-09-03 14:13:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"819412a579ab37bf25c6fff8c971f804e12ec499","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-09-04 07:52:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e645e134f770818448540a64ce5492f130882b8c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 18:40:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24b23b00f2b76867ae069d8f55b13db022c1cf81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 18:41:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4331186b7049eff2c986f10b55c84f5d6167df0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 20:25:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea8dba6e90454f028f1765da7254e2ad\n\n- grenade https://zuul.opendev.org/t/openstack/build/625606639056448ea7b4a91c588d702e : SUCCESS in 1h 03m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd7f2ac1e89a4e199ae45e537d28d6f2 : SUCCESS in 1h 31m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4317de709fdc46758c149000b3662049 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/414c823779d041fcb7a47da9ec7e1402 : SUCCESS in 17m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/461fcba7a73d41be935fa1acebafa4a8 : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/763df20efd894b26af7b881ce210d2f9 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/519b1fee0a18411d9d00fe07fb1d6e3c : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da03dbfbf400470f9ddeeacfbe8d652b : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/45194bd5db034172b04a9113656223dc : SUCCESS in 8m 15s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f3a9bfe727434926a587108c4993dcd4 : SUCCESS in 1h 05m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dfc294e00b91437db2c53430d1a8743c : SUCCESS in 1h 38m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef05e04a7368454db6a3046feb136dc6 : SUCCESS in 1h 12m 19s\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/15c3b60a72b7409a81fb42b2e0fd1936 : SUCCESS in 1h 04m 31s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/0a08458b974e4f37b36d3ea67ef5d95b : SUCCESS in 39m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5aa796cd5b9243558550d8f0c6723961 : SUCCESS in 47m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d34d2a95d64e45049722b89480eeaaff : SUCCESS in 54m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/faab59f5471e4fb883f1d47d04248fb5 : SUCCESS in 1h 09m 25s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/39196f0dc98b47f9b7465e2224e6aac8 : SUCCESS in 1h 04m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/57200f14961c4f8b8aab99c6ff83253a : SUCCESS in 1h 21m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/baac511ff15049db8997a0fd4ce256c4 : SUCCESS in 35m 28s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/8ac1b08405da4e21b1daa54e2029478b : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bd3b5b391f7d80f1a54a09c4e684003cc03d4945","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 20:25:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e01a91e6b432fa69432b3d41ff4cfb13eaaf836f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 20:26:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b9470faac9c45d7a4de0036f15fd561\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6715b8ae20a145a2ba0125f1e0da48c6 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"015bf36eec1b6fabfb1197f832f41e4a35f647d2","revisions":{"015bf36eec1b6fabfb1197f832f41e4a35f647d2":{"kind":"REWORK","_number":1,"created":"2026-09-02 13:10:28.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/27/1003527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/27/1003527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/27/1003527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/27/1003527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/27/1003527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/27/1003527/1"}}},"commit":{"parents":[{"commit":"0a19cf44f2b257cec6b7c070e9a6df0cc3500172","subject":"Merge \"Detach ports from vm before deleting them in Neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0a19cf44f2b257cec6b7c070e9a6df0cc3500172"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-09-02 12:57:32.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-09-02 13:08:09.000000000","tz":330},"subject":"Skip test_router_rescheduling test if there is no L3 agent available","message":"Skip test_router_rescheduling test if there is no L3 agent available\n\nPreviously the test was skipped only if there wasn\u0027t\n\"l3_agent_scheduler\" API extension available.\nBut since [1] this extension is available with ml2/ovn as well\nbut as the test is only relevant when L3 agents are deployed.\n\nSo the test is now skipped also when there is no L3 agent\navailable in the deployment.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/982792\n\nRelated-Issue: https://redhat.atlassian.net/browse/OSPRH-36303\nChange-Id: Ie59e431d70e21f9d1bc4fd66d94bba69722d503c\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/015bf36eec1b6fabfb1197f832f41e4a35f647d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/015bf36eec1b6fabfb1197f832f41e4a35f647d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
