)]}'
{"id":"openstack%2Ftempest~1004029","triplet_id":"openstack%2Ftempest~master~I79a05b5cda923aa33a9aa05898d5998b5be4f336","project":"openstack/tempest","branch":"master","topic":"qa-2026-2-release","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-08 01:57:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I79a05b5cda923aa33a9aa05898d5998b5be4f336","subject":"Use stable constraint in tox to release new tag for 2026.2","status":"MERGED","created":"2026-09-04 19:28:33.000000000","updated":"2026-09-08 01:59:08.000000000","submitted":"2026-09-08 01:57:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004029-qa-2026-2-release","meta_rev_id":"c981076ba9f0e3a70026dcf27e34f8b0ee930721","_number":1004029,"virtual_id_number":1004029,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 01:57:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2026-09-05 01:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2026-09-05 01:21:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 21:42:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 22:29:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"472a4ff380475b17de7fddccc31b732e57baec0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 19:28:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a53b2c3d81573573271956019395af0959e2400a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 21:42:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa26774973ca44d8987c300300b0b5f1\n\n- grenade https://zuul.opendev.org/t/openstack/build/e9ab56ca4fa44cab9cf4cacf4cb016e5 : SUCCESS in 45m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0deb6689e0964e9e8c378bf220917141 : SUCCESS in 1h 25m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2e39d39b0994c5293a301285c715632 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36522a6077949509663ee2001a47424 : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8df1e70438da482d882e21310174fde4 : SUCCESS in 32m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6263a4d585334142947ddb4acb3686db : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed22cacf9e15465b8ac74f9ff1df0aa1 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdbcab3b26084d34ba51414e5db89ca3 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96ae45caebb54303ae7e93a44b0447ce : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5c45168bf7504c119a95d2721ca66826 : SUCCESS in 7m 35s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/383a7281934d44e68cb8eba4128c3a40 : SUCCESS in 1h 23m 11s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/2f175e01bfc84dba836766a303ab0262 : SUCCESS in 52m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cea259de70a94d3dbb3ccbbb12d25f2a : SUCCESS in 2h 00m 29s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/cecfa70aaf474e8289f07e6f0125ee41 : SUCCESS in 2h 07m 38s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/ab7ad4731d1145cdafc88650b78c6f3c : SUCCESS in 55m 02s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/c3439ad7424543299cd4b4086232f401 : SUCCESS in 1h 34m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/76d8bfe0a12c45eeb8f25de908175da1 : SUCCESS in 1h 12m 41s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/71d455e641ee42cfb9065455ed8f0836 : SUCCESS in 8m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b775bd2d2c5a4d868bbbf5690dd5ea0a : SUCCESS in 32m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4b10ceb2c754439ab6a72aa1686ca6fa : SUCCESS in 1h 29m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c92540a0ec044c4882342f71bd21c84f : SUCCESS in 23m 29s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/6c4a5df990454a148a0af5009fd83ed5 : SUCCESS in 1h 27m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/11c0dffeb70c4a05ac07f2ceb45060b9 : SUCCESS in 1h 18m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/60f5f78cf77a40a18573be1d7c9c6d2f : SUCCESS in 1h 15m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9ea80dbff6d1465c8ea787764b49e464 : SUCCESS in 1h 07m 15s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/8f5811230b794ab7b52f38ddcfecbd2d : FAILURE in 58m 58s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/9d193a5367d3482c9f24e9b2e8a486ca : FAILURE in 59m 55s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/efba4dce3b5240b4b1e29bf72750d29b : SUCCESS in 1h 14m 59s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/39a8f4238c5e4e6ebcc084a1c81c8261 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8ce67d612ea917d7f461700d6adbd26b51e526bd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-05 01:21:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"549268966707b115adf53cea5f2e62993a08d1b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 03:20:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21684013d3fe489ca5e32fc4b7c0551e\n\n- grenade https://zuul.opendev.org/t/openstack/build/43820a49f7a54375821bebe86450320b : SUCCESS in 30m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bb7eb863097478e802b58dabdf34016 : SUCCESS in 1h 16m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a72f52d7b5c4511bfd8b4264622f512 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc559d9ef28a407f8e24d8f55662478c : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87cdf988e2364372b0802c217301261d : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5620d8f3149445e94d6522d3c944cde : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d8b4a191795416fa124dca3e51e2009 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7f843cea4fb472d98315bf41ea12759 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c5331b9a4994d0bbd203cd17da09fde : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e9dcc992269445e7ae4eac626a50e5d6 : SUCCESS in 4m 27s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/6e07eb56df694f97a6c60078c0a82431 : SUCCESS in 1h 19m 37s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/aec9fbf5731544a1b61c4946636e95c3 : SUCCESS in 1h 00m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ef5d9d458292438eaf043ac66705a9dd : SUCCESS in 1h 55m 25s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/121a6c34ecd64e40a25688c4e15ee229 : SUCCESS in 1h 42m 10s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/ed8339e9b7e2421c88d3232e83cfbf35 : SUCCESS in 1h 40m 02s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/9a443cea8c5b4716b5dcd80752a2ee23 : FAILURE in 1h 26m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/70f1c1ec9a8f4369baa543ecde0795a4 : SUCCESS in 35m 45s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/19bf417193a34c7dbb428cc423d5f801 : SUCCESS in 6m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b84a86d8944c4f008a77251e4df01d4f : SUCCESS in 56m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/057126ebdcbc4df58c516c06f60cd8e0 : SUCCESS in 1h 37m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4c81037495d04aac87fb14ea39c83fa4 : SUCCESS in 36m 29s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/8151d7565c964e5b99ccec2de84771c1 : SUCCESS in 1h 27m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/edfa87d8627c44e988530aa767318e18 : SUCCESS in 45m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e25f78e801a4174b5b354e09a912161 : SUCCESS in 1h 11m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d7a7ff9d283149bb898e32b4712820e7 : SUCCESS in 32m 29s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/25fa660894784252aad2408b06498c26 : SUCCESS in 1h 17m 29s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/b2b6dd07a7584f538de09fd71d4c67eb : SUCCESS in 45m 35s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/08ef1c811f414b61844326f9011d6571 : SUCCESS in 33m 27s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3599a5f8ae0941b08045e1d451ac2fd7 : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c40567885c531a3dabf353798dc08d6616b09e12","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-05 04:41:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"468ca402859308f8a1d70ee5ea300d5e17af080e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 06:08:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c38fc3584361418ca04e7bbb823c7fdf\n\n- grenade https://zuul.opendev.org/t/openstack/build/b485923f24094c8c9dc548ced1ab2eae : SUCCESS in 34m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b81ae8971ee44aa984a86216d0cc520 : SUCCESS in 54m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c20952e4abb54a7994c5c36223fb9ae3 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acfd34d75ee640cf9fc9f752993ca131 : TIMED_OUT in 31m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62921b7ed6ff4c7e9ad60ea997e756cb : SUCCESS in 13m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae990023362e4afa81f760991fb5bead : SUCCESS in 2m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3bbffde025e64fbf8b9380952f8c2a13 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7900cb8df994c47bf6bea7236575c0b : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/366b0d2affb14503b99a2e620f6a677d : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e571d11b03af4e23b508ff7b62933d9b : SUCCESS in 10m 42s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d6c91bcffe354afa8634ed3251d9899d : SUCCESS in 1h 01m 00s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/5d7671f9bbaf4f25ae1ca1bc452a8ad9 : SUCCESS in 32m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8ef4edb9209440f79651556c648c2d65 : SUCCESS in 1h 07m 34s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/70bf097431624185862eb3c4ff351853 : SUCCESS in 1h 01m 55s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/50ad319e1ab044699493891522716b57 : SUCCESS in 54m 23s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/d28c208314224f4ebb12576a959b3fff : SUCCESS in 1h 25m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ccd914d2474245aeb8280712a8da4123 : SUCCESS in 50m 38s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/94f5e035861542c7a91653c3dfb30fb0 : SUCCESS in 25m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20a26769baf5459983ef626da03d61fa : SUCCESS in 27m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ef0732b450bc4802937d66062bb80895 : SUCCESS in 44m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b4b8f6d047654ce4a7da90efdebf1118 : SUCCESS in 48m 22s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/2eed1900dd444ab08a8f7bfff19bf284 : SUCCESS in 44m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ad51c378936742a284b0452a4cf2b5e1 : SUCCESS in 44m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebcdda82e8c2495e89f4bed1f408f10f : SUCCESS in 1h 08m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d42ce4db2aeb4e2e96c9243d9f120bbc : SUCCESS in 1h 10m 11s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/fc64a3b76710448682115993183d85d6 : SUCCESS in 1h 09m 22s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/aa87d6cbd7834100a76e77b21e057535 : SUCCESS in 1h 10m 16s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/f6dff53965aa4b09830f0542b78f5452 : SUCCESS in 1h 08m 35s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c9fa2ed3c35544efa42c066dfbe43889 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"979b81cc8f8ab6a4644c20911f06e2565d4166ba","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 22:29:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d13b78d95ca22f2f10df80fdb6a86ff6fc282839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 00:17:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6261ce5ba44d47658903a158b54ca588\n\n- grenade https://zuul.opendev.org/t/openstack/build/b9ec013a72554fca8037d0642677f5c9 : SUCCESS in 43m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1119ab288105403dbab712639e1815fa : SUCCESS in 47m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65635d50c886472883a34f162a348374 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aaf3ea005a44b0ab1406558d2a70949 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/270fab2dea1640579f28ae92b929e7f8 : SUCCESS in 33m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/350ec0cdac8c4221ae9a7ea3d8b04d0e : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5136a6390074fd3997648b5d61a6150 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cfbc44b8ca1b4607b26c29053c373400 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c4175567b934680b2250ca9d895e77e : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b18933357f804fe7a232f1cbde2a4c6a : SUCCESS in 7m 21s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b41732ffc3f4478eaea9f65dafb17dee : SUCCESS in 48m 48s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/60cdd183bd3a49b0a73edb153b05930f : SUCCESS in 56m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0f04539c9ce244eeb69b6f3aa48a8b60 : SUCCESS in 52m 40s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/9f51c9917d8c4237aabbf305ad4a303e : SUCCESS in 1h 42m 18s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/ea1e21a4a6d943e08ed6d421ce13b4e4 : SUCCESS in 1h 16m 48s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/58432200b87d4c1199222c7eec630a3a : SUCCESS in 1h 24m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e4f5dbf724444b59b9354caad5a81062 : SUCCESS in 1h 15m 31s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/651eb5d2440a47a7a202842383f158d8 : SUCCESS in 11m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62a613cb15e649d59788e8a5eeb7a70b : SUCCESS in 33m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6d2c24b767c544bdb6685eb0ba7fb2db : SUCCESS in 1h 38m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a10ed5be869c4a1fa4785bdb9abbe656 : SUCCESS in 50m 36s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/97de5d358b8b4bf38903fccaa717e0e2 : SUCCESS in 1h 02m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dd6cd1efda524788a2573b4563b3c960 : SUCCESS in 48m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/781149764e7e4a24ad54408fadc12eee : SUCCESS in 1h 08m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c19ed13a855e44769cb785a76e36078e : SUCCESS in 1h 03m 59s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2c8cd50551f94202b7aa6e5b0826f8ab : FAILURE in 41m 08s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/9d0f830cf36c460190d0c2fce6641764 : SUCCESS in 1h 14m 42s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/f7a0d27084e94665b0d8447a45e3a069 : SUCCESS in 35m 36s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b0c07b0d7d514a609328d7bae1f2d3be : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b7f5919ab3fd727451754f023ad0a501f389c3dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 00:18:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2b1db3688f032102e39924c963cb094b944bb0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 01:57:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40f4dce3661c48c18307920874fbd850\n\n- grenade https://zuul.opendev.org/t/openstack/build/83650a9f57044c9bbb4e500156f55582 : SUCCESS in 32m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82808a76c3b448c99fe0d2e91c7147e4 : SUCCESS in 1h 33m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/622f26b46f3241848872efbcf87070fc : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/797810c3f46b4a1ea94cff6f37a7ce3b : SUCCESS in 31m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b437ee33aae40d48fd53f5028c00f80 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ffb34ef907b744a780a5fb1acb9aeed8 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c80b326bd784dd9b233a6be69b5ce40 : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5de32729b44145449b311ce4f0198bed : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e5e22d1ba54a43869e6f6e892b13478e : SUCCESS in 7m 04s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9e1770b461cf42cdbd86983675213415 : SUCCESS in 1h 22m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ce31b99c6756486ab57f4ff88d3b4f62 : SUCCESS in 1h 29m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7749679e071400aab53423f798bc6a5 : SUCCESS in 1h 10m 19s\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/038c4d885905485d80a72ebf22a8f6fc : SUCCESS in 39m 56s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/1498f2670ebc4bceb9719e0f1bf2a657 : SUCCESS in 32m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb4726a66ff24e0996bfce7753cec9bb : SUCCESS in 33m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fff11b7f67c2400a91011db3d858e1c4 : SUCCESS in 48m 28s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4691d2d455c54f45a5d1cc3b42763cde : SUCCESS in 1h 10m 11s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/8ab57b165ac742e9bc15dab41fb85af2 : SUCCESS in 1h 01m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9279c00d05f64f79922a99fb5f59d726 : SUCCESS in 1h 24m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/101b3a81e126407895a5e7e2040d61fc : SUCCESS in 52m 12s","accounts_in_message":[],"_revision_number":1},{"id":"cb0e8f95bfaf419c7d8ae213a306baaa1997fcd8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 01:57:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c981076ba9f0e3a70026dcf27e34f8b0ee930721","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 01:59:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d07d1009af294c6eb620ca92f212c4f3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b2086189ec2419b9315198c77d6c8e9 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bc51c4f823746228b18438f81c628e9 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8edabfd5976b6aaf0e40649cefd908ccddcb0d48","revisions":{"8edabfd5976b6aaf0e40649cefd908ccddcb0d48":{"kind":"REWORK","_number":1,"created":"2026-09-04 19:28:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/29/1004029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/29/1004029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/1004029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/1004029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/1004029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/29/1004029/1"}}},"commit":{"parents":[{"commit":"4af3622e5db45aabbbe4e7850b97590db45a4283","subject":"Add releasenote to tag the Tempest for 2026.2 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4af3622e5db45aabbbe4e7850b97590db45a4283"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 19:27:01.000000000","tz":-240},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 19:27:01.000000000","tz":-240},"subject":"Use stable constraint in tox to release new tag for 2026.2","message":"Use stable constraint in tox to release new tag for 2026.2\n\nWe are going to release Tempest new tag to declare the start of\nsupport for stable/2026.2. So that new tag use stable/2026.2\nconstraint in the tox env.\n\nFor example: if anyone use Tempest new tag in future say\n1 year later then tox env also should use the 2026.2 contstraint for\ncompatibility instead master.\n\nAlso, making ironic-tempest-bios-ipmi-direct-tinyipa as\nnon voting to merge his change.\n\nThese need to be reverted once release is done.\n\nChange-Id: I79a05b5cda923aa33a9aa05898d5998b5be4f336\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8edabfd5976b6aaf0e40649cefd908ccddcb0d48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8edabfd5976b6aaf0e40649cefd908ccddcb0d48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
