)]}'
{"id":"openstack%2Ftempest~1004030","triplet_id":"openstack%2Ftempest~master~I1e2436ba8efeccc2f195ef930554d7f44c9ed29f","project":"openstack/tempest","branch":"master","topic":"qa-2026-2-release","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-05 01:22:02.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I1e2436ba8efeccc2f195ef930554d7f44c9ed29f","subject":"Switch to the master constraint in tox.ini","status":"MERGED","created":"2026-09-04 19:30:48.000000000","updated":"2026-09-08 02:37:03.000000000","submitted":"2026-09-08 02:18:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004030-qa-2026-2-release","meta_rev_id":"e4bea6dad493054765a14384d1920c63e1e5c978","_number":1004030,"virtual_id_number":1004030,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 02:18:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2026-09-05 01:22:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2026-09-05 01:22:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 21:33:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a184bcff0cb9575a396cbf58d5a2b6a9615a65e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 19:30:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6be09af5d98115c2a0ca0dbe17d02ec12b56f09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 21:33:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a111e8a1440c4548bc8d30c514e257b2\n\n- grenade https://zuul.opendev.org/t/openstack/build/24aca5f56c8f4faa9aff7b07d8f283e8 : SUCCESS in 48m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d767c9a8754481aa3942c60b7df15aa : SUCCESS in 1h 42m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd3618fd9b88492c854a1f95b176708d : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bbc7046a96946d889d3d3ef0b224e55 : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce443e9406174a9fa8e6393d6ff95fa5 : SUCCESS in 32m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e21325f395540a4b5bdbdea6dbf9b1c : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f41c56802134773bc0b8bbd33ee8c26 : SUCCESS in 6m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d01d40e039d45f58b1e259049439095 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f639973bd1d44aeb83756c3eb24ec07 : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/618018fd2b4b46feb796cec6bdb1999e : SUCCESS in 8m 58s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/db09af4e2cf2418eb49e6d0e66ff75c0 : SUCCESS in 1h 22m 07s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/764a14e1185a4c568ffdaa8d8d4a4e81 : SUCCESS in 41m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8ec75e3892bc409f852a7131d422323c : SUCCESS in 1h 21m 42s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/3ead7705302d4c14ae59d66a56a79269 : SUCCESS in 1h 46m 29s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/26d7a75cb864415fa22edcbe895f2859 : SUCCESS in 1h 55m 40s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/1e509c2a24cf4b488bfa2d7b90c05001 : SUCCESS in 1h 48m 52s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1b5c91577f534db2b038f539128bd87a : FAILURE in 39m 46s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/50820317e6564048b6fb150c819d66f9 : SUCCESS in 12m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84cf438fc7184279b12103508b26b3f2 : SUCCESS in 1h 06m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/62d38897d00d445aa38991808eee3c7e : FAILURE in 1h 48m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dfe1c2877ae5450e9c88647819f90d98 : SUCCESS in 51m 44s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/4fd530f2da274223a5c79033441e4c09 : SUCCESS in 1h 38m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/655834e20ab3464ca2cb8adeca3fdeb5 : SUCCESS in 1h 28m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b461dc1e792e4675a3fd6c10b97f9320 : SUCCESS in 43m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c458a4be7b454e15999fbec55f4cd219 : SUCCESS in 1h 09m 27s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/d7912f6d477342e78edd62529b40d9a1 : SUCCESS in 1h 06m 07s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/739a5d05045549c592491d2a1216bd42 : SUCCESS in 1h 50m 36s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/8f310eba60934bafba6e0afeccfbf1b1 : SUCCESS in 1h 05m 37s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/bdd47b928b554c86bb6a504b7bfe597c : SUCCESS in 1h 12m 26s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f0c355e9af35415d8ae2d5bbc86931ae : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"aa1875725063bbf51e4242f1a1c12b282d2b2102","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-05 01:22:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c540637c09972ad05adfaabe3f85d9bb239156db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 03:22:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6f2e03de3ac4218acc957ce44693943\n\n- grenade https://zuul.opendev.org/t/openstack/build/f2d808c4f45b42d090b34967451af6fd : SUCCESS in 1h 04m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff2ed5a7a71340a7a210831f89650a21 : SUCCESS in 1h 32m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56d766f370074403ac72d61f8dbf2d62 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9318131754c4e448b5c2393450fb5ec : SUCCESS in 13m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/189319e4c4ad486884e247a4033247e8 : SUCCESS in 25m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/550147e6b46b4488a4191e6eff10b785 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bef54209fc444382ad6daaad5512f492 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7254349e7a649b881998e84c88449d0 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9d76228f8a749fa9a4015b5a33a375a : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/76288e3c34e44aafa1f002da241e54da : SUCCESS in 7m 38s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/353b6b603aeb49f2a59e57a920d8a3f7 : SUCCESS in 1h 18m 00s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/13bebf504081423ca75e707fe113c4f5 : SUCCESS in 54m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a0aaa228c3c342679c9f47a699f2acbb : SUCCESS in 57m 54s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/a571d893bad24cfe9464edbfc46610a2 : SUCCESS in 1h 53m 40s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/b538aa2a356a4b4293975459c6aea302 : SUCCESS in 1h 08m 15s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/8436ceb3140a46a985adaffd2f0895ee : SUCCESS in 1h 38m 29s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/abc45383d0a544fbb17b5acea9dac00c : SUCCESS in 1h 07m 39s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/38ba79d4ae214b149750ac2d11adedb2 : SUCCESS in 12m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de3486ffe1724c5b9220b2c2872ae8bc : SUCCESS in 52m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7a813bf2e19848739bb99a01d75427aa : SUCCESS in 1h 38m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1d4f85a6e8a54bb8850d37c67299467d : SUCCESS in 55m 58s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/e7213764f6c04a709d33d5c9bbef26d5 : SUCCESS in 1h 32m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9c3e04f5d5254056837322462afdff46 : SUCCESS in 1h 28m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb783b741d174738adbbec4cca3e5a7d : SUCCESS in 1h 12m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/374a2d280c844c77b7391e3cb1aaa3ae : SUCCESS in 1h 03m 55s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/24f898dd3e8d45a798319d24d032dbfe : SUCCESS in 1h 19m 32s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/8e0ff4cc50824bb1a35032334346f705 : SUCCESS in 1h 39m 36s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/cfc516f73ca043ef92d74d56c3431b90 : SUCCESS in 31m 10s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a39204d8322c4e1ab3f2c06236e658e1 : SUCCESS in 29m 21s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5d20316e4a384445b481a4c86b52a536 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d59e7f54c5edda257cb240202a99577214d3354e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 00:19:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af104ff08614b3ac11912781e55fa742ef553f1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 02:18:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d80dc8718414da68729db1eb4f65cc6\n\n- grenade https://zuul.opendev.org/t/openstack/build/8c626f043ab04606bad6b97e6181c181 : SUCCESS in 45m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52c911d58a094fbca18a89bd596ea47a : SUCCESS in 1h 44m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea178c32d718485590487c0cf9ced550 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8141e499965c424ba6b1f424b99b60df : SUCCESS in 32m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa030f2abd6f4d0fbf46545279b3de0b : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/126633c73fc44b4b84ef9e172d8840dd : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1dfe26e72424059bd2341123e443b22 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29dd455581cb44a6900934fc90da978a : SUCCESS in 4m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3414420b20424a5ca71e3a1453bfdc5b : SUCCESS in 7m 13s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ce749fc9d444415f95c4ea73e8d97496 : SUCCESS in 1h 27m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd13e561acd14b4ea6a2eb5722bba9de : SUCCESS in 1h 49m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0828bba31011470482d29d46d69da797 : SUCCESS in 1h 10m 50s\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/c6174714dfa54e8893229b2bcca9f1cb : SUCCESS in 1h 34m 01s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/5b88ce8cf1394ce1be92c3d233606753 : SUCCESS in 49m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9d9ba1b84555482989bf438eed6f9b2c : SUCCESS in 33m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de98e625621b410e9526e10f90368975 : SUCCESS in 57m 13s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b8beb782fb394a5f84f9ab9633835650 : SUCCESS in 1h 08m 37s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/5837ff1d98994ed78776eb459c7c01f8 : SUCCESS in 45m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e4c03534b04848e7b8c1f6920f877695 : SUCCESS in 47m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4dd2cb9db005407d912bf418ec62a910 : SUCCESS in 23m 45s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/095ec9cd4bd84dd0a205456e99c81c8b : SUCCESS in 45m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a02cfcfc75f3e6514919557c21030f2a83d1dea3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 02:18:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e4bea6dad493054765a14384d1920c63e1e5c978","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 02:37:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f9fa1d40f904adb8822d1395f942856\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0155b95f6304753bd95451546d6a8e2 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a3f075635c94ecd9f9005313fd01859 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fa3a9de4217b4e006d26320f9f2fb4cc4eb1d7a","revisions":{"1fa3a9de4217b4e006d26320f9f2fb4cc4eb1d7a":{"kind":"REWORK","_number":1,"created":"2026-09-04 19:30:48.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/30/1004030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/1004030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/1004030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/1004030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/1004030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/1004030/1"}}},"commit":{"parents":[{"commit":"8edabfd5976b6aaf0e40649cefd908ccddcb0d48","subject":"Use stable constraint in tox to release new tag for 2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8edabfd5976b6aaf0e40649cefd908ccddcb0d48"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 19:30:23.000000000","tz":-240},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 19:30:23.000000000","tz":-240},"subject":"Switch to the master constraint in tox.ini","message":"Switch to the master constraint in tox.ini\n\nTo release a new tag for Tempest 2026.2, we modified the\ncompatible constraint in tox.in\n\nAs new tag is released now so we can switch back to the master\nconstraint in tox.ini.\n\nChange-Id: I1e2436ba8efeccc2f195ef930554d7f44c9ed29f\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1fa3a9de4217b4e006d26320f9f2fb4cc4eb1d7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1fa3a9de4217b4e006d26320f9f2fb4cc4eb1d7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
