)]}'
{"id":"openstack%2Ftempest~1004412","triplet_id":"openstack%2Ftempest~master~If4cf4fff2bc0c1da14697674ae57178e1cb951fc","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"32704":{"account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"last_update":"2026-09-08 12:02:47.000000000","reason":"\u003cGERRIT_ACCOUNT_32704\u003e replied on the change","reason_account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-09-09 05:04:14.000000000","reason":"Change was submitted"},"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2026-09-09 05:04:14.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-09-07 12:47:16.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-09 03:16:59.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"If4cf4fff2bc0c1da14697674ae57178e1cb951fc","subject":"Temporary skip web-download test","status":"MERGED","created":"2026-09-07 11:40:26.000000000","updated":"2026-09-10 15:26:33.000000000","submitted":"2026-09-09 05:04:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004412","meta_rev_id":"84d3197302a27fc2d04003a88e3ad8d1191d9295","_number":1004412,"virtual_id_number":1004412,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-09 05:04:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-09-08 12:02:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":1,"date":"2026-09-07 11:52:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-09-09 03:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-07 12:24:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-09-09 03:16:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":2,"date":"2026-09-07 11:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-09-09 03:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-07 12:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 11:52:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-09-07 12:24:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-07 12:24:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-09-07 12:26:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-09-07 13:51:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 12:02:34.000000000","updated_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"reviewer":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"0a69755a52f67b2d6dbf051d639a50661a68e03a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-07 11:40:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"554321c978a4e6e20a1da62b6b65603d472a55e0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-07 11:52:01.000000000","message":"Patch Set 1: Code-Review+1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"013517acc5d51b4e0612609e8fb222a802439d41","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 12:24:21.000000000","message":"Patch Set 1: Code-Review+1 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"191c3d20a86bd279e8843943f575feb305858c4e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-07 12:27:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f526673bf4da603e09830fa90be6abf7715d212","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-07 12:30:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4817b6a47d0f72ce749951df45b8327cdebd5137","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-07 12:47:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"096c379ff531ef07571cd8b087ab19efae44d6a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 13:51:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c80b83dd694040198f7c9d6c8c542b95\n\n- grenade https://zuul.opendev.org/t/openstack/build/70a3624e960c431d9e4fb51001beb5c6 : SUCCESS in 1h 04m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1fcd8cf309e42888d215509a14b40b4 : SUCCESS in 1h 36m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70af73c0a6c2401ca5f91cc513edd227 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0452ff37d041e3963d05478ad3ea48 : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ac19c0ba7874b958eb91bd97dfa15c8 : SUCCESS in 35m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4569eb6023e7401ca2bad71aeddaea5d : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb965922d7b6425b8f88c7d39d34604e : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/478b2d4309ab4308ab4f2f8342390680 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd7e7ba3a9484b9a99a416a68cff71f9 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/faddd0d4445d4b0eb05606111c0858ed : SUCCESS in 7m 23s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/6f22331b4729429695c68b55184931eb : SUCCESS in 1h 17m 51s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/9b91a0ce6ed3478e86dccbbabddf7795 : SUCCESS in 58m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/062d0a71b6924416a16356c080795995 : SUCCESS in 1h 25m 48s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/5421310bf2714fbcbbb42d5f55faea24 : SUCCESS in 1h 52m 17s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/e7e93c3a25e043e6a66f7a18da3dbb92 : SUCCESS in 1h 46m 13s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/ad9cb1bf0b8f4a33b69e66ecc3220368 : SUCCESS in 1h 33m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/96ecaad3cfbe46f99d407413b528e8ab : SUCCESS in 1h 06m 11s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/93321e6b62b443639f82442d8d640980 : SUCCESS in 12m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f7a65fba0554cfaab438b56790effcd : SUCCESS in 53m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3b2c250d000b45f4b88a5d0656398e8c : SUCCESS in 1h 47m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/23c7e2dabd40419387529b3dcaede01a : SUCCESS in 41m 58s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/6201fe25d40548e9867445f220104902 : SUCCESS in 1h 21m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3714eab7020d4538bf13bfecac090928 : SUCCESS in 1h 15m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/79af99f57a7543cf8720fa2dcca1caa9 : SUCCESS in 1h 09m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bebdc75dd5a14a0bb4c4f92f969dab1e : SUCCESS in 59m 15s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/11dc10aa366841d8a42a92b2596a536c : SUCCESS in 1h 08m 19s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/c353dd3057e3444987605fb56fb558e0 : SUCCESS in 1h 22m 40s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/23b0722a27cf4d3fa3ce6e1fd9db2bcf : SUCCESS in 58m 21s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9279c7a7da094abe9adcbf0f3526e49e : SUCCESS in 1h 11m 44s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/534d3add532a43c7a34c2ed451c2db77 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a2b064daaf63a2ae31e750e6478cdb93e4377837","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-09-08 12:02:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6699a23a80927523dd6d4e8384a1d7648ed1ebe","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-09-08 12:02:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cbd40cff9a94f370bb39309bea14a7894332b73f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-09 03:16:59.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ee19d2cb1ebebd88f33a0b579de1879e863b526","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 03:18:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbc836fa07247e5086b756443769a68317d9138c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 05:04:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1b98da11aa84b919ac05901757855bd\n\n- grenade https://zuul.opendev.org/t/openstack/build/59bdbdfdc39b42238880cbee7eca5548 : SUCCESS in 1h 09m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97eb4bea8d8947e98b56d0fa6b20ab2f : SUCCESS in 1h 26m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dee063d68ec45d4aefab0f52ffffcc6 : SUCCESS in 13m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/160e495a17dc4129aa7e8b218d90cfa5 : SUCCESS in 35m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de3b1af10a40464a841876d86aea870d : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/456fa86064ee423c9b0ea7f69cbb94e3 : SUCCESS in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fad606602f44421d9328365fe1c42ce9 : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2eadb0c869354ef2b4b63feae8d375b4 : SUCCESS in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c198519580d454988f5b990b67b741a : SUCCESS in 8m 23s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/63e0c1312e334233a281116ae72f8169 : SUCCESS in 1h 27m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/527531e72fbd49d682d456ce4ed0f3a4 : SUCCESS in 1h 12m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/08e5aab163124d35b774ed6a60ddb89f : SUCCESS in 1h 13m 20s\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/faf113213806436baf95c3af7eafe454 : SUCCESS in 1h 15m 30s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/7714e7219e394247bb511c44558100fe : SUCCESS in 1h 09m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db939d6659774b06998354f86039e35e : SUCCESS in 51m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e67d1e9b74a24b9295f9c156062ece87 : SUCCESS in 50m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1b83ee3c9efc4dc89bd7a94d0e8ad0b4 : SUCCESS in 1h 09m 52s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/270d00c05c7248339dfcb981b7b1f1ee : SUCCESS in 1h 05m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0fd2d9abde3247d69a3a3b442b8fade5 : SUCCESS in 1h 34m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/475e72c526264f729c9406299184f33e : SUCCESS in 43m 42s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d5215965a8fc4e289662e9ab487b273f : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2d54c9c31e2d4f26725f204e06a8c98fb8e0d5d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 05:04:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"027695364971a1a5963e43f3d977afb76605aade","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 05:06:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1141e7331724ea39c5679fe7bb3a8a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dac936edd3d4231ae7b14aff77f97f0 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1},{"id":"84d3197302a27fc2d04003a88e3ad8d1191d9295","tag":"autogenerated:gerrit:revert","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-10 15:26:33.000000000","message":"Created a revert of this change as Ib8776c9b0bde155cad22f6c8cd9a873063578e8c","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"323a7562b20eaa859d420d8611e2e656fcd70d55","revisions":{"323a7562b20eaa859d420d8611e2e656fcd70d55":{"kind":"REWORK","_number":1,"created":"2026-09-07 11:40:26.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/12/1004412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/1004412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/1004412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/1004412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/1004412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/1004412/1"}}},"commit":{"parents":[{"commit":"6d858a41e571787dfb6649829b24d0207e043580","subject":"Merge \"Don\u0027t send immutable fields when updating a credential\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6d858a41e571787dfb6649829b24d0207e043580"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-09-07 11:36:52.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-09-07 11:40:17.000000000","tz":180},"subject":"Temporary skip web-download test","message":"Temporary skip web-download test\n\nthis is now sporadically broken due to LP#2166488\nand blocks merging of CVE patches to stable branches.\n\nWill be reverted ASAP after stable branches fixes are merged.\n\nChange-Id: If4cf4fff2bc0c1da14697674ae57178e1cb951fc\nRelated: #2166488\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/323a7562b20eaa859d420d8611e2e656fcd70d55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/323a7562b20eaa859d420d8611e2e656fcd70d55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
