)]}'
{"id":"openstack%2Ftempest~1005072","triplet_id":"openstack%2Ftempest~master~Ib8776c9b0bde155cad22f6c8cd9a873063578e8c","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"32704":{"account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"last_update":"2026-09-15 13:31:43.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-09-10 16:36:46.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2026-09-15 13:31:43.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-09-11 10:26:50.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-11 17:51:50.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-15 13:31:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib8776c9b0bde155cad22f6c8cd9a873063578e8c","subject":"Revert \"Temporary skip web-download test\"","status":"MERGED","created":"2026-09-10 15:26:33.000000000","updated":"2026-09-15 13:32:49.000000000","submitted":"2026-09-15 13:31:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"revert_of":1004412,"submission_id":"1005072","meta_rev_id":"1aaa5527db29b60bc4c588f190b14109d4c9b33c","_number":1005072,"virtual_id_number":1005072,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 13:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2026-09-14 18:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2026-09-14 18:53:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"reviewers":{"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 15:26:33.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-10 15:26:33.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-09-10 15:26:33.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 15:26:33.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-09-10 15:26:33.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-09-10 15:26:33.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"state":"REVIEWER"},{"updated":"2026-09-14 13:58:57.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"}],"messages":[{"id":"922800300195540b0518dc83d2c067272fba6b3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-10 15:26:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c765f2ff19a55b87a00f97ab4786fc1a8975fd02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 17:21:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7d8ac4c4cc24294957b884b571f6fc3\n\n- grenade https://zuul.opendev.org/t/openstack/build/f99f8d204cb8461d98e8a388a78eb6c8 : SUCCESS in 53m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32dc2a5556c6410b82969d1ffa3e7866 : SUCCESS in 43m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb0bae720d43441b8f46050283b75c9a : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4d621933e546aa9671eb9285b028b5 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f692306076bf4b3eb04ad3057b02cb75 : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fce010a8a2140f9b6aa6ce3ad659e53 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0e6de4bad4340a79075d538753e92a0 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c33f401e14804d8ab02b6be96d50387c : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aca95ce468ff45768dbacc30ddbda9e0 : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c6fe8b85b88e4669a9194f4b3be0decf : SUCCESS in 4m 45s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/65d412df84834c87820b8ba7779cad09 : SUCCESS in 44m 53s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/fc8ed0b97451493786584941cc5d65a2 : SUCCESS in 52m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/704110421533491898b5903a68aef250 : SUCCESS in 54m 04s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/67e17f9fa6a84558b1f138a7c52db8aa : FAILURE in 46m 28s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/8b2764c22b7f441d9fbc52dca384240e : FAILURE in 47m 41s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/01201e29b1f741918580f77d37a43553 : SUCCESS in 1h 38m 30s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/11be8f1e98494d6ba1c594558b480d3d : SUCCESS in 33m 16s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3603e1a18f804654b89a3bf15bf4738b : SUCCESS in 12m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bb26e9b952e4008a25afb5ffd8c1357 : SUCCESS in 59m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e6558b375d5c412fa1f5e6ce43337a1b : SUCCESS in 52m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/825d9302511a454593e245ed84d177a9 : SUCCESS in 57m 26s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/e32c70c21e3f4e038b445e0f39140891 : SUCCESS in 1h 47m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/340c0f1a01ab4fab9a9e217a0d36b0a8 : SUCCESS in 49m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/975c4ff0ab244243840200f1188a644e : SUCCESS in 1h 16m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b5dfa9ae7d4f40b6ba8b7ce86d4e3496 : SUCCESS in 1h 05m 29s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/19b081f918fd49a39a604c4f5aac44bb : SUCCESS in 1h 00m 01s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/6382dfd19d0844a182486f58dd933ecf : SUCCESS in 1h 24m 37s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/61365ecd11be4e8cb426149c7b9c3413 : SUCCESS in 1h 03m 38s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/03e9c9fa63ac48288346770cd53fe068 : SUCCESS in 1h 16m 18s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8d8c0f7e9d024dcf9d7c6d2ffd336c03 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4287fdbcb9ce282b461b97c28d27e68f5b15130c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-10 20:33:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a188919684a89e4333288121787f9831354ed4c6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-11 10:26:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eabe56ecbfd6bfd0df16a07cff6cee5f388f7102","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-09-11 11:38:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c70c41bc2172caf690fbc6dc9e13cacb4f8fc9b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-11 17:51:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0881ed877fc926fb1e5b7061cf5b9f29776caa02","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-12 13:29:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3b247ef50a6021bab1f4ec28147f5d9eaceafc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 15:39:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/708a41267fd44d14960eddc105378d17\n\n- grenade https://zuul.opendev.org/t/openstack/build/934cd93e6b3049bf9ccf4fcc3fc7c41d : FAILURE in 1h 45m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab3bde2ad1ff4a01a1b0ffa73e831481 : SUCCESS in 42m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f981fb968a8f4a7a93e55509df4d20a0 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62aa945f6ff14955a7c03e26ecc2d2e4 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d716dfc3ea8c4f6abd38ffe2703c76fe : SUCCESS in 26m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1309bee42464323bc0db1248654d86b : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c7425fc65114fdeb57377e63a15eb76 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1dd0a5434917408dadf3198de08b2387 : SUCCESS in 5m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3469b57bb6944d09a72061ab43083720 : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cc90988ad2374907ba3fbafcf3233b21 : SUCCESS in 7m 51s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4a68e20e63ee4e2c9d141dc33f5a21a5 : SUCCESS in 1h 30m 33s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/63d1b12c132945d1a2dad9bb5358e892 : SUCCESS in 1h 03m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8e2f792cb02f44c7a6334e05a0e08905 : SUCCESS in 1h 57m 29s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/47a28d20d767432382ef5e70a3c507f6 : SUCCESS in 2h 03m 48s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/57f2d38fcb054a748afca8066305d170 : SUCCESS in 1h 54m 23s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/454b7e66451745ce8666572a3d093d38 : SUCCESS in 1h 29m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fc98764d6e0b4675a2acee22a427bc55 : SUCCESS in 1h 17m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a12612ad170e450e8d58ae40abd091d3 : SUCCESS in 11m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae9e161f2463452db331a7dafe002795 : SUCCESS in 1h 08m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d9c1bee8a83940acbd9f2bc2308ab878 : SUCCESS in 1h 45m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fd50e9b7c30a46a49eb46eb4d5426cd0 : SUCCESS in 46m 32s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/24710a757263446dac2dbe9c0b6501bd : SUCCESS in 1h 44m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/217769b8a8bc4ad783d30e33c5ae55df : SUCCESS in 1h 05m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c182d6c5c5a945a4bc4b4339dcadd04a : SUCCESS in 39m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ffb1ca5cac7c4d64a99788d3d8739a09 : SUCCESS in 50m 42s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/3135c366ad95440186d2a60775e15c6c : SUCCESS in 46m 46s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/83d85705bb884bed858eab9fdb045982 : SUCCESS in 38m 37s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/4c4059ba93ea4ea9b9c064002f229c61 : SUCCESS in 1h 00m 14s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/24e3b347adcc4909b520716790f504ac : SUCCESS in 34m 21s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/214869bf9fa943eca12b139a3d8ec502 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0c3334330559cd6bf9d9eea878ecaac72fc7afff","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-09-14 13:58:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ed3b83a7f0834f61e5702faca17a8cbe54901c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 16:01:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9a0081c10cc451a94bd78cfe4e918d3\n\n- grenade https://zuul.opendev.org/t/openstack/build/f2c99ca3c8c74dd1b67c88f206070246 : SUCCESS in 1h 07m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2ad0adb6b5f4e409e519b90696ca32b : SUCCESS in 1h 19m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10100814e3b3410dab92e9e3f7a596e5 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51d328092c8f4ecd87c735ca33178083 : SUCCESS in 13m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c2b0f6aec6740c3adfa41b690637e4b : SUCCESS in 25m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a30d9a02344426bae3975651bbc2220 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35771a988aa94e9085d057e17d5990f5 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fce938fd504b4c5a9d93047d004aaee8 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4aa9936b2f6149b399edc25a2d1a5817 : SUCCESS in 5m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c82313ca5f24339ada169056bac0925 : SUCCESS in 7m 26s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/aa766b662e774b4daa79c3f35a106d84 : SUCCESS in 59m 16s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/fa133a7d14194821b6cc97bdafde6ed5 : SUCCESS in 54m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/889870080168416397d3731bbc1771ea : SUCCESS in 1h 10m 23s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/b426d137b156415784dacae51365dbcf : SUCCESS in 1h 44m 39s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/a4e1ce799ccb4d44a7bb87269383ec98 : SUCCESS in 53m 46s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/998fac56d29c48c0bb266f6afc3b18fd : SUCCESS in 1h 54m 29s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/142699bed6cf4f06a40aeb6df302e35c : SUCCESS in 1h 06m 16s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/155abc3a9e1c4112a31d378706ea772d : SUCCESS in 5m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0df118e283c64ae18ba453a2714a62ad : SUCCESS in 33m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a92498071ac4823b113f61743d0cfc2 : SUCCESS in 38m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9329b5b9b6064c24b8048fd2636a7e0a : SUCCESS in 53m 52s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/604d08fa8c3f4dd1ae381b777d482b43 : SUCCESS in 1h 37m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6bb1ad061e3b4e5f91ebf86a8b5b10b0 : SUCCESS in 1h 29m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3d8efbcffa24c768f70f051d7c0a399 : SUCCESS in 1h 04m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ee570ab72a5431bb074fd2b4f8a5ecc : SUCCESS in 1h 06m 06s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/0d107b433c5c46548597602ee2c2adf6 : SUCCESS in 1h 03m 23s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/6101b23d8f6044a88d4b842b9f84afb2 : SUCCESS in 1h 25m 44s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/efaaf45e77674f41b8a5d2089da46dfe : SUCCESS in 1h 18m 22s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3b14bbadd37e4daf84dd91d3c17c6802 : SUCCESS in 1h 26m 18s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c4228b6d53e146e28411b015fd0d6e12 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"48caab07bd158facaaca11d49e8b0054ba6630f3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-14 18:53:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0ac3f6f6cdb9f28bfb084177396c71166dd11cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 18:54:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09e7bf21d0adaff9b9b9cad0fd0081c6fb2a019f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 20:22:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb96e5a7652141c58eb3088644b7b70b\n\n- grenade https://zuul.opendev.org/t/openstack/build/15b7934b5cab44d6b90cde4d91ac28fc : SUCCESS in 1h 03m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b865d5f1b37d4a0fad40fd02c46efe00 : SUCCESS in 1h 06m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df7127c4f8ed4caeafe2eca6d4d2fe69 : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74402ed8766d480bb14615e4adac21a8 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/673ece7f6e1e4f659a07a9ce40ef94bf : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/526ba300cdf54c7eb49c498eb761d558 : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6df21734ca444487a7fe2f2dde48469f : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ca02d2a154547dc97b68f03b415118d : SUCCESS in 7m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/76df0e3f5012412082cb9f3d164dec5c : SUCCESS in 4m 03s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c4fc196b0e30459b83c4fa76313c9b1e : SUCCESS in 1h 09m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9256b5ebf2374a42b6ee8cee536903ce : SUCCESS in 1h 23m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a74ea7d46ef4447ea2722e48b83a7a77 : SUCCESS in 1h 10m 58s\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/462820faf723460bb4390df0a16eef93 : FAILURE in 37m 08s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/98c41055bb3941a3b84a4b453d6ff02e : SUCCESS in 1h 01m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/073e5b5ed2024a0b8393cb2d6e204ad2 : SUCCESS in 1h 02m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a461db3481054137af85ec598be78cb1 : SUCCESS in 55m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7aff89dbd2374a72a6e5a34eef8e4886 : SUCCESS in 1h 08m 06s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/7f94f0086f934d2bba004d9710a77800 : SUCCESS in 43m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e56595018c934d269982ee0bcfdd5001 : SUCCESS in 51m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4e282412a2fc4a7bab773397e20d4acd : SUCCESS in 49m 53s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/8ef2c493cd3f459ba1d4e52fe3cc077e : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8932a9da5caf027eab11017092d4cee8ec313055","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-15 09:25:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"534b3bd25514cddf76e425f3261b7f1b9ad38e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 11:42:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7678569da1e74205af82a9e841452c14\n\n- grenade https://zuul.opendev.org/t/openstack/build/85a17752e84845a89b73788979abaff4 : SUCCESS in 1h 03m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85f4aae9b44846c3aa7a2c3d1d217eec : SUCCESS in 2h 05m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d35a464a657c4edf97c3780a8ecb1051 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50415edf84ce448087c01c03fef8b713 : SUCCESS in 13m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3e47d5d436941c49927e960a14fc104 : SUCCESS in 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf4fd879a4bf464ebb8c7d2b688bbc4e : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e5e2a655cf24b4aaecabcc4e7143617 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fbdf287fe8c94d969c263e523f7cbdea : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0034969e7ca4b4fa9394e04a5d2638a : SUCCESS in 4m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/28eb6b856e0e4397b1641c60ad5364be : SUCCESS in 9m 32s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/daeb8d84ab664466b322a38fca542993 : SUCCESS in 1h 30m 51s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/1045ddc017894cbbb8326ee259c19d3c : SUCCESS in 40m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cb617560102c4806aa3f08d07a40cbdc : SUCCESS in 1h 45m 41s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/da9b667aa6b24631a3f9e054dc83eaa0 : SUCCESS in 1h 06m 42s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/831deaa9d418438db17aa4b199e87ca7 : SUCCESS in 1h 39m 46s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/441bd5c32a5e4d0abf770fd4d4e65d7b : SUCCESS in 1h 37m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/11e16a8332b745238c161c64f0d94caf : SUCCESS in 40m 37s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a6ff2e391834491da17e784cbebb0cb2 : SUCCESS in 12m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74a5c4a8d5ce414680780539acea2915 : SUCCESS in 1h 02m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7677b5fc151545d9b8f67c19342afe58 : SUCCESS in 54m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/35167996dbf7458ab44f16f8d74c2c56 : SUCCESS in 57m 01s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/f52aa960cefc4ac1bdec40e80aca81fe : SUCCESS in 58m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ec84ac3f54ae45d58552ae194c7be09c : SUCCESS in 1h 19m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/75af1a53909b45ae9b4679c50f9bc601 : SUCCESS in 1h 06m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/02765084ef614897b9bae54393b0ed0f : SUCCESS in 43m 10s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/c0f036f438344d5aaa57dc4619229832 : SUCCESS in 48m 19s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/55b3593b69704328b311b4a55ff07910 : SUCCESS in 1h 37m 30s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/bd694a7af18040eeb8c7d0a7902947d2 : SUCCESS in 1h 03m 19s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/190b30ef0ec545b4a3c72205e6cc407e : SUCCESS in 30m 50s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/99f95fd2121649adba8f6a145d3dfa68 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"79afa045a6529e827b90553402d7b69ba6e4c17d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 11:43:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ef945ae5c4002fad2cf5208843b821de5368147","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 13:31:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/049bac5455724d4d96ad157ec0be66e9\n\n- grenade https://zuul.opendev.org/t/openstack/build/d42a58c67c034ea58ec378a66e250455 : SUCCESS in 1h 00m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10864e566f274a81a0079164a26ea963 : SUCCESS in 1h 34m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6bbeda2fd945788684c740e353d666 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddb8be31b70a462d91bbba6402ff4e80 : SUCCESS in 30m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a8bb879f1640a983c0effe6c0d4d67 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/06afcc0e80574816beec2e6c9b1b1fc2 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57293a123edd42b6a1324883dca85d4c : SUCCESS in 6m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/caf29f8ec3d944c99740114406efaf41 : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/abdbaac16ae744429e06a59cb1e5ea0a : SUCCESS in 7m 30s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/8424584e94f449f8a651330e36652eee : SUCCESS in 1h 04m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7668f85a995c4bbda6a9eef4b9a2091f : SUCCESS in 1h 40m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf38a64e9c2d42ecb9b3d00b58e6250b : SUCCESS in 1h 11m 41s\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/cf2cf7d31eff47c6b27f402b4bc03e28 : SUCCESS in 1h 35m 23s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/9da74c68c76c49b7a9793a4b10dacc80 : SUCCESS in 55m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c88f78e88aaf4551be66c173ea806792 : SUCCESS in 1h 01m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4b27023e292447a802a3fcf918b02c6 : SUCCESS in 27m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fd9bcc4cec664bf1b652b105f29b1e75 : SUCCESS in 35m 40s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/88599ee52bcd4b54a9f99fe6193c9984 : SUCCESS in 42m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2edfad7d30344c66a5f6ca3bb35184ec : SUCCESS in 1h 43m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b6e269c86ae74d729409b039bd628e3c : SUCCESS in 24m 18s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/dab9c37fffb74537af63337c4592e2ff : SUCCESS in 1h 00m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7545b24e663815cc8cc56e86e3b221aa12df82f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 13:31:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1aaa5527db29b60bc4c588f190b14109d4c9b33c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 13:32:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73543ca20ea34e589ea29d8fc3884dd4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c1bce144eb646a9899a7856816d94af : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb96e36b70ff0879c082561d1bcc6e91c816f34a","revisions":{"eb96e36b70ff0879c082561d1bcc6e91c816f34a":{"kind":"REWORK","_number":1,"created":"2026-09-10 15:26:33.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/72/1005072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/72/1005072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/72/1005072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/72/1005072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/72/1005072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/72/1005072/1"}}},"commit":{"parents":[{"commit":"323a7562b20eaa859d420d8611e2e656fcd70d55","subject":"Temporary skip web-download test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/323a7562b20eaa859d420d8611e2e656fcd70d55"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","date":"2026-09-10 15:26:33.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2026-09-10 15:26:33.000000000","tz":0},"subject":"Revert \"Temporary skip web-download test\"","message":"Revert \"Temporary skip web-download test\"\n\nThis reverts commit 323a7562b20eaa859d420d8611e2e656fcd70d55.\n\nReason for revert: patches to stable branches have been merged\n\nRelated: #2166488\nChange-Id: Ib8776c9b0bde155cad22f6c8cd9a873063578e8c\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eb96e36b70ff0879c082561d1bcc6e91c816f34a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eb96e36b70ff0879c082561d1bcc6e91c816f34a"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
