)]}'
{"id":"openstack%2Ftempest~1006015","triplet_id":"openstack%2Ftempest~master~Iae9131cd49ea5990cd27a35c617b95b0fb314c3e","project":"openstack/tempest","branch":"master","topic":"bug/2133818","attention_set":{"36743":{"account":{"_account_id":36743,"name":"ranikim","display_name":"jihyeonkim","email":"jh011.kim@samsung.com","username":"ranikim"},"last_update":"2026-09-21 13:19:22.000000000","reason":"\u003cGERRIT_ACCOUNT_32704\u003e replied on the change","reason_account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}}},"removed_from_attention_set":{"32704":{"account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"last_update":"2026-09-21 13:19:22.000000000","reason":"\u003cGERRIT_ACCOUNT_32704\u003e replied on the change","reason_account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}}},"hashtags":[],"change_id":"Iae9131cd49ea5990cd27a35c617b95b0fb314c3e","subject":"Log config file only when it exists","status":"NEW","created":"2026-09-17 11:26:02.000000000","updated":"2026-09-21 13:19:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa356246d6360fc39f4382faa1e69d23e8df4a56","_number":1006015,"virtual_id_number":1006015,"owner":{"_account_id":36743,"name":"ranikim","display_name":"jihyeonkim","email":"jh011.kim@samsung.com","username":"ranikim"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-17 13:46:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 13:46:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 13:10:41.000000000","updated_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"reviewer":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"31dab2321adeb3e3216ddac96caff6dea44ab0a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36743,"name":"ranikim","display_name":"jihyeonkim","email":"jh011.kim@samsung.com","username":"ranikim"},"date":"2026-09-17 11:26:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ce1369f9eae96651e32eb1d0ab85f8203e45d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 13:46:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ec17e864e4d4bfdb7c70318cea3edd4\n\n- grenade https://zuul.opendev.org/t/openstack/build/0d3fb4334ebf4bb886f2a26535c7bace : SUCCESS in 1h 04m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc28b33ee3f343e592d219fdf6a60cc9 : SUCCESS in 1h 26m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b08a73283bbb40378ff51b0ab396747d : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df071ac742ba41949e8cf049426bd963 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ff8fb023c474dfd9e616a40672885e4 : SUCCESS in 32m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a39b4e94a5ab4aa893dac78e2aef7062 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6e5b090d18f49ea93e247a2073d9c1d : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cebbb7a1c83042c88012484f2d908917 : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b43c13dc377496aad227e0457fe9a7e : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/82f1a42cd8ab4d7fbd9b9675a2935374 : SUCCESS in 6m 45s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5d807bb54fbd49cba2afb4ff13a22b10 : SUCCESS in 1h 06m 34s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/5ca946328aa841b49f47d2545a2a440c : SUCCESS in 1h 05m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/df321c65127e402f8f2852c6455742d2 : SUCCESS in 2h 14m 41s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/d4486225ada9473d9872d615aa04af6b : SUCCESS in 1h 42m 40s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/0bf9ccae460c4a5697dcf961b47f022c : SUCCESS in 1h 46m 38s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/c5b345ff9daf408f8258c030483967b1 : SUCCESS in 1h 44m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1ab6e0bcbf674b229bf498c438dbce68 : SUCCESS in 1h 02m 47s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/5b6e100928624b9fb64534c043a80892 : SUCCESS in 16m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b26d91378130433ab1be12103b6c4e48 : SUCCESS in 52m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bdc684b1f6174c0882122b8d746862c3 : SUCCESS in 1h 40m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4e1682892d444575bdec4395d9a7f2c7 : SUCCESS in 47m 43s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/bd687001d918483992d0f05c25eb1284 : SUCCESS in 1h 31m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/772197d14bd24dada15d085ed0038440 : SUCCESS in 1h 35m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4667f9b9e26a4e608ae1e07561e2bf37 : SUCCESS in 1h 04m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8bf7f04bd19343dda67ed966d9df2fa5 : SUCCESS in 1h 02m 49s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/680e338e47c2483f8720c96b2138dbba : FAILURE in 54m 35s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/649497293f3a4b70a2a94679a10a9013 : SUCCESS in 1h 01m 49s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/7a9a4f9e07fa416393d0e6d2fe154516 : SUCCESS in 1h 12m 58s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/64044ccdb18a4baeaedbe9c300bf087b : SUCCESS in 1h 06m 23s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e4d0984df5504dffa0a0d0d3f5c99d07 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"fa356246d6360fc39f4382faa1e69d23e8df4a56","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-09-21 13:19:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3583a8fb18fc1e2caf258a1fd2950e4578f4ca7","revisions":{"c3583a8fb18fc1e2caf258a1fd2950e4578f4ca7":{"kind":"REWORK","_number":1,"created":"2026-09-17 11:26:02.000000000","uploader":{"_account_id":36743,"name":"ranikim","display_name":"jihyeonkim","email":"jh011.kim@samsung.com","username":"ranikim"},"ref":"refs/changes/15/1006015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/15/1006015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/1006015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/1006015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/1006015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/15/1006015/1"}}},"commit":{"parents":[{"commit":"08d2d53f6aca8c760b5e8581cea5656259d1ce58","subject":"Merge \"Changed some more Neutron code to use project_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/08d2d53f6aca8c760b5e8581cea5656259d1ce58"}]}],"author":{"name":"jihyeon.kim","email":"jh011.kim@samsung.com","date":"2026-09-17 11:21:25.000000000","tz":540},"committer":{"name":"jihyeon.kim","email":"jh011.kim@samsung.com","date":"2026-09-17 11:21:25.000000000","tz":540},"subject":"Log config file only when it exists","message":"Log config file only when it exists\n\nWhen no config file is found, Tempest falls back to\n/etc/tempest/tempest.conf but still logs that it is using that file.\nSkip the message when the fallback file does not exist.\n\nCloses-Bug: #2133818\nChange-Id: Iae9131cd49ea5990cd27a35c617b95b0fb314c3e\nSigned-off-by: jihyeon.kim \u003cjh011.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c3583a8fb18fc1e2caf258a1fd2950e4578f4ca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c3583a8fb18fc1e2caf258a1fd2950e4578f4ca7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
