)]}'
{"id":"openstack%2Ftempest~1007666","triplet_id":"openstack%2Ftempest~master~I618cd104ca3c481374d35581460142838fb7c4ca","project":"openstack/tempest","branch":"master","topic":"bug/2166205","attention_set":{"36743":{"account":{"_account_id":36743,"name":"ranikim","display_name":"jihyeonkim","email":"jh011.kim@samsung.com","username":"ranikim"},"last_update":"2026-09-30 12:09:45.000000000","reason":"\u003cGERRIT_ACCOUNT_32704\u003e replied on the change","reason_account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}}},"removed_from_attention_set":{"32704":{"account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"last_update":"2026-09-30 12:09:45.000000000","reason":"\u003cGERRIT_ACCOUNT_32704\u003e replied on the change","reason_account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}}},"hashtags":[],"change_id":"I618cd104ca3c481374d35581460142838fb7c4ca","subject":"Handle paginated Cinder user messages","status":"NEW","created":"2026-09-28 08:34:04.000000000","updated":"2026-09-30 12:09:45.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"e8493edd7c536c7566e86bc9ed7e06ec04f55f4f","_number":1007666,"virtual_id_number":1007666,"owner":{"_account_id":36743,"name":"ranikim","display_name":"jihyeonkim","email":"jh011.kim@samsung.com","username":"ranikim"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-28 10:45:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-28 10:45:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-30 12:05:50.000000000","updated_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"reviewer":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"8b68720b264c196285c1ec651f8bbbd7c71d817b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36743,"name":"ranikim","display_name":"jihyeonkim","email":"jh011.kim@samsung.com","username":"ranikim"},"date":"2026-09-28 08:34:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08fe387c83ecc9d3325925c360f4fd31c45df8c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 10:45:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a750734d7424e36a220a8165a64329e\n\n- grenade https://zuul.opendev.org/t/openstack/build/2f46b2db7b4e4b66b5ed9761452bdf7a : SUCCESS in 31m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f84b72b3baf84904acca6db46b82eb48 : SUCCESS in 1h 24m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ecc8d99fd684262b80ca1f0104307b9 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d8b423168d45e381352b586b38a488 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/772175d164fc48fb9932d67cfd4f362b : SUCCESS in 33m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2d87df6f7204bda8923f4ef244ad28c : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4af10cce1e34480bbf635fc1496dfb51 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/943a50c90abb445b923d4183d79a8b3f : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ce009e570634bb397b3d0bed8e95695 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/970e290d198b4f84a7548b5744fe6283 : SUCCESS in 4m 09s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/64bf9e3cc31048378443011f571b7884 : SUCCESS in 41m 35s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/38a4bf82760e460bad32029c5685f1d0 : SUCCESS in 50m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/399323725cfc42a381b6d2909897f52f : SUCCESS in 2h 03m 47s (non-voting)\n- tempest-full-2026-2 https://zuul.opendev.org/t/openstack/build/7f114d7ced80463e8eb10d81e97f740e : SUCCESS in 1h 36m 50s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/d5eb456afb1b4c4187876acd26159b98 : SUCCESS in 1h 54m 42s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/11537dc8bcca4374864ac813477cc9bb : SUCCESS in 1h 24m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cf772f67e69f498ea4ba01f22aa51ab5 : SUCCESS in 1h 02m 15s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/5c99a69416d6470b892f6b642ebfe1db : FAILURE in 4m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccb33348027e45aca8988193adba07d9 : SUCCESS in 1h 03m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7a827affe14e458398a1542b11f752de : SUCCESS in 59m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7385f06c4d8140d79c529f19188f9065 : SUCCESS in 43m 13s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/f1155c79af6b4a83b5b515a4f7b10cf0 : SUCCESS in 1h 45m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7d926d03ea724bfdafb55adf3fdb0a59 : SUCCESS in 1h 42m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9759a11cf674b1fa4a0f033cfb132ec : SUCCESS in 42m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/18b7f0b2f4db40ab95e623f30fadb85d : SUCCESS in 1h 14m 23s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/0e655e00b5524d4abef14fd46f09c719 : SUCCESS in 1h 14m 40s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/6c3cbb72b05e4d34b50b46ada814372a : SUCCESS in 1h 33m 48s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/6f3d25ec9109433da98f5d50cbaf1f7c : SUCCESS in 1h 06m 37s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4cdc1448464c44319d1ce00d50fbfb91 : SUCCESS in 1h 07m 06s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1cb155b7086848fda3dbf4304b399d60 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4596ea8c77da2257b9a01361ad65d084cdb60687","author":{"_account_id":36743,"name":"ranikim","display_name":"jihyeonkim","email":"jh011.kim@samsung.com","username":"ranikim"},"date":"2026-09-29 01:37:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29f7377c11977aca4de19d633fcf5f49c03a41d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 03:25:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1237a52d92414403af36a23e22c14e72\n\n- grenade https://zuul.opendev.org/t/openstack/build/8e78b286869443f888863ed1bba787f7 : SUCCESS in 36m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e51e0699ae943e4873f9698c4372058 : SUCCESS in 1h 10m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5078823780ec4ee1a24cb4ce5ad955d9 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c39edc41e3b649f6982d3041cac07d25 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdcb926ef9bb493486315b70e2785fd6 : SUCCESS in 33m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d520e4c54fc948839994e011d3b5fae8 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d994a308252647c5906a214ad7cb312b : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15dc4f9ee6b44c03bcd60fe965045ad9 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9e6bcf7b758475d94772902176b2e08 : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8a7facbe993e4c569c33340573bb132d : SUCCESS in 6m 24s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/afb8c9165fcb4217ae0274df3fcedc6a : SUCCESS in 39m 06s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/f78c614e29bf4c168f8f08d6b47ee5c3 : SUCCESS in 53m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/29ce17890f4c454c8aba9d80d57ccc19 : SUCCESS in 1h 22m 52s (non-voting)\n- tempest-full-2026-2 https://zuul.opendev.org/t/openstack/build/48619eb28dcc4ef3ae7fe9e3a48b7ffa : SUCCESS in 1h 42m 15s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/32e665fc6af343c1861b518a692331b5 : SUCCESS in 1h 15m 44s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/6b2c4201ebe94db28cc2bc749eac13ce : SUCCESS in 1h 14m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3239e0fe1fe84e68a5e548c1d8f498b7 : SUCCESS in 1h 11m 14s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/1bf011b2400d4569bf32f1cfa69695de : FAILURE in 7m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da90f281bff6491286a9c0fc868a231b : SUCCESS in 34m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/24abaa75733542be9ad126c20494c8ef : SUCCESS in 43m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c1a058a9c97a4ae49231f9ab679fde4d : SUCCESS in 34m 41s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/d7ad84d82d0847c3bdff70dad694063a : SUCCESS in 41m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/399a2cb4316f4666885bbb440dc00dd5 : SUCCESS in 54m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/46290f12f4cf4a67aee3146e3d3e9854 : SUCCESS in 35m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a4355bdf9dea46328d597921a9238a60 : SUCCESS in 30m 14s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/94520766e3e54194ab237f4983b191aa : SUCCESS in 34m 17s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/1ee1a307358940fda373118fdcca1553 : SUCCESS in 1h 21m 54s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/316b76a4e9b741c38c77f2de07c70e1a : SUCCESS in 1h 12m 17s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b80e7dcd436e4439aaf0588cc28ce220 : SUCCESS in 1h 03m 53s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5f6996193ccd4fd8b4649aa3ce793e7e : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3feab00a3dd658bdb4498bb26162de2f9e872486","author":{"_account_id":36743,"name":"ranikim","display_name":"jihyeonkim","email":"jh011.kim@samsung.com","username":"ranikim"},"date":"2026-09-29 03:37:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e8493edd7c536c7566e86bc9ed7e06ec04f55f4f","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-09-30 12:09:45.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75fcdee82ddf6f8d4f9b59eb5720ff1dcb2d28da","revisions":{"75fcdee82ddf6f8d4f9b59eb5720ff1dcb2d28da":{"kind":"REWORK","_number":1,"created":"2026-09-28 08:34:04.000000000","uploader":{"_account_id":36743,"name":"ranikim","display_name":"jihyeonkim","email":"jh011.kim@samsung.com","username":"ranikim"},"ref":"refs/changes/66/1007666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/66/1007666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/1007666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/1007666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/1007666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/66/1007666/1"}}},"commit":{"parents":[{"commit":"5692e968f9e26bd86559efa765f6e5f47f6bf72d","subject":"Merge \"Add Tempest gate job for stable/2026.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5692e968f9e26bd86559efa765f6e5f47f6bf72d"}]}],"author":{"name":"jihyeon.kim","email":"jh011.kim@samsung.com","date":"2026-09-28 08:37:36.000000000","tz":540},"committer":{"name":"jihyeon.kim","email":"jh011.kim@samsung.com","date":"2026-09-28 08:38:50.000000000","tz":540},"subject":"Handle paginated Cinder user messages","message":"Handle paginated Cinder user messages\n\nCinder includes messages_links when a message list is paginated.\nAllow the field in the response schema and follow subsequent pages\nwhen looking for the message created by the test.\n\nCloses-Bug: #2166205\nChange-Id: I618cd104ca3c481374d35581460142838fb7c4ca\nSigned-off-by: jihyeon.kim \u003cjh011.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/75fcdee82ddf6f8d4f9b59eb5720ff1dcb2d28da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/75fcdee82ddf6f8d4f9b59eb5720ff1dcb2d28da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
