)]}'
{"id":"openstack%2Ftempest~117193","triplet_id":"openstack%2Ftempest~master~Ib6de34fc0d86dc2a1b7859e4d3dc11488799809d","project":"openstack/tempest","branch":"master","topic":"bp/client-checks-success","hashtags":[],"change_id":"Ib6de34fc0d86dc2a1b7859e4d3dc11488799809d","subject":"Add client response checking for object storage service","status":"ABANDONED","created":"2014-08-27 11:17:22.000000000","updated":"2016-06-03 05:09:19.000000000","total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f4a7cb48c884f5868a5eaeaa2cd25a46c8602ffa","_number":117193,"virtual_id_number":117193,"owner":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"all":[{"_account_id":1216,"name":"gholt","email":"z-launchpad@brim.net","username":"gholt"},{"_account_id":512,"name":"Mike Barton","username":"5UOzf3BLEHnnEzgp","inactive":true},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":866,"name":"Chmouel Boudjnah","email":"chmouel@chmouel.com","username":"chmouel"},{"value":-1,"date":"2014-09-29 10:01:57.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},{"date":"2014-09-08 02:22:32.000000000","_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"_account_id":2696,"name":"Darrell Bishop","email":"dbishop@nvidia.com","username":"darrellb"},{"date":"2014-09-10 02:45:40.000000000","_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":2828,"name":"Michael Barton","email":"mike@weirdlooking.com","username":"redbo"},{"value":-1,"date":"2014-09-08 01:21:30.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":1009,"name":"Greg Lange","email":"greglange@gmail.com","username":"greglange"},{"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},{"_account_id":792,"name":"OpenStack Hudson","email":"hudson@openstack.org","username":"hudson-openstack"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"recommended":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"disliked":{"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1216,"name":"gholt","email":"z-launchpad@brim.net","username":"gholt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":512,"name":"Mike Barton","username":"5UOzf3BLEHnnEzgp","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":866,"name":"Chmouel Boudjnah","email":"chmouel@chmouel.com","username":"chmouel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"value":-1,"date":"2014-09-11 21:44:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2014-09-08 20:03:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2696,"name":"Darrell Bishop","email":"dbishop@nvidia.com","username":"darrellb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2828,"name":"Michael Barton","email":"mike@weirdlooking.com","username":"redbo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1009,"name":"Greg Lange","email":"greglange@gmail.com","username":"greglange"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":792,"name":"OpenStack Hudson","email":"hudson@openstack.org","username":"hudson-openstack"},{"value":-1,"date":"2014-11-14 22:31:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":1,"date":"2014-09-09 00:34:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"all":[{"_account_id":1216,"name":"gholt","email":"z-launchpad@brim.net","username":"gholt"},{"_account_id":512,"name":"Mike Barton","username":"5UOzf3BLEHnnEzgp","inactive":true},{"value":-1,"date":"2014-09-23 21:17:34.000000000","_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":866,"name":"Chmouel Boudjnah","email":"chmouel@chmouel.com","username":"chmouel"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"_account_id":2696,"name":"Darrell Bishop","email":"dbishop@nvidia.com","username":"darrellb"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":2828,"name":"Michael Barton","email":"mike@weirdlooking.com","username":"redbo"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":1009,"name":"Greg Lange","email":"greglange@gmail.com","username":"greglange"},{"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},{"_account_id":792,"name":"OpenStack Hudson","email":"hudson@openstack.org","username":"hudson-openstack"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1216,"name":"gholt","email":"z-launchpad@brim.net","username":"gholt"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":512,"name":"Mike Barton","username":"5UOzf3BLEHnnEzgp","inactive":true},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":866,"name":"Chmouel Boudjnah","email":"chmouel@chmouel.com","username":"chmouel"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":2696,"name":"Darrell Bishop","email":"dbishop@nvidia.com","username":"darrellb"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":2828,"name":"Michael Barton","email":"mike@weirdlooking.com","username":"redbo"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1009,"name":"Greg Lange","email":"greglange@gmail.com","username":"greglange"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":792,"name":"OpenStack Hudson","email":"hudson@openstack.org","username":"hudson-openstack"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":512,"name":"Mike Barton","username":"5UOzf3BLEHnnEzgp","inactive":true},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":792,"name":"OpenStack Hudson","email":"hudson@openstack.org","username":"hudson-openstack"},{"_account_id":866,"name":"Chmouel Boudjnah","email":"chmouel@chmouel.com","username":"chmouel"},{"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},{"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"_account_id":1009,"name":"Greg Lange","email":"greglange@gmail.com","username":"greglange"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"_account_id":1216,"name":"gholt","email":"z-launchpad@brim.net","username":"gholt"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},{"_account_id":2696,"name":"Darrell Bishop","email":"dbishop@nvidia.com","username":"darrellb"},{"_account_id":2828,"name":"Michael Barton","email":"mike@weirdlooking.com","username":"redbo"},{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":512,"name":"Mike Barton","username":"5UOzf3BLEHnnEzgp","inactive":true},"reviewer":{"_account_id":512,"name":"Mike Barton","username":"5UOzf3BLEHnnEzgp","inactive":true},"state":"REVIEWER"},{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":792,"name":"OpenStack Hudson","email":"hudson@openstack.org","username":"hudson-openstack"},"reviewer":{"_account_id":792,"name":"OpenStack Hudson","email":"hudson@openstack.org","username":"hudson-openstack"},"state":"REVIEWER"},{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":866,"name":"Chmouel Boudjnah","email":"chmouel@chmouel.com","username":"chmouel"},"reviewer":{"_account_id":866,"name":"Chmouel Boudjnah","email":"chmouel@chmouel.com","username":"chmouel"},"state":"REVIEWER"},{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},"reviewer":{"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},"state":"REVIEWER"},{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},"reviewer":{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},"state":"REVIEWER"},{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":1009,"name":"Greg Lange","email":"greglange@gmail.com","username":"greglange"},"reviewer":{"_account_id":1009,"name":"Greg Lange","email":"greglange@gmail.com","username":"greglange"},"state":"REVIEWER"},{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":1216,"name":"gholt","email":"z-launchpad@brim.net","username":"gholt"},"reviewer":{"_account_id":1216,"name":"gholt","email":"z-launchpad@brim.net","username":"gholt"},"state":"REVIEWER"},{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":2696,"name":"Darrell Bishop","email":"dbishop@nvidia.com","username":"darrellb"},"reviewer":{"_account_id":2696,"name":"Darrell Bishop","email":"dbishop@nvidia.com","username":"darrellb"},"state":"REVIEWER"},{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":2828,"name":"Michael Barton","email":"mike@weirdlooking.com","username":"redbo"},"reviewer":{"_account_id":2828,"name":"Michael Barton","email":"mike@weirdlooking.com","username":"redbo"},"state":"REVIEWER"},{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"reviewer":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"state":"REVIEWER"},{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},"reviewer":{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},"state":"REVIEWER"},{"updated":"2014-08-27 16:53:41.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2014-09-08 01:21:30.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-08 02:22:32.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2014-09-08 20:03:19.000000000","updated_by":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"reviewer":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"state":"REVIEWER"},{"updated":"2014-09-09 00:34:50.000000000","updated_by":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"reviewer":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"state":"REVIEWER"},{"updated":"2014-09-10 02:45:40.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"},{"updated":"2014-09-11 21:44:28.000000000","updated_by":{"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},"reviewer":{"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},"state":"REVIEWER"},{"updated":"2014-09-23 21:17:34.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2014-09-29 10:01:57.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2014-11-14 22:31:33.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"ca374b944af1342b69e736847a6e4ab94e8318f9","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-08-27 11:17:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e06b57e119bbdc57269f3964fc81e4687371a70c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-27 12:31:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/93/117193/1/check/check-tempest-dsvm-full/4fe38ec : SUCCESS in 1h 00m 37s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/93/117193/1/check/check-tempest-dsvm-postgres-full/ffc01d3 : SUCCESS in 1h 12m 28s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/93/117193/1/check/check-tempest-dsvm-neutron-full/7854269 : SUCCESS in 1h 02m 25s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/93/117193/1/check/check-tempest-dsvm-neutron-heat-slow/c7f7caf : SUCCESS in 26m 05s\n- check-grenade-dsvm http://logs.openstack.org/93/117193/1/check/check-grenade-dsvm/1b67969 : SUCCESS in 44m 25s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/93/117193/1/check/gate-tempest-dsvm-large-ops/77a529c : SUCCESS in 26m 50s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/93/117193/1/check/gate-tempest-dsvm-neutron-large-ops/f8506f3 : SUCCESS in 25m 24s\n- gate-tempest-pep8 http://logs.openstack.org/93/117193/1/check/gate-tempest-pep8/a82bfe3 : SUCCESS in 3m 22s\n- gate-tempest-docs http://docs-draft.openstack.org/93/117193/1/check/gate-tempest-docs/d237a5a/doc/build/html/ : SUCCESS in 4m 09s\n- gate-tempest-python27 http://logs.openstack.org/93/117193/1/check/gate-tempest-python27/f01e610 : SUCCESS in 4m 25s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/93/117193/1/check/check-grenade-dsvm-partial-ncpu/58c5a38 : SUCCESS in 42m 53s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/93/117193/1/check/check-tempest-dsvm-ironic-nv/85c4452 : SUCCESS in 38m 38s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/93/117193/1/check/check-tempest-dsvm-ironic-postgres-nv/b4d4de2 : SUCCESS in 40m 32s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/93/117193/1/check/check-tempest-dsvm-full-icehouse/b90f4eb : SUCCESS in 53m 28s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/1/check/check-tempest-dsvm-neutron-icehouse/dcde0e9 : SUCCESS in 1h 00m 35s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/93/117193/1/check/check-tempest-dsvm-neutron-heat-slow-icehouse/97df648 : SUCCESS in 22m 42s\n- check-grenade-dsvm-icehouse http://logs.openstack.org/93/117193/1/check/check-grenade-dsvm-icehouse/53f7056 : FAILURE in 26m 39s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/1/check/check-grenade-dsvm-neutron-icehouse/a8a7669 : FAILURE in 30m 09s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/93/117193/1/check/check-tempest-dsvm-postgres-full-icehouse/d3f93bb : SUCCESS in 49m 52s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/93/117193/1/check/check-tempest-dsvm-neutron-full-icehouse/09ad882 : SUCCESS in 1h 00m 42s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/93/117193/1/check/check-tempest-dsvm-virtual-ironic-nv/0261e4f : SUCCESS in 35m 51s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/93/117193/1/check/gate-tempest-dsvm-large-ops-icehouse/817a4b4 : SUCCESS in 25m 52s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/93/117193/1/check/gate-tempest-dsvm-neutron-large-ops-icehouse/60e1f5c : SUCCESS in 22m 26s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/93/117193/1/check/check-grenade-dsvm-partial-ncpu-icehouse/2fbfe64 : FAILURE in 26m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c6d0e882bd5be0163caab30a3136597fbb1d8c16","author":{"_account_id":10016,"name":"Joseph Lanoux","email":"joseph.lanoux@hpe.com","username":"lanoux"},"date":"2014-08-27 12:56:22.000000000","message":"Patch Set 1:\n\nHi Daisuke. That\u0027s sad: I registered on the etherpad I was working on the object storage service.","accounts_in_message":[],"_revision_number":1},{"id":"9f7e0deaabb9674ba29d5de9905cfd620d1fa0af","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-08-27 16:52:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nIn addition to the inline comment, I am confused about the success code values. It seems code that was checking for any 2xx is now going through a client that is checking for specific responses. It has been the position of the swift core team in the past that swift apis are not specified to return any particular response code and that a change in the response should not be considered an api change. This should be clarified and made consistent.","accounts_in_message":[],"_revision_number":1},{"id":"ddba943830956a1849277bbb435d87f961f0e13f","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2014-08-27 17:20:15.000000000","message":"Patch Set 1: Code-Review-1\n\nDavid Kranz is correct; any 2xx status code indicates success. Testing for a specific code (like 201 or 204) is stricter than the Swift API is.\n\nIt\u0027s worth noting that Swift is simply following HTTP here. From RFC 7231 (which supersedes RFC 2616):\n\n    6.3.  Successful 2xx\n    \n       The 2xx (Successful) class of status code indicates that the client\u0027s\n       request was successfully received, understood, and accepted.\n\nHonestly, code like this:\n\n    self.expected_success(204, resp.status)\n\nstrikes me as wrong at a fundamental level. HTTP tells us what success means: a status code between 200 and 299, inclusive. If someone wants to be pedantic about it (and there are those who do, and that\u0027s fine too), that person can gather up all instances of 2xx-series status codes from RFC 7230-7235 and state that \"success\" means a status code of 200, 201, 202, 203, 204, 205, or 206.\n\nThen, one could write:\n\n    self.expected_success(resp.status)\n\nand leave the definition of \"success\" to HTTP. Because really, is there a meaningful difference between a 204 No Content response and a 200 OK response with Content-Length: 0? I don\u0027t think there is.","accounts_in_message":[],"_revision_number":1},{"id":"7c569ff7c296a0c65b329c477d95b4480ad9a219","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-08-28 01:22:05.000000000","message":"Patch Set 1:\n\nHi Joseph. I am so sorry. I missed the etherpad. If you do not mind, I will take over the role of this bp for swift.\n\n\nHi David and Sam. Thanks for your review. I was just trying to help improve tests by checking status code more strictly. \n\nBut, you are correct. Okay, I will modify to check a status code among 200-206 as done in each original test case. As Sam mentioned,\n\n    self.expected_success(resp.status)\n\nis intuitive, but this needs great refactoring so I will prefer \n\n    self.expected_success(HTTP_SUCCESS, resp.status)","accounts_in_message":[],"_revision_number":1},{"id":"0360e55b2f435b57dff338cb3a239fd1dbbfd26e","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-08-28 01:36:20.000000000","message":"Patch Set 1:\n\n@Sam: I don\u0027t really know the history of this but I would guess that when nova and swift came together \"in the beginning\", nova documented specific codes and all the other projects have followed that lead.","accounts_in_message":[],"_revision_number":1},{"id":"957b320511a0d641df2e2cd4b22b0858ca066247","author":{"_account_id":10016,"name":"Joseph Lanoux","email":"joseph.lanoux@hpe.com","username":"lanoux"},"date":"2014-08-28 08:08:24.000000000","message":"Patch Set 1:\n\nOk then.","accounts_in_message":[],"_revision_number":1},{"id":"f3089461552c0f1343eb9c15b57ec37f0f8479c8","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-08-28 17:08:35.000000000","message":"Patch Set 1: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/93/117193/1/23989/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":1},{"id":"d72395c52416f545c71e6bf402cee2e9e6429721","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2014-08-28 20:25:00.000000000","message":"Patch Set 1:\n\nI think this:\n\n    self.expected_success(HTTP_SUCCESS, resp.status)\n\nwill be sufficient. It appears that HTTP_SUCCESS is defined as (200, 201, 202, 203, 204, 205, 206), and that\u0027s broad enough for Swift.\n\nWith respect to Nova, I feel that it was a mistake to document particular status codes. The result is an API that has the syntax of HTTP but not the semantics, and HTTP\u0027s semantics are much more important than the syntax. Of course, this change isn\u0027t the place to fix that... it might not even be fixable unless someone discovers time travel.","accounts_in_message":[],"_revision_number":1},{"id":"d58c3fb859f3a36c34c05d1e3a126bfc3b188e81","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-08-29 04:46:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2841829c39a0b387385e19967f451ad393db59f4","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-08-29 05:15:57.000000000","message":"Patch Set 2: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/93/117193/2/24279/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":2},{"id":"fa4e8bd7cb40e2f33db73851f8ab8f45bc7dc2f0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-29 06:14:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/93/117193/2/check/check-tempest-dsvm-full/936b8b9 : FAILURE in 16m 36s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/93/117193/2/check/check-tempest-dsvm-postgres-full/d9bcbe3 : FAILURE in 17m 40s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/93/117193/2/check/check-tempest-dsvm-neutron-full/a2bb922 : FAILURE in 20m 19s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/93/117193/2/check/check-tempest-dsvm-neutron-heat-slow/2567ea8 : FAILURE in 26m 03s\n- check-grenade-dsvm http://logs.openstack.org/93/117193/2/check/check-grenade-dsvm/69cd634 : FAILURE in 20m 09s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/93/117193/2/check/gate-tempest-dsvm-large-ops/07a5fe0 : FAILURE in 19m 31s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/93/117193/2/check/gate-tempest-dsvm-neutron-large-ops/a2ca9df : FAILURE in 19m 20s\n- gate-tempest-pep8 http://logs.openstack.org/93/117193/2/check/gate-tempest-pep8/23bdb6e : SUCCESS in 2m 53s\n- gate-tempest-docs http://docs-draft.openstack.org/93/117193/2/check/gate-tempest-docs/008f4c3/doc/build/html/ : SUCCESS in 3m 48s\n- gate-tempest-python27 http://logs.openstack.org/93/117193/2/check/gate-tempest-python27/f87e6c3 : FAILURE in 4m 09s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/93/117193/2/check/check-grenade-dsvm-partial-ncpu/15b4c76 : FAILURE in 17m 46s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/93/117193/2/check/check-tempest-dsvm-ironic-nv/a5ca1b9 : FAILURE in 19m 01s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/93/117193/2/check/check-tempest-dsvm-ironic-postgres-nv/7f3478f : FAILURE in 17m 50s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/93/117193/2/check/check-tempest-dsvm-full-icehouse/7dbc574 : FAILURE in 15m 12s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/2/check/check-tempest-dsvm-neutron-icehouse/beaf489 : FAILURE in 15m 12s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/93/117193/2/check/check-tempest-dsvm-neutron-heat-slow-icehouse/0f91350 : FAILURE in 17m 40s\n- check-grenade-dsvm-icehouse http://logs.openstack.org/93/117193/2/check/check-grenade-dsvm-icehouse/42bbd5d : FAILURE in 32m 45s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/2/check/check-grenade-dsvm-neutron-icehouse/dba98dc : FAILURE in 34m 25s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/93/117193/2/check/check-tempest-dsvm-postgres-full-icehouse/61bfc93 : FAILURE in 18m 54s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/93/117193/2/check/check-tempest-dsvm-neutron-full-icehouse/58b0d87 : FAILURE in 15m 21s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/93/117193/2/check/check-tempest-dsvm-virtual-ironic-nv/26e131d : FAILURE in 38m 01s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/93/117193/2/check/gate-tempest-dsvm-large-ops-icehouse/f619b82 : FAILURE in 17m 50s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/93/117193/2/check/gate-tempest-dsvm-neutron-large-ops-icehouse/59823c9 : FAILURE in 17m 16s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/93/117193/2/check/check-grenade-dsvm-partial-ncpu-icehouse/c1f89ba : FAILURE in 33m 11s","accounts_in_message":[],"_revision_number":2},{"id":"c145de922342c331a22843e75a123dfec0bcc105","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-08-29 08:07:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"da6d56ba3060fd9b81edad9f0e18cde13ae42ed7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-29 09:22:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/93/117193/3/check/check-tempest-dsvm-full/5bf77cc : FAILURE in 1h 07m 55s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/93/117193/3/check/check-tempest-dsvm-postgres-full/20790f6 : FAILURE in 1h 02m 14s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/93/117193/3/check/check-tempest-dsvm-neutron-full/d2d92ad : FAILURE in 53m 38s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/93/117193/3/check/check-tempest-dsvm-neutron-heat-slow/c4bfffa : SUCCESS in 22m 05s\n- check-grenade-dsvm http://logs.openstack.org/93/117193/3/check/check-grenade-dsvm/388af5a : FAILURE in 28m 05s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/93/117193/3/check/gate-tempest-dsvm-large-ops/3ec6266 : SUCCESS in 27m 27s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/93/117193/3/check/gate-tempest-dsvm-neutron-large-ops/c888c73 : SUCCESS in 21m 45s\n- gate-tempest-pep8 http://logs.openstack.org/93/117193/3/check/gate-tempest-pep8/24450ac : SUCCESS in 4m 32s\n- gate-tempest-docs http://docs-draft.openstack.org/93/117193/3/check/gate-tempest-docs/d663296/doc/build/html/ : SUCCESS in 2m 38s\n- gate-tempest-python27 http://logs.openstack.org/93/117193/3/check/gate-tempest-python27/1a9e18c : SUCCESS in 4m 26s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/93/117193/3/check/check-grenade-dsvm-partial-ncpu/a88adbc : FAILURE in 30m 10s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/93/117193/3/check/check-tempest-dsvm-ironic-nv/334dcf3 : FAILURE in 37m 56s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/93/117193/3/check/check-tempest-dsvm-ironic-postgres-nv/8b4e93c : FAILURE in 37m 14s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/93/117193/3/check/check-tempest-dsvm-full-icehouse/9828859 : FAILURE in 51m 55s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/3/check/check-tempest-dsvm-neutron-icehouse/a9524a8 : FAILURE in 59m 02s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/93/117193/3/check/check-tempest-dsvm-neutron-heat-slow-icehouse/17d44a7 : SUCCESS in 22m 27s\n- check-grenade-dsvm-icehouse http://logs.openstack.org/93/117193/3/check/check-grenade-dsvm-icehouse/e42aa8c : FAILURE in 32m 29s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/3/check/check-grenade-dsvm-neutron-icehouse/2b3f9ca : FAILURE in 33m 54s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/93/117193/3/check/check-tempest-dsvm-postgres-full-icehouse/b5728fe : FAILURE in 47m 21s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/93/117193/3/check/check-tempest-dsvm-neutron-full-icehouse/c9ef975 : FAILURE in 54m 38s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/93/117193/3/check/check-tempest-dsvm-virtual-ironic-nv/814e576 : FAILURE in 31m 02s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/93/117193/3/check/gate-tempest-dsvm-large-ops-icehouse/5ea7a41 : SUCCESS in 23m 41s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/93/117193/3/check/gate-tempest-dsvm-neutron-large-ops-icehouse/3cc8067 : SUCCESS in 21m 17s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/93/117193/3/check/check-grenade-dsvm-partial-ncpu-icehouse/da04833 : FAILURE in 42m 27s","accounts_in_message":[],"_revision_number":3},{"id":"03407d03f02c21f2b53df32ae08f47927f12bd65","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-08-29 09:32:36.000000000","message":"Patch Set 3: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/93/117193/3/24320/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":3},{"id":"709bf024ba1030e89db2675d0a7b1bee090a8105","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-08-29 09:50:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e46bd2b960c25157de8910554e9dabd38d5aa841","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-29 11:38:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-full http://logs.openstack.org/93/117193/4/check/check-tempest-dsvm-full/a2e1268 : SUCCESS in 1h 02m 19s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/93/117193/4/check/check-tempest-dsvm-postgres-full/488b420 : SUCCESS in 1h 00m 18s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/93/117193/4/check/check-tempest-dsvm-neutron-full/2cead6e : SUCCESS in 52m 22s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/93/117193/4/check/check-tempest-dsvm-neutron-heat-slow/b292723 : SUCCESS in 25m 27s\n- check-grenade-dsvm http://logs.openstack.org/93/117193/4/check/check-grenade-dsvm/18ca1c5 : SUCCESS in 48m 42s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/93/117193/4/check/gate-tempest-dsvm-large-ops/60e1d77 : SUCCESS in 25m 12s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/93/117193/4/check/gate-tempest-dsvm-neutron-large-ops/63da0db : SUCCESS in 18m 37s\n- gate-tempest-pep8 http://logs.openstack.org/93/117193/4/check/gate-tempest-pep8/6bad4b8 : SUCCESS in 4m 23s\n- gate-tempest-docs http://docs-draft.openstack.org/93/117193/4/check/gate-tempest-docs/d5d57f6/doc/build/html/ : SUCCESS in 4m 46s\n- gate-tempest-python27 http://logs.openstack.org/93/117193/4/check/gate-tempest-python27/fc3c9fb : SUCCESS in 3m 51s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/93/117193/4/check/check-grenade-dsvm-partial-ncpu/43c46b1 : SUCCESS in 41m 54s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/93/117193/4/check/check-tempest-dsvm-ironic-nv/ba3e791 : SUCCESS in 36m 24s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/93/117193/4/check/check-tempest-dsvm-ironic-postgres-nv/3565ff7 : SUCCESS in 38m 56s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/93/117193/4/check/check-tempest-dsvm-full-icehouse/11d91a4 : SUCCESS in 57m 57s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/4/check/check-tempest-dsvm-neutron-icehouse/091d896 : SUCCESS in 1h 01m 43s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/93/117193/4/check/check-tempest-dsvm-neutron-heat-slow-icehouse/16b570a : SUCCESS in 23m 14s\n- check-grenade-dsvm-icehouse http://logs.openstack.org/93/117193/4/check/check-grenade-dsvm-icehouse/c8b34f6 : SUCCESS in 43m 53s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/4/check/check-grenade-dsvm-neutron-icehouse/4bfb478 : FAILURE in 14m 11s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/93/117193/4/check/check-tempest-dsvm-postgres-full-icehouse/d906c83 : SUCCESS in 51m 23s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/93/117193/4/check/check-tempest-dsvm-neutron-full-icehouse/7c30ff3 : SUCCESS in 1h 00m 56s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/93/117193/4/check/check-tempest-dsvm-virtual-ironic-nv/48b64dd : SUCCESS in 36m 46s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/93/117193/4/check/gate-tempest-dsvm-large-ops-icehouse/845a812 : SUCCESS in 18m 48s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/93/117193/4/check/gate-tempest-dsvm-neutron-large-ops-icehouse/9792340 : SUCCESS in 20m 03s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/93/117193/4/check/check-grenade-dsvm-partial-ncpu-icehouse/b5e1b4a : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":4},{"id":"31ee7edc5f40acd11a7dc0f38f13b6d52d31622c","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-08-29 12:48:43.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9c38488ab4e3819b18fa8117e4cef5ce5685f5e9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-08-29 12:58:40.000000000","message":"Patch Set 4: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/93/117193/4/24357/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":4},{"id":"e5824d87c881ee54e9f4bd4a44298f0af3b570dc","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2014-08-29 18:38:46.000000000","message":"Patch Set 4: Code-Review-1\n\n(8 comments)\n\nThere are a number of places where particular 2xx status codes are still being expected. I have attempted to point them all out.","accounts_in_message":[],"_revision_number":4},{"id":"556808d8db3ab66ab856f1e260629923fec10910","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-01 01:13:09.000000000","message":"Patch Set 4:\n\n(10 comments)\n\nThanks David and Sam. Please see my replies.","accounts_in_message":[],"_revision_number":4},{"id":"da9d67483d26a240c45ee67492d7270ae3205f34","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-05 01:55:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"41458692c2ab3b270563ac994b60725e9ed689b0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-05 03:07:52.000000000","message":"Patch Set 5: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/93/117193/5/24979/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":5},{"id":"26c77350b6c3942c40c8a941ec0cb74c7ed50f48","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-05 03:13:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-full http://logs.openstack.org/93/117193/5/check/check-tempest-dsvm-full/6e59608 : SUCCESS in 59m 37s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/93/117193/5/check/check-tempest-dsvm-postgres-full/651839f : SUCCESS in 1h 16m 17s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/93/117193/5/check/check-tempest-dsvm-neutron-full/9744e59 : SUCCESS in 1h 07m 21s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/93/117193/5/check/check-tempest-dsvm-neutron-heat-slow/a84b37a : SUCCESS in 24m 02s\n- check-grenade-dsvm http://logs.openstack.org/93/117193/5/check/check-grenade-dsvm/a0cf0d4 : SUCCESS in 46m 43s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/93/117193/5/check/gate-tempest-dsvm-large-ops/f423444 : SUCCESS in 25m 28s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/93/117193/5/check/gate-tempest-dsvm-neutron-large-ops/21cd0d8 : SUCCESS in 26m 49s\n- gate-tempest-pep8 http://logs.openstack.org/93/117193/5/check/gate-tempest-pep8/0ea1fd4 : SUCCESS in 4m 11s\n- gate-tempest-docs http://docs-draft.openstack.org/93/117193/5/check/gate-tempest-docs/36ea055/doc/build/html/ : SUCCESS in 2m 33s\n- gate-tempest-python27 http://logs.openstack.org/93/117193/5/check/gate-tempest-python27/17719f3 : SUCCESS in 3m 56s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/93/117193/5/check/check-grenade-dsvm-partial-ncpu/4d275d0 : SUCCESS in 44m 42s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/93/117193/5/check/check-tempest-dsvm-ironic-nv/c23ad50 : SUCCESS in 40m 16s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/93/117193/5/check/check-tempest-dsvm-ironic-postgres-nv/5a857ca : SUCCESS in 46m 44s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/93/117193/5/check/check-tempest-dsvm-full-icehouse/fc6b676 : SUCCESS in 51m 17s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/5/check/check-tempest-dsvm-neutron-icehouse/9706df6 : SUCCESS in 1h 01m 47s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/93/117193/5/check/check-tempest-dsvm-neutron-heat-slow-icehouse/b5f4fb9 : SUCCESS in 22m 53s\n- check-grenade-dsvm-icehouse http://logs.openstack.org/93/117193/5/check/check-grenade-dsvm-icehouse/1f34fab : SUCCESS in 40m 57s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/5/check/check-grenade-dsvm-neutron-icehouse/b64f286 : FAILURE in 31m 18s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/93/117193/5/check/check-tempest-dsvm-postgres-full-icehouse/1fbbc36 : SUCCESS in 47m 16s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/93/117193/5/check/check-tempest-dsvm-neutron-full-icehouse/87dcc34 : SUCCESS in 52m 58s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/93/117193/5/check/check-tempest-dsvm-virtual-ironic-nv/970cb48 : SUCCESS in 35m 54s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/93/117193/5/check/gate-tempest-dsvm-large-ops-icehouse/bde496a : SUCCESS in 26m 22s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/93/117193/5/check/gate-tempest-dsvm-neutron-large-ops-icehouse/b4e124d : SUCCESS in 18m 16s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/93/117193/5/check/check-grenade-dsvm-partial-ncpu-icehouse/5db7bd2 : SUCCESS in 41m 02s","accounts_in_message":[],"_revision_number":5},{"id":"00fa6685e53b172b2127e3f65642a2041ffa1369","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-09-05 19:11:43.000000000","message":"Patch Set 5:\n\nSo I am not going to vote on this until the swift team works this out. It seems clear to me that there are a few inconsistencies in the doc, but that the intent is that all apis return 2xx. For example, the create methods for account and container do that.","accounts_in_message":[],"_revision_number":5},{"id":"ca1333ee7d5b9da64135ccabe6fd282b6a7beca1","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2014-09-05 19:52:24.000000000","message":"Patch Set 5:\n\nYes, that\u0027s the intention of the Swift team. I went through the docs at one point and scrubbed out specific success codes, though it appears that either I did not get them all or they have crept back in.\n\nI\u0027m not going to -1 this patch since it matches the docs, but I also won\u0027t +1 it since I think the docs are wrong.","accounts_in_message":[],"_revision_number":5},{"id":"10dce305c2a157cf031b27174f437d3da70be154","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-08 01:03:19.000000000","message":"Patch Set 5:\n\nOK. I realized that most Swift team members agreed that success response code of all APIs is 2xx and docs are now inconsistent with team\u0027s intent.","accounts_in_message":[],"_revision_number":5},{"id":"89edbe5a34bc14db40819446bfd17b42cce5b75d","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-08 01:11:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6c1b0c3852cd9416b0ba32cbc73523367be92694","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-08 01:21:30.000000000","message":"Patch Set 6: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/93/117193/6/25238/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":6},{"id":"071e1d3efa67be016a1ece90bc153e626d393131","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-08 02:14:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-full/76f03c7 : SUCCESS in 45m 28s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-postgres-full/06abb6e : SUCCESS in 1h 02m 18s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-full/e30c14f : SUCCESS in 59m 43s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-heat-slow/2fe3aa7 : SUCCESS in 24m 57s\n- check-grenade-dsvm http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm/d9877f5 : SUCCESS in 36m 45s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-large-ops/b22dba0 : SUCCESS in 20m 46s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-neutron-large-ops/7f0e278 : SUCCESS in 19m 43s\n- gate-tempest-pep8 http://logs.openstack.org/93/117193/6/check/gate-tempest-pep8/0dcad04 : SUCCESS in 2m 38s\n- gate-tempest-docs http://docs-draft.openstack.org/93/117193/6/check/gate-tempest-docs/d9999de/doc/build/html/ : SUCCESS in 2m 38s\n- gate-tempest-python27 http://logs.openstack.org/93/117193/6/check/gate-tempest-python27/18e34c8 : SUCCESS in 2m 31s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-partial-ncpu/4dbefb6 : SUCCESS in 36m 02s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-ironic-nv/3ba1ad8 : SUCCESS in 36m 04s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-ironic-postgres-nv/97ddcb7 : SUCCESS in 36m 05s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-full-icehouse/9d0f471 : SUCCESS in 45m 34s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-icehouse/967bab0 : SUCCESS in 55m 26s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-heat-slow-icehouse/682c69c : SUCCESS in 25m 02s\n- check-grenade-dsvm-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-icehouse/440a947 : FAILURE in 27m 43s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-neutron-icehouse/226f7c7 : FAILURE in 12m 51s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-postgres-full-icehouse/db8cd68 : SUCCESS in 46m 55s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-full-icehouse/5424e61 : SUCCESS in 50m 13s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-virtual-ironic-nv/c03c295 : SUCCESS in 30m 45s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-large-ops-icehouse/88b49eb : SUCCESS in 21m 26s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-neutron-large-ops-icehouse/06f30e6 : SUCCESS in 21m 02s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-partial-ncpu-icehouse/a2d7327 : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":6},{"id":"2a2b04329e633ad5ee089cefcaaa29c584f74b8c","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-09-08 02:22:32.000000000","message":"Patch Set 6:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-grenade-dsvm-icehouse: https://bugs.launchpad.net/bugs/1331274\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"4afbcdb1f3188b2b41a3c48a69f435a7a13d3292","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-08 02:53:26.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"cdcc756e0ad38ef06fc179f7cdce0ebbb1b90055","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-08 04:11:55.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-full/91a4c31 : SUCCESS in 56m 26s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-postgres-full/22ed561 : SUCCESS in 1h 18m 24s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-full/f2ff515 : SUCCESS in 1h 05m 20s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-heat-slow/ce832a1 : SUCCESS in 22m 59s\n- check-grenade-dsvm http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm/8cce587 : SUCCESS in 44m 23s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-large-ops/962f155 : SUCCESS in 24m 17s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-neutron-large-ops/b27e26a : SUCCESS in 22m 25s\n- gate-tempest-pep8 http://logs.openstack.org/93/117193/6/check/gate-tempest-pep8/ce0f31b : SUCCESS in 3m 54s\n- gate-tempest-docs http://docs-draft.openstack.org/93/117193/6/check/gate-tempest-docs/ae1efa4/doc/build/html/ : SUCCESS in 3m 18s\n- gate-tempest-python27 http://logs.openstack.org/93/117193/6/check/gate-tempest-python27/486997c : SUCCESS in 2m 52s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-partial-ncpu/8aa66fd : SUCCESS in 39m 02s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-ironic-nv/45a78c5 : SUCCESS in 34m 06s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-ironic-postgres-nv/d0e79cb : SUCCESS in 36m 35s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-full-icehouse/96023af : SUCCESS in 48m 18s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-icehouse/2e57bca : SUCCESS in 55m 02s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-heat-slow-icehouse/4d0d51d : SUCCESS in 43m 13s\n- check-grenade-dsvm-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-icehouse/2291a4b : SUCCESS in 40m 18s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-neutron-icehouse/3b9277e : FAILURE in 32m 26s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-postgres-full-icehouse/0c8c21a : FAILURE in 7m 55s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-full-icehouse/033042e : SUCCESS in 54m 32s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-virtual-ironic-nv/b307dec : SUCCESS in 29m 35s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-large-ops-icehouse/5140434 : SUCCESS in 21m 07s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-neutron-large-ops-icehouse/8918da4 : SUCCESS in 21m 04s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-partial-ncpu-icehouse/4b49eec : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":6},{"id":"2c769c01816a10ef3b1d463e07dfd39f1ce4262e","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-09-08 04:12:07.000000000","message":"Patch Set 6:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-postgres-full-icehouse: https://bugs.launchpad.net/bugs/1270710\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"ccc93dc9062e152a64a31021cb8936abd5b560d3","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-08 04:13:53.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"65196f1fc10a71b4910a7314640060481db5ca4d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-08 05:28:31.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-full/67e5ae9 : SUCCESS in 55m 31s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-postgres-full/a48ee86 : SUCCESS in 1h 14m 17s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-full/0c21454 : SUCCESS in 1h 02m 27s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-heat-slow/aac52b6 : SUCCESS in 22m 18s\n- check-grenade-dsvm http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm/71559d5 : FAILURE in 30m 19s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-large-ops/63af30d : SUCCESS in 25m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-neutron-large-ops/40ada00 : SUCCESS in 25m 40s\n- gate-tempest-pep8 http://logs.openstack.org/93/117193/6/check/gate-tempest-pep8/c2e4f6a : SUCCESS in 3m 04s\n- gate-tempest-docs http://docs-draft.openstack.org/93/117193/6/check/gate-tempest-docs/06aea57/doc/build/html/ : SUCCESS in 3m 37s\n- gate-tempest-python27 http://logs.openstack.org/93/117193/6/check/gate-tempest-python27/9580de0 : SUCCESS in 2m 58s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-partial-ncpu/15896db : SUCCESS in 37m 06s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-ironic-nv/f6fa0dc : SUCCESS in 34m 52s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-ironic-postgres-nv/622a191 : SUCCESS in 39m 23s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-full-icehouse/95ba852 : SUCCESS in 48m 17s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-icehouse/caa8cfb : SUCCESS in 55m 18s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-heat-slow-icehouse/80c2665 : SUCCESS in 35m 47s\n- check-grenade-dsvm-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-icehouse/81a642a : SUCCESS in 40m 24s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-neutron-icehouse/cbab348 : FAILURE in 33m 24s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-postgres-full-icehouse/63cc7dd : FAILURE in 8m 38s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-full-icehouse/5e02446 : SUCCESS in 50m 13s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-virtual-ironic-nv/b788162 : SUCCESS in 29m 29s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-large-ops-icehouse/09d0128 : SUCCESS in 20m 30s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-neutron-large-ops-icehouse/9fc0ee3 : SUCCESS in 17m 54s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-partial-ncpu-icehouse/20e44ce : SUCCESS in 37m 45s","accounts_in_message":[],"_revision_number":6},{"id":"15002f1c0aec9b58f2e8ed34708a620a304037cd","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-09-08 05:28:44.000000000","message":"Patch Set 6:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-grenade-dsvm: https://bugs.launchpad.net/bugs/1331274\n- check-tempest-dsvm-postgres-full-icehouse: https://bugs.launchpad.net/bugs/1270710\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"9b7d2f45cca9d588501fb75de5b1cfc2599b1d06","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-08 06:18:17.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b6f3e537e16e75b6df9b7eb7488ef5e02154ffa6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-08 07:19:46.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-full/9cb0b86 : SUCCESS in 54m 34s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-postgres-full/03fdaad : SUCCESS in 1h 01m 15s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-full/ed9f630 : FAILURE in 18m 32s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-heat-slow/65d9f08 : FAILURE in 14m 22s\n- check-grenade-dsvm http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm/b229a90 : SUCCESS in 48m 16s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-large-ops/54d8206 : SUCCESS in 26m 32s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-neutron-large-ops/e2d39d7 : SUCCESS in 18m 09s\n- gate-tempest-pep8 http://logs.openstack.org/93/117193/6/check/gate-tempest-pep8/9313c82 : SUCCESS in 2m 53s\n- gate-tempest-docs http://docs-draft.openstack.org/93/117193/6/check/gate-tempest-docs/f8ae5fc/doc/build/html/ : SUCCESS in 3m 50s\n- gate-tempest-python27 http://logs.openstack.org/93/117193/6/check/gate-tempest-python27/47b0bf4 : SUCCESS in 2m 48s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-partial-ncpu/c92ad1e : SUCCESS in 43m 12s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-ironic-nv/08b9933 : SUCCESS in 39m 04s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-ironic-postgres-nv/fa98ff2 : SUCCESS in 40m 15s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-full-icehouse/32318fa : SUCCESS in 44m 21s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-icehouse/56e3818 : SUCCESS in 53m 23s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-heat-slow-icehouse/61ae307 : SUCCESS in 20m 29s\n- check-grenade-dsvm-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-icehouse/2ac9fb6 : SUCCESS in 42m 48s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-neutron-icehouse/3dae41b : FAILURE in 33m 14s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-postgres-full-icehouse/d8d5034 : SUCCESS in 46m 55s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-full-icehouse/c237c6f : FAILURE in 56m 28s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-virtual-ironic-nv/6250765 : SUCCESS in 28m 07s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-large-ops-icehouse/b206cc8 : SUCCESS in 20m 36s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-neutron-large-ops-icehouse/56487ac : SUCCESS in 20m 23s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-partial-ncpu-icehouse/d388938 : SUCCESS in 40m 28s","accounts_in_message":[],"_revision_number":6},{"id":"5a5dcb5b61b9ca8a58824736ac2eec310a4d6ec0","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-09-08 07:20:00.000000000","message":"Patch Set 6:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-neutron-full: unrecognized error\n- check-tempest-dsvm-neutron-heat-slow: https://bugs.launchpad.net/bugs/1270710\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"ea333dddbf038e60b390684e1325e388b4a75688","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-08 07:20:39.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b609dd4562b6a21033bc44961dc5551bd2f6cb0d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-08 08:25:05.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-full/7c22b75 : SUCCESS in 44m 02s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-postgres-full/5392911 : SUCCESS in 1h 00m 25s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-full/ae32cda : SUCCESS in 1h 04m 22s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-heat-slow/ec57ff7 : SUCCESS in 19m 16s\n- check-grenade-dsvm http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm/61c4f07 : SUCCESS in 41m 05s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-large-ops/26090a8 : SUCCESS in 19m 06s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-neutron-large-ops/c102562 : SUCCESS in 18m 07s\n- gate-tempest-pep8 http://logs.openstack.org/93/117193/6/check/gate-tempest-pep8/50ddb9d : SUCCESS in 4m 22s\n- gate-tempest-docs http://docs-draft.openstack.org/93/117193/6/check/gate-tempest-docs/8faf188/doc/build/html/ : SUCCESS in 2m 50s\n- gate-tempest-python27 http://logs.openstack.org/93/117193/6/check/gate-tempest-python27/b64b77e : SUCCESS in 2m 47s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-partial-ncpu/d9656dd : SUCCESS in 36m 47s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-ironic-nv/a0a10d1 : SUCCESS in 34m 23s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-ironic-postgres-nv/435cdd0 : SUCCESS in 34m 31s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-full-icehouse/4f56859 : SUCCESS in 50m 14s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-icehouse/d780073 : FAILURE in 59m 04s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-heat-slow-icehouse/ba52ff2 : SUCCESS in 21m 36s\n- check-grenade-dsvm-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-icehouse/fe27afc : SUCCESS in 39m 54s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-neutron-icehouse/eb8965c : FAILURE in 38m 23s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-postgres-full-icehouse/c3d6f81 : SUCCESS in 48m 02s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-full-icehouse/b59f9bc : SUCCESS in 52m 34s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-virtual-ironic-nv/6ce11d0 : SUCCESS in 28m 51s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-large-ops-icehouse/c05ea98 : SUCCESS in 19m 46s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-neutron-large-ops-icehouse/9cd8cb5 : SUCCESS in 19m 15s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-partial-ncpu-icehouse/94f11f9 : SUCCESS in 39m 06s","accounts_in_message":[],"_revision_number":6},{"id":"cc3c4e578a418311f46edfb72ae69ecabc5fc8f5","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-09-08 08:25:19.000000000","message":"Patch Set 6:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-neutron-icehouse: https://bugs.launchpad.net/bugs/1323658\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"0cdee6985fafa815287100d3273ddafc2bbf7bca","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-08 09:21:14.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"cf50f7e17f4d8480f2b3ce0a767c8832dff61579","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-08 10:41:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-full/4e22419 : SUCCESS in 46m 56s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-postgres-full/5351eac : SUCCESS in 1h 14m 33s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-full/78bc5f4 : SUCCESS in 1h 02m 05s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-heat-slow/b45dda5 : SUCCESS in 24m 00s\n- check-grenade-dsvm http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm/3028752 : SUCCESS in 46m 26s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-large-ops/f10cc0c : SUCCESS in 25m 52s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-neutron-large-ops/cbbc820 : SUCCESS in 25m 44s\n- gate-tempest-pep8 http://logs.openstack.org/93/117193/6/check/gate-tempest-pep8/ca4cdf4 : SUCCESS in 4m 08s\n- gate-tempest-docs http://docs-draft.openstack.org/93/117193/6/check/gate-tempest-docs/7a9901f/doc/build/html/ : SUCCESS in 2m 48s\n- gate-tempest-python27 http://logs.openstack.org/93/117193/6/check/gate-tempest-python27/c6c685d : SUCCESS in 2m 48s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-partial-ncpu/97b276d : SUCCESS in 44m 37s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-ironic-nv/2977b38 : SUCCESS in 43m 01s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-ironic-postgres-nv/7f3ac26 : SUCCESS in 43m 10s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-full-icehouse/12acff9 : SUCCESS in 48m 45s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-icehouse/8fa23cb : SUCCESS in 58m 42s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-heat-slow-icehouse/be84b0d : SUCCESS in 21m 45s\n- check-grenade-dsvm-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-icehouse/216f48d : SUCCESS in 38m 58s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-neutron-icehouse/9e98668 : FAILURE in 36m 20s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-postgres-full-icehouse/c7e1416 : SUCCESS in 48m 49s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-neutron-full-icehouse/b766fa4 : SUCCESS in 59m 57s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/93/117193/6/check/check-tempest-dsvm-virtual-ironic-nv/79ad8c6 : SUCCESS in 27m 49s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-large-ops-icehouse/7769812 : SUCCESS in 21m 34s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/93/117193/6/check/gate-tempest-dsvm-neutron-large-ops-icehouse/4873069 : SUCCESS in 20m 06s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/93/117193/6/check/check-grenade-dsvm-partial-ncpu-icehouse/e60ecc7 : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":6},{"id":"1570a524b1e9e5176672503cf00b967dbc42a35b","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-08 11:28:56.000000000","message":"Patch Set 6:\n\nxenserver: recheck","accounts_in_message":[],"_revision_number":6},{"id":"097c033d683ca96b29073d2fc8f70816af5c58a1","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-08 11:37:54.000000000","message":"Patch Set 6:\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/93/117193/6/25312/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":6},{"id":"c91056480499f7c445238f7a6cd02b83b041714b","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-09-08 20:03:19.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"808819fec206c53736e2df1e860a3013f677d32d","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2014-09-09 00:31:31.000000000","message":"Patch Set 6: Code-Review-1\n\nI have one question related to this change.\nOn Swift, the status codes of success are not defined?\n\nThis patch would relax the status code checks to all http success codes(20X). but some previous checks were more strict, I prefer strict checks as possible.","accounts_in_message":[],"_revision_number":6},{"id":"8d1b0f79504b19394cb80ec43b1d7c794918d746","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2014-09-09 00:34:50.000000000","message":"Patch Set 6: Code-Review+1\n\nI like it. Thanks!\n\n(I\u0027ll see if I can get the docs fixed, too, so we don\u0027t have this trouble any more.)","accounts_in_message":[],"_revision_number":6},{"id":"de3345f946a3c1d8ab739757de8bb309ed394d1a","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-10 01:04:25.000000000","message":"Patch Set 6:\n\nKenichi,\n\nSwift team\u0027s policy says a successful response code is in 200-206 range. It is not restricted to one code.\n\nAs you mentioned, some previous tests restricted to one successful code, but those did not meet team\u0027s policy. Some docs says the opposite now, but they will be fixed in the future.","accounts_in_message":[],"_revision_number":6},{"id":"b1ddf5e192f8b0e6879510385c75dbd298a3a9a3","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-10 01:04:49.000000000","message":"Patch Set 6:\n\nxenserver: recheck","accounts_in_message":[],"_revision_number":6},{"id":"d6a4ae93f64b0fa9bce5a7e091a88b1d05ddf795","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2014-09-10 01:08:10.000000000","message":"Patch Set 6:\n\nHi Daisuke,\n\nThanks for your explanation, can you point some related documents about succeeded codes?\nI\u0027d like to check them before approving.","accounts_in_message":[],"_revision_number":6},{"id":"156ad2ebc64bff37b4120497f450bae5571d8838","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-10 01:26:03.000000000","message":"Patch Set 6:\n\nFollowing two documents say that container PUT results are 201 or 202, and object PUT results are 201 in success. However, these will be fixed soon.\n\nObject PUT: http://docs.openstack.org/api/openstack-object-storage/1.0/content/create-update-object.html\n\nContainer PUT: http://docs.openstack.org/api/openstack-object-storage/1.0/content/create-container.html","accounts_in_message":[],"_revision_number":6},{"id":"45db60171f9ee04132c21689db0d43073167c69e","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2014-09-10 02:03:14.000000000","message":"Patch Set 6:\n\n@Daisuke,\n\nAs you said, the specified docs are against current this patch now. The docs shows status codes clearly, and I\u0027m not sure they will be changed in the future.\n\nMy main concern is that tempest cannot block backwards incompatible changes of Swift by this patch. In many cases, clients would check status codes for knowing their request success or not, and the expected status codes are hard-coded in the clients like Tempest which is before applied with this patch. If relaxing status code checks and changing them, such clients will be broken.\n\nOn Nova, new API (was called \"Nova v3 API\") tried to change status codes also for more accurate codes, but the API has been dropped now because of backwards incompatibilities.\nSo I\u0027d like to confirm this kind of change carefully.","accounts_in_message":[],"_revision_number":6},{"id":"869879d73208778d3c165caddff7baa9c41748c4","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2014-09-10 02:45:40.000000000","message":"Patch Set 6:\n\nTo be very clear, Swift\u0027s policy is and always has been to respond with 2xx series on success, 4xx series on client error, and 5xx on server error.\n\nSee http://www.w3.org/Protocols/rfc2616/rfc2616-sec6.html#sec6.1.1 and the newer http://tools.ietf.org/html/rfc7231#section-6","accounts_in_message":[],"_revision_number":6},{"id":"f8d5ab644cace644d80ec8b16d3eead4ad80b9e4","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2014-09-10 02:57:51.000000000","message":"Patch Set 6:\n\n@John\n\nThanks for joining.\nI understand the policy and that is common policy for RESTful services.\n\nAccording to the docs which are specified by Daisuke and current tempest code, success codes of each Swift API seems to be defined as each value: 201 or something now.\n\nHowever, will/can these status codes be changed in the range 20X in the future? Is that a Swift policy?","accounts_in_message":[],"_revision_number":6},{"id":"91bb1fb7da2247120e1f97fb394de7a91b1ad20c","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2014-09-10 03:50:48.000000000","message":"Patch Set 6:\n\nI\u0027m agreeing with everything that Sam has said already in this conversation. Swift\u0027s policy is error code classes, as I stated above.\n\nI think the confusion is what the clients should do as a matter of policy when talking to Swift versus what Swift sends[1]. Clearly, Swift isn\u0027t sending one of 100 different successful status codes to a good write request (ie 200-299). Most of those aren\u0027t even formally defined by RFC.\n\nWhat I am saying is that Swift\u0027s policy (and API) is that a client should treat a 2xx response code as a success. I believe the docs currently list one or more specific codes, but the intention is and always has been for Swift to send error code classes (ie 2xx, 4xx, 5xx) as defined in the RFC[2]. Obviously certain response codes have defined meanings (eg 206, 404, 501), and Swift will send the best possible response code to reflect the status of the request. The point is that returning eg 200 vs 201 vs 202 all mean \"success\" and if we someday also return 222 or any other 2xx code, that is still considered a successful response to a request.\n\nThis is Swift\u0027s official policy.\n\nIf the docs do not include this information, I believe the docs should be updated.\n\n\n[1] http://en.wikipedia.org/wiki/Robustness_principle\n\n[2] http://tools.ietf.org/html/rfc7231#section-6","accounts_in_message":[],"_revision_number":6},{"id":"74d9c14b280081110687b2b57093a3a38e2461ae","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2014-09-10 07:19:04.000000000","message":"Patch Set 6:\n\n@John\n\nBasically, OpenStack projects should not change success status code even if the changed code is more accurate than before, because of OpenStack API change guideline.\n\nhttps://wiki.openstack.org/wiki/APIChangeGuidelines#Generally_Not_Acceptable\n\n Generally Not Acceptable\n [..]\n * Changing which response code is returned on success\n\nThat is my sticky point on this patch.","accounts_in_message":[],"_revision_number":6},{"id":"b09f0cffbccdc19b99c8d8dddfce1d07b29858e2","author":{"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},"date":"2014-09-11 21:44:28.000000000","message":"Patch Set 6: Code-Review-1\n\nThe documentation should not be changed to something as vague as 200-299 inclusive. (https://review.openstack.org/#/c/120622/) \n\nI\u0027m investigating whether a cloud provider is the one who eventually decides what swift returns. This is an important deciding point especially for swift where you can run ceph or gluster underlying but want to implement the Object Storage API. If the swift code says 200 or 204, we need to document and test for that.","accounts_in_message":[],"_revision_number":6},{"id":"636f5e50eb8f45d794500b841aaba88c7970ff03","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2014-09-12 22:13:30.000000000","message":"Patch Set 6:\n\nI am confused, and I hope I am simply misreading things. Is the objection that OpenStack Tempest should test OpenStack Swift to ensure its compatibility with Ceph?","accounts_in_message":[],"_revision_number":6},{"id":"94981cf9af2d035a7bca01ff14a23bf14bd468a8","author":{"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},"date":"2014-09-12 22:45:20.000000000","message":"Patch Set 6:\n\nHi Samuel, ha, no, I think that OpenStack tempest tests should test OpenStack project\u0027s code. :)","accounts_in_message":[],"_revision_number":6},{"id":"793ca74378fb242c179556956b12a466b7a3a5b8","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2014-09-14 00:19:22.000000000","message":"Patch Set 6:\n\nI have left a comment on the docs patch suggesting that we document the codes Swift does return, but document that clients should respond to response code classes rather than specific codes.\n\nFrom there, I\u0027m not sure what the best end-result in tempest is, but I\u0027m fine with either way.\n\nMy main goal is that clients do not rely on a specific code (because that leads to brittle code).","accounts_in_message":[],"_revision_number":6},{"id":"d032fa66552450438df3cf251d773501ca654236","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2014-09-15 09:01:03.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"475f7ece3eedbda80fd95769bd4bf8ddc1de87cf","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2014-09-15 14:31:40.000000000","message":"Patch Set 6:\n\n# I wrote the same thing on the doc patch review, but I think that is important for this patch also then write it again:\n\nAs a developer working for REST API, I can understand that developers want to change the success codes to more accurate codes if they find. However, what should we expect to clients with such success code changes?\n\nIf we cannot promise stable success codes, clients cannot handle such success codes even if swift provides more accurate codes. Clients should expect the success codes as unstable and they cannot switch their behaviors based on the success codes. That seems a dilemma.","accounts_in_message":[],"_revision_number":6},{"id":"c13b58c7c77059fb3e6cede484c5ecb751a58db9","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2014-09-19 04:20:18.000000000","message":"Patch Set 6:\n\nI would like to move forward this discussion, and as everyone might think, I also would like to look for a happy medium.\n\nThe root problem is the difference between Swift\u0027s philosophy and OpenStack\u0027s (or compute-related projects\u0027) philosophy. And, I do NOT believe that OpenStack\u0027s philosophy always takes precedence of individual project\u0027s one. Especially, Swift has the property of eventual consistency. Therefore, response codes, contents, and metadata vary depending on internal states.\n\nOn the other hand, Tempest has the responsibility of API stability, so Kenichi\u0027s sticking point is also important.\n\nThe following is just an idea from my side. Swift team naturally try to return more appropriate class of success codes in their developing and reviewing phase. Therefore, for example, it seems unlikely that PUT request returns 203-206. From this point of view, Tempest can check response code in the narrower range (but not specific one) as follows:\n\nPUT -\u003e [200, 201, 202]\nGET -\u003e [200, 204, 206]\nHEAD -\u003e [200, 204]\nPOST, DELETE -\u003e [200, 202, 204]\n\nWhat do you think this idea? I want to hear opinions from Swift and QA side.","accounts_in_message":[],"_revision_number":6},{"id":"663c00cd2eafb53d6cb0a44be4e45aa67919e115","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2014-09-23 21:17:34.000000000","message":"Patch Set 6: Workflow-1\n\nJust a -1 place-holder here until it\u0027s clarified which codes to use.","accounts_in_message":[],"_revision_number":6},{"id":"6de404ac67ac575f3da0cc0645bac31cd420931f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-29 10:01:57.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"b820d5ee05a1eb5b1e8dc8428f0564a33cfcf982","author":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"date":"2014-11-14 16:27:47.000000000","message":"Patch Set 6:\n\nHey,\nCan we move on with this one ?\n\nI forgot about this patch and submitted a new patch here : https://review.openstack.org/#/c/134555/\n\nThis patch implements somehow what Daisuke Morita said in his last comment (\"PUT -\u003e [200, 201, 202] GET -\u003e [200, 204, 206] HEAD -\u003e [200, 204] POST, DELETE -\u003e [200, 202, 204]\")\n\nThis way, Swift devs are free to use new status for new features (no Tempest test to block these new features) and they got a warning (Tempest fails) if they change the status code of a feature already tested in tempest. Then one can submit a patch to Tempest if needed, such a small change would be merged quickly imo.\n\nMy patch here https://review.openstack.org/#/c/134555/ tries to make a compromise : several status code are accepted for the same API call ([200, 201, 202], same behavior of what Tempest accepts now) but does not accept a wide range of status code (200-299).","accounts_in_message":[],"_revision_number":6},{"id":"18b148f389a2385cdafff3d59e31a138d91b6afa","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2014-11-14 22:31:33.000000000","message":"Patch Set 6: Code-Review-1\n\nThis change seems to be taking on two things at once.\n\n1) consolidate the status code checking to the client instead of the tests\n2) change the way tempest validates status codes\n\nI think we\u0027re stalled out on #2 - but #1 can happen purely as a clean up, there should be basically no disagreement about DRYing that out.\n\nThe other referenced here [1] only attempts to deal with #1.  We should merge that.\n\nThen after that lands, we can rebase this or spin up a new patch to just take on #2 and continue this discussion there if needed.\n\n1. https://review.openstack.org/#/c/134555/","accounts_in_message":[],"_revision_number":6},{"id":"942e2ef06d85bce650ea267ae0a5b62974c76db7","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2016-06-03 05:09:19.000000000","message":"Abandoned\n\nAs http://lists.openstack.org/pipermail/openstack-dev/2016-May/096204.html , we start abandoning patches have gotten old without any updating after negative feedback. Please restore if necessary to restart this again.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6d568d8efbd3dd90ccf9f43b3b476279fb3b82d2","revisions":{"00400ab24c3ddbc8a6d8ee5321fc5633535e1d39":{"kind":"REWORK","_number":1,"created":"2014-08-27 11:17:22.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/93/117193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/117193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/117193/1"}}},"commit":{"parents":[{"commit":"9c1ce58a22cfa1143fa8e8b4ddfc6f57d31db27e","subject":"Merge \"Add service tags to tempest/api/orchestration/stacks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9c1ce58a22cfa1143fa8e8b4ddfc6f57d31db27e"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2014-08-27 11:11:11.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2014-08-27 11:17:15.000000000","tz":540},"subject":"Add client response checking for object storage service","message":"Add client response checking for object storage service\n\n* Checks for response status codes to client side were added\n* Checks for response status codes from all tests were removed\n\nPartially implements: blueprint client-checks-success\n\nChange-Id: Ib6de34fc0d86dc2a1b7859e4d3dc11488799809d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/00400ab24c3ddbc8a6d8ee5321fc5633535e1d39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/00400ab24c3ddbc8a6d8ee5321fc5633535e1d39"}]},"branch":"refs/heads/master"},"d9e2b40788f1ee9a7a8dabf249d3b79c61cf2db9":{"kind":"REWORK","_number":2,"created":"2014-08-29 04:46:59.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/93/117193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/117193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/117193/2"}}},"commit":{"parents":[{"commit":"f0536c7b78d78a61625c6e4544945b5134c2bcc8","subject":"Merge \"Add functools.wraps on all function decorators\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f0536c7b78d78a61625c6e4544945b5134c2bcc8"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2014-08-27 11:11:11.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2014-08-29 04:46:50.000000000","tz":540},"subject":"Add client response checking for object storage service","message":"Add client response checking for object storage service\n\n* Checks for response status codes to client side were added\n* Checks for response status codes from all tests were removed\n\nPartially implements: blueprint client-checks-success\n\nChange-Id: Ib6de34fc0d86dc2a1b7859e4d3dc11488799809d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d9e2b40788f1ee9a7a8dabf249d3b79c61cf2db9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d9e2b40788f1ee9a7a8dabf249d3b79c61cf2db9"}]},"branch":"refs/heads/master"},"4cdcb2697ce921a87291cfe084a77707e02b7575":{"kind":"REWORK","_number":3,"created":"2014-08-29 08:07:58.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/93/117193/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/117193/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/117193/3"}}},"commit":{"parents":[{"commit":"f0536c7b78d78a61625c6e4544945b5134c2bcc8","subject":"Merge \"Add functools.wraps on all function decorators\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f0536c7b78d78a61625c6e4544945b5134c2bcc8"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2014-08-27 11:11:11.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2014-08-29 08:07:51.000000000","tz":540},"subject":"Add client response checking for object storage service","message":"Add client response checking for object storage service\n\n* Checks for response status codes to client side were added\n* Checks for response status codes from all tests were removed\n\nPartially implements: blueprint client-checks-success\n\nChange-Id: Ib6de34fc0d86dc2a1b7859e4d3dc11488799809d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4cdcb2697ce921a87291cfe084a77707e02b7575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4cdcb2697ce921a87291cfe084a77707e02b7575"}]},"branch":"refs/heads/master"},"c581c08a3ded830457180912779c0acc27ddb6a2":{"kind":"REWORK","_number":4,"created":"2014-08-29 09:50:16.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/93/117193/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/117193/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/117193/4"}}},"commit":{"parents":[{"commit":"68988e346ae9d1d1534747027e9bfbc23af89d4c","subject":"Merge \"Fix hardcoded gateway in IPV6 subnet api tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/68988e346ae9d1d1534747027e9bfbc23af89d4c"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2014-08-27 11:11:11.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2014-08-29 09:50:09.000000000","tz":540},"subject":"Add client response checking for object storage service","message":"Add client response checking for object storage service\n\n* Checks for response status codes to client side were added\n* Checks for response status codes from all tests were removed\n\nPartially implements: blueprint client-checks-success\n\nChange-Id: Ib6de34fc0d86dc2a1b7859e4d3dc11488799809d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c581c08a3ded830457180912779c0acc27ddb6a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c581c08a3ded830457180912779c0acc27ddb6a2"}]},"branch":"refs/heads/master"},"32c191c2bfd877fbb9f14e7d1699caa87efa2aa2":{"kind":"REWORK","_number":5,"created":"2014-09-05 01:55:26.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/93/117193/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/117193/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/117193/5"}}},"commit":{"parents":[{"commit":"a5c669d7f0c0383edc8e5b961a4fd4b32044d04f","subject":"Properly detect grenade in check_logs script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a5c669d7f0c0383edc8e5b961a4fd4b32044d04f"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2014-08-27 11:11:11.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2014-09-05 01:55:18.000000000","tz":540},"subject":"Add client response checking for object storage service","message":"Add client response checking for object storage service\n\n* Checks for response status codes to client side were added\n* Checks for response status codes from all tests were removed\n\nPartially implements: blueprint client-checks-success\n\nChange-Id: Ib6de34fc0d86dc2a1b7859e4d3dc11488799809d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/32c191c2bfd877fbb9f14e7d1699caa87efa2aa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/32c191c2bfd877fbb9f14e7d1699caa87efa2aa2"}]},"branch":"refs/heads/master"},"6d568d8efbd3dd90ccf9f43b3b476279fb3b82d2":{"kind":"REWORK","_number":6,"created":"2014-09-08 01:11:46.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/93/117193/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/117193/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/117193/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/117193/6"}}},"commit":{"parents":[{"commit":"ed286f811a2db9ad9184fa26e787ca45e26854f4","subject":"Merge \"Don\u0027t run lbaas scenario test as smoke test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ed286f811a2db9ad9184fa26e787ca45e26854f4"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2014-08-27 11:11:11.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2014-09-08 01:11:38.000000000","tz":540},"subject":"Add client response checking for object storage service","message":"Add client response checking for object storage service\n\n* Checks for response status codes to client side were added\n* Checks for response status codes from all tests were removed\n\nPartially implements: blueprint client-checks-success\n\nChange-Id: Ib6de34fc0d86dc2a1b7859e4d3dc11488799809d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6d568d8efbd3dd90ccf9f43b3b476279fb3b82d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6d568d8efbd3dd90ccf9f43b3b476279fb3b82d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
