)]}'
{"id":"openstack%2Ftempest~207294","triplet_id":"openstack%2Ftempest~master~Ia12d55c34ef540f2488e998883a2c4a4da81e30f","project":"openstack/tempest","branch":"master","topic":"service-client-unit-tests","hashtags":[],"change_id":"Ia12d55c34ef540f2488e998883a2c4a4da81e30f","subject":"Add unit test for quotas_client","status":"MERGED","created":"2015-07-30 05:12:28.000000000","updated":"2015-08-24 10:07:59.000000000","submitted":"2015-08-24 10:07:57.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4014dee210c4b78d42a5401944d2da9208101134","_number":207294,"virtual_id_number":207294,"owner":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},{"value":0,"date":"2015-08-20 01:42:45.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":2,"date":"2015-08-24 10:07:57.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-08-18 07:35:49.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":10300,"name":"Pranali Deore","email":"pranalideore@gmail.com","username":"PranaliDeore"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"all":[{"value":0,"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2015-08-24 08:57:00.000000000","_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":2,"date":"2015-08-19 22:26:34.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":10300,"name":"Pranali Deore","email":"pranalideore@gmail.com","username":"PranaliDeore"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"all":[{"value":0,"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2015-08-24 08:57:00.000000000","_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":10300,"name":"Pranali Deore","email":"pranalideore@gmail.com","username":"PranaliDeore"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":10300,"name":"Pranali Deore","email":"pranalideore@gmail.com","username":"PranaliDeore"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10300,"name":"Pranali Deore","email":"pranalideore@gmail.com","username":"PranaliDeore"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-08-11 14:37:28.000000000","updated_by":{"_account_id":10300,"name":"Pranali Deore","email":"pranalideore@gmail.com","username":"PranaliDeore"},"reviewer":{"_account_id":10300,"name":"Pranali Deore","email":"pranalideore@gmail.com","username":"PranaliDeore"},"state":"REVIEWER"},{"updated":"2015-08-18 07:35:49.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-19 22:26:34.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2015-08-20 01:42:45.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-24 08:57:00.000000000","updated_by":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"reviewer":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"state":"REVIEWER"},{"updated":"2015-08-24 10:07:57.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"8c43751c57dc523fa20b1d62c613c78a7c600ba5","author":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"date":"2015-07-30 05:12:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9c226318cda06022be8105d0898d4e48aab9b0d","author":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"date":"2015-07-30 06:05:14.000000000","message":"Topic changed from quotas to service-client-unit-tests","accounts_in_message":[],"_revision_number":1},{"id":"c1690fbe503cd5a28994cbcf83c94d32bd1d349b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-30 07:37:50.000000000","message":"Patch Set 1:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/94/207294/1/6416/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":1},{"id":"27e30eea01467c94bae2b0c20b6f82dd4ba368ba","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-30 18:26:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-python34 http://logs.openstack.org/94/207294/1/check/gate-tempest-python34/abbef9a/ : SUCCESS in 2m 14s\n- gate-tempest-pep8 http://logs.openstack.org/94/207294/1/check/gate-tempest-pep8/3ea6390/ : SUCCESS in 1m 37s\n- gate-tempest-docs http://docs-draft.openstack.org/94/207294/1/check/gate-tempest-docs/6b072d7//doc/build/html/ : SUCCESS in 2m 28s\n- gate-tempest-python27 http://logs.openstack.org/94/207294/1/check/gate-tempest-python27/031faa9/ : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"50cb0b6d4cdae35b9acf42a41b863f4249ba2ace","author":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"date":"2015-07-31 15:20:59.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks for your work on this. Could you please squash your changes so that you submit one change per class that you are testing. Currently you are submitting one change per method, but that leads too many many small changes that increases the review burden.\n\nNote, that we usually favor small patches but as this is a \"work from scratch\", I think it\u0027s best to submit a comprehensive first patch.","accounts_in_message":[],"_revision_number":1},{"id":"3b9fe400d74e0dee006514945475a497f88cd4da","author":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"date":"2015-08-03 05:22:24.000000000","message":"Patch Set 1: Workflow-1\n\nThank you for your comments, Jordan. I\u0027ll comprehend each patches I had submitted.","accounts_in_message":[],"_revision_number":1},{"id":"05a656ac34f623417bf9e2d8c9cfaa4aae2a36ec","author":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"date":"2015-08-03 06:19:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"05b88fa0b724dbaafe21b5f1796898cc6378176f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-03 06:24:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-python34 http://logs.openstack.org/94/207294/2/check/gate-tempest-python34/f04ea7d/ : SUCCESS in 3m 03s\n- gate-tempest-pep8 http://logs.openstack.org/94/207294/2/check/gate-tempest-pep8/fce1727/ : SUCCESS in 1m 42s\n- gate-tempest-docs http://docs-draft.openstack.org/94/207294/2/check/gate-tempest-docs/d864d40//doc/build/html/ : SUCCESS in 2m 11s\n- gate-tempest-python27 http://logs.openstack.org/94/207294/2/check/gate-tempest-python27/f310e9f/ : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":2},{"id":"d64951d64a69fbf4f11cddbe8b921ec2eb6f9f78","author":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"date":"2015-08-03 06:32:36.000000000","message":"Patch Set 3: Commit message was updated","accounts_in_message":[],"_revision_number":3},{"id":"304223c63dcb2ec981b6c5359212c1118942015b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-03 06:35:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-python34 http://logs.openstack.org/94/207294/3/check/gate-tempest-python34/3a08a01/ : SUCCESS in 2m 44s\n- gate-tempest-pep8 http://logs.openstack.org/94/207294/3/check/gate-tempest-pep8/e84feaf/ : SUCCESS in 1m 57s\n- gate-tempest-docs http://docs-draft.openstack.org/94/207294/3/check/gate-tempest-docs/3a781e8//doc/build/html/ : SUCCESS in 1m 56s\n- gate-tempest-python27 http://logs.openstack.org/94/207294/3/check/gate-tempest-python27/1184c17/ : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":3},{"id":"256e02bf8448ee3f2f19d9f0df3f4b0851a6d38d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-08-03 07:29:09.000000000","message":"Patch Set 3:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/94/207294/3/6756/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":3},{"id":"87acf539a1265d6e10dc1659ccf555e56f6e4875","author":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"date":"2015-08-03 07:55:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nThanks for squashing your changes, it will really speed up the review process.\n\nIt looks for to me, just a small fix to have a \"better looking code\".","accounts_in_message":[],"_revision_number":3},{"id":"c868fed528d91f66ab8dfdbf44c79da399995411","author":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"date":"2015-08-03 07:56:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b0c5d2d80da574f1e1e49552c1b22647b4d7ca2a","author":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"date":"2015-08-03 08:07:09.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"39cddedd7780f4e6a0fb6c235dc9d3f018b39eec","author":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"date":"2015-08-03 10:08:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3d9920ae16c52506efa8596fe479576a319ba881","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-03 10:26:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-python34 http://logs.openstack.org/94/207294/4/check/gate-tempest-python34/88094fd/ : SUCCESS in 2m 41s\n- gate-tempest-pep8 http://logs.openstack.org/94/207294/4/check/gate-tempest-pep8/4998000/ : SUCCESS in 1m 21s\n- gate-tempest-docs http://docs-draft.openstack.org/94/207294/4/check/gate-tempest-docs/83df09d//doc/build/html/ : SUCCESS in 1m 29s\n- gate-tempest-python27 http://logs.openstack.org/94/207294/4/check/gate-tempest-python27/adb2044/ : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":4},{"id":"abc62ff291f54e5e51215002ab50038b9ad2219f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-08-03 11:13:11.000000000","message":"Patch Set 4:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/94/207294/4/6783/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":4},{"id":"09f479514a23eab6a41219f075f2c6892eed767d","author":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"date":"2015-08-06 08:40:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d02f62668c4fb28fe526aa1f5c8f080b28f96faa","author":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"date":"2015-08-06 08:48:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"826bf3834bea9803725af34419d341d1e9d54ea8","author":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"date":"2015-08-06 08:49:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8c87a64750019830e92a888de17d63651c8f4f26","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-06 08:53:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-python34 http://logs.openstack.org/94/207294/5/check/gate-tempest-python34/db44506/ : SUCCESS in 2m 44s\n- gate-tempest-pep8 http://logs.openstack.org/94/207294/5/check/gate-tempest-pep8/304a8be/ : SUCCESS in 1m 26s\n- gate-tempest-docs http://docs-draft.openstack.org/94/207294/5/check/gate-tempest-docs/e378284//doc/build/html/ : SUCCESS in 1m 30s\n- gate-tempest-python27 http://logs.openstack.org/94/207294/5/check/gate-tempest-python27/139a110/ : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":5},{"id":"e4d6891e32f2fb3e8f8265fb573705d608bad0f1","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-08-06 09:47:52.000000000","message":"Patch Set 5:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/94/207294/5/7274/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":5},{"id":"e058d7e984f6289ce8f6267372b29ef0d44a94ed","author":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"date":"2015-08-11 14:01:39.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6910f4e6b43817423503f81b3d97e70fbf2fa712","author":{"_account_id":10300,"name":"Pranali Deore","email":"pranalideore@gmail.com","username":"PranaliDeore"},"date":"2015-08-11 14:37:28.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ca3a7443f53449f03487706db9210966a51af006","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2015-08-13 00:39:15.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)\n\nMost part is good for me, just few comments.","accounts_in_message":[],"_revision_number":5},{"id":"ef0b1409ac99cafa7ccd505d713d6f8a80d164e8","author":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"date":"2015-08-18 05:38:04.000000000","message":"Patch Set 5:\n\nThank you for your review. I fix problems you pointed out.","accounts_in_message":[],"_revision_number":5},{"id":"d4210e5a279c4527375b2e5cb44c7eb381d7d5ce","author":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"date":"2015-08-18 05:38:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5e0ab85096da4f72d8beb10cfc80b832f9829542","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-18 05:43:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-python34 http://logs.openstack.org/94/207294/6/check/gate-tempest-python34/af16dd0/ : SUCCESS in 2m 56s\n- gate-tempest-pep8 http://logs.openstack.org/94/207294/6/check/gate-tempest-pep8/4f06e29/ : SUCCESS in 1m 53s\n- gate-tempest-docs http://docs-draft.openstack.org/94/207294/6/check/gate-tempest-docs/19c537b//doc/build/html/ : SUCCESS in 1m 56s\n- gate-tempest-python27 http://logs.openstack.org/94/207294/6/check/gate-tempest-python27/613f5cf/ : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":6},{"id":"ed75cb861d865f43b3f2db1a87b24a9a98f27afe","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-08-18 07:35:49.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- SRIOV-Tempest http://144.76.193.39/ci-artifacts/207294/6/SRIOV-Tempest : SUCCESS in 55m 46s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"a42339af166c27a7c0b5995642468637e08210a6","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2015-08-19 22:26:34.000000000","message":"Patch Set 6: Code-Review+2\n\nlgtm, thanks for updating","accounts_in_message":[],"_revision_number":6},{"id":"67f255c43b9cbb1dab17579aae9af4e796bb7f66","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-08-20 01:42:45.000000000","message":"Patch Set 6:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/94/207294/6/8721/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":6},{"id":"e78ecd519d353491b93270b6a5f3f3c91d94755c","author":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"date":"2015-08-24 08:57:00.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"21e5777195995988c29aca664f8761a02a8e538a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-24 08:57:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"9d14df9ca0d5c7b191689d285fb6682d33bee74a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-24 10:07:57.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tempest-python34 http://logs.openstack.org/94/207294/6/gate/gate-tempest-python34/7ba478f/ : SUCCESS in 2m 48s\n- gate-tempest-pep8 http://logs.openstack.org/94/207294/6/gate/gate-tempest-pep8/b1c35bd/ : SUCCESS in 1m 52s\n- gate-tempest-docs http://docs-draft.openstack.org/94/207294/6/gate/gate-tempest-docs/9d6a6f6//doc/build/html/ : SUCCESS in 1m 39s\n- gate-tempest-python27 http://logs.openstack.org/94/207294/6/gate/gate-tempest-python27/5b1b92c/ : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":6},{"id":"4014dee210c4b78d42a5401944d2da9208101134","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-24 10:07:59.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3f73184355683063d145e721d5d5d5bc23bde2ad","revisions":{"a6906c777d7f933b8fa529c072600b7ec546db7a":{"kind":"REWORK","_number":1,"created":"2015-07-30 05:12:28.000000000","uploader":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"ref":"refs/changes/94/207294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/207294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/207294/1"}}},"commit":{"parents":[{"commit":"c4bfc797ff01116d00f0ebc485c66a8b26d3ac80","subject":"Merge \"Reuse mocked body value between tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c4bfc797ff01116d00f0ebc485c66a8b26d3ac80"}]}],"author":{"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","date":"2015-07-30 05:10:28.000000000","tz":0},"committer":{"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","date":"2015-07-30 05:10:28.000000000","tz":0},"subject":"Add unit test for method show_quota_set","message":"Add unit test for method show_quota_set\n\nThis patch adds unit test for quotas_client.\nSpecific about method show_quota_set.\n\nChange-Id: Ia12d55c34ef540f2488e998883a2c4a4da81e30f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a6906c777d7f933b8fa529c072600b7ec546db7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a6906c777d7f933b8fa529c072600b7ec546db7a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c4bfc797ff01116d00f0ebc485c66a8b26d3ac80","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1e94a9a457db680c0afd720266458c1b64cdbeea":{"kind":"REWORK","_number":2,"created":"2015-08-03 06:19:47.000000000","uploader":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"ref":"refs/changes/94/207294/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/207294/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/207294/2"}}},"commit":{"parents":[{"commit":"3101ebddd7bfb214d8de277e365499f04667f492","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3101ebddd7bfb214d8de277e365499f04667f492"}]}],"author":{"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","date":"2015-08-03 06:18:15.000000000","tz":0},"committer":{"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","date":"2015-08-03 06:18:15.000000000","tz":0},"subject":"Add unit test for method quotas_client","message":"Add unit test for method quotas_client\n\nThis patch adds unit test for quotas_client.\n\nChange-Id: Ia12d55c34ef540f2488e998883a2c4a4da81e30f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1e94a9a457db680c0afd720266458c1b64cdbeea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1e94a9a457db680c0afd720266458c1b64cdbeea"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3101ebddd7bfb214d8de277e365499f04667f492","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"162bb2f94091ccd19d877ac980718c81fb80f775":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2015-08-03 06:32:36.000000000","uploader":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"ref":"refs/changes/94/207294/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/207294/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/207294/3"}}},"commit":{"parents":[{"commit":"3101ebddd7bfb214d8de277e365499f04667f492","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3101ebddd7bfb214d8de277e365499f04667f492"}]}],"author":{"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","date":"2015-08-03 06:18:15.000000000","tz":0},"committer":{"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","date":"2015-08-03 06:32:36.000000000","tz":0},"subject":"Add unit test for quotas_client","message":"Add unit test for quotas_client\n\nThis patch adds unit test for quotas_client.\n\nChange-Id: Ia12d55c34ef540f2488e998883a2c4a4da81e30f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/162bb2f94091ccd19d877ac980718c81fb80f775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/162bb2f94091ccd19d877ac980718c81fb80f775"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3101ebddd7bfb214d8de277e365499f04667f492","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"daddbfb2b33b6bb78ed8d577df12aac4f4bbb129":{"kind":"REWORK","_number":4,"created":"2015-08-03 10:08:56.000000000","uploader":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"ref":"refs/changes/94/207294/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/207294/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/207294/4"}}},"commit":{"parents":[{"commit":"3101ebddd7bfb214d8de277e365499f04667f492","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3101ebddd7bfb214d8de277e365499f04667f492"}]}],"author":{"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","date":"2015-08-03 06:18:15.000000000","tz":0},"committer":{"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","date":"2015-08-03 10:07:42.000000000","tz":0},"subject":"Add unit test for quotas_client","message":"Add unit test for quotas_client\n\nThis patch adds unit test for quotas_client.\n\nChange-Id: Ia12d55c34ef540f2488e998883a2c4a4da81e30f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/daddbfb2b33b6bb78ed8d577df12aac4f4bbb129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/daddbfb2b33b6bb78ed8d577df12aac4f4bbb129"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3101ebddd7bfb214d8de277e365499f04667f492","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"227a78364eb86e9205eb40756646c6c89787acdc":{"kind":"REWORK","_number":5,"created":"2015-08-06 08:48:59.000000000","uploader":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"ref":"refs/changes/94/207294/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/207294/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/207294/5"}}},"commit":{"parents":[{"commit":"3101ebddd7bfb214d8de277e365499f04667f492","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3101ebddd7bfb214d8de277e365499f04667f492"}]}],"author":{"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","date":"2015-08-03 06:18:15.000000000","tz":0},"committer":{"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","date":"2015-08-06 08:48:10.000000000","tz":0},"subject":"Add unit test for quotas_client","message":"Add unit test for quotas_client\n\nThis patch adds unit test for quotas_client.\n\nChange-Id: Ia12d55c34ef540f2488e998883a2c4a4da81e30f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/227a78364eb86e9205eb40756646c6c89787acdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/227a78364eb86e9205eb40756646c6c89787acdc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3101ebddd7bfb214d8de277e365499f04667f492","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3f73184355683063d145e721d5d5d5bc23bde2ad":{"kind":"REWORK","_number":6,"created":"2015-08-18 05:38:51.000000000","uploader":{"_account_id":17104,"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","username":"yasui"},"ref":"refs/changes/94/207294/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/207294/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/207294/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/207294/6"}}},"commit":{"parents":[{"commit":"3101ebddd7bfb214d8de277e365499f04667f492","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3101ebddd7bfb214d8de277e365499f04667f492"}]}],"author":{"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","date":"2015-08-03 06:18:15.000000000","tz":0},"committer":{"name":"Kenji Yasui","email":"k-yasui@ti.jp.nec.com","date":"2015-08-18 05:38:25.000000000","tz":0},"subject":"Add unit test for quotas_client","message":"Add unit test for quotas_client\n\nThis patch adds unit test for quotas_client.\n\nChange-Id: Ia12d55c34ef540f2488e998883a2c4a4da81e30f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3f73184355683063d145e721d5d5d5bc23bde2ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3f73184355683063d145e721d5d5d5bc23bde2ad"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3101ebddd7bfb214d8de277e365499f04667f492","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
