)]}'
{"id":"openstack%2Ftempest~223247","triplet_id":"openstack%2Ftempest~master~Iffd6fa21fbba6ffb914f92d4ba1ad989a8c6ff95","project":"openstack/tempest","branch":"master","topic":"service-client-migration","hashtags":[],"change_id":"Iffd6fa21fbba6ffb914f92d4ba1ad989a8c6ff95","subject":"Remove tests.base module and use tempest-lib\u0027s","status":"ABANDONED","created":"2015-09-14 18:14:16.000000000","updated":"2015-09-14 22:08:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cc110b0c5f85f263ef131998d299ada91fc2e323","_number":223247,"virtual_id_number":223247,"owner":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"date":"2015-09-14 22:08:07.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":-1,"date":"2015-09-14 21:14:46.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-1,"date":"2015-09-14 21:11:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-09-14 20:56:34.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2015-09-14 21:11:22.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2015-09-14 21:14:46.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2015-09-14 22:08:07.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e68e9397f46ededeb8ac24471ad6189ae34f2200","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2015-09-14 18:14:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5097b67b781fa2fb73adf10453f23052131623ac","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2015-09-14 18:26:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e320009751abc994949232a4e23652b2f3d7f3cc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-14 18:43:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-python34 http://logs.openstack.org/47/223247/2/check/gate-tempest-python34/f54a57c/ : SUCCESS in 3m 08s\n- gate-tempest-pep8 http://logs.openstack.org/47/223247/2/check/gate-tempest-pep8/1fab3ce/ : SUCCESS in 1m 37s\n- gate-tempest-docs http://docs-draft.openstack.org/47/223247/2/check/gate-tempest-docs/3a7d334//doc/build/html/ : SUCCESS in 1m 36s\n- gate-tempest-python27 http://logs.openstack.org/47/223247/2/check/gate-tempest-python27/3307406/ : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":2},{"id":"5bb0f702e3cfcb897b83ca14a3c19f610167d725","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-14 19:27:51.000000000","message":"Patch Set 2:\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/47/223247/2/11674/index.html : SUCCESS in 57 m 40 s","accounts_in_message":[],"_revision_number":2},{"id":"1971f9627eb4bac03057197bb5c6b05afb34acf5","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2015-09-14 20:56:34.000000000","message":"Patch Set 2: Code-Review-1\n\nThe code itself looks good, but I\u0027m not convinced we should create this kind of dependency between tempest unit tests an tempest-lib ones.","accounts_in_message":[],"_revision_number":2},{"id":"869c7912bdb7ef62ba2894ffa5930a765735b335","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2015-09-14 21:08:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8f8da928eee852498f6ed49bfd6b310062c65d2","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2015-09-14 21:11:22.000000000","message":"Patch Set 3: Code-Review-1\n\nI definitely don\u0027t like doing this we should keep the unit tests isolated from the lib as much as possible. Let\u0027s just update the imports in the migration patch. We don\u0027t need to get everything done in tempest so it\u0027s a straight cp. We can modify things slightly in the tempest-lib patch.","accounts_in_message":[],"_revision_number":3},{"id":"e21b4c57168d366e239c00229af3d67c22161431","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-14 21:14:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-python34 http://logs.openstack.org/47/223247/3/check/gate-tempest-python34/b99321a/ : FAILURE in 1m 59s\n- gate-tempest-pep8 http://logs.openstack.org/47/223247/3/check/gate-tempest-pep8/68d4a1b/ : FAILURE in 1m 50s\n- gate-tempest-docs http://docs-draft.openstack.org/47/223247/3/check/gate-tempest-docs/7b12c74//doc/build/html/ : SUCCESS in 1m 56s\n- gate-tempest-python27 http://logs.openstack.org/47/223247/3/check/gate-tempest-python27/9aef10e/ : FAILURE in 2m 01s","accounts_in_message":[],"_revision_number":3},{"id":"b47c3a1479c7c0d56c59ea3a4cee5c1d77893887","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2015-09-14 21:16:20.000000000","message":"Abandoned\n\nYeah, After discussing with Andrea, I have the same idea.\nNow this is unnecessary.","accounts_in_message":[],"_revision_number":3},{"id":"e165891594a9c8185b521969b14e6b920daacfa1","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-14 22:08:07.000000000","message":"Patch Set 3:\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/47/223247/3/11679/index.html : SUCCESS in 54 m 05 s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ab753b72bf7a6ce93b219302e388eb6f7f65d0d2","revisions":{"526184b1d31106defcde2b1641fc40c7654914ea":{"kind":"REWORK","_number":1,"created":"2015-09-14 18:14:16.000000000","uploader":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"ref":"refs/changes/47/223247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/47/223247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/47/223247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/47/223247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/47/223247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/47/223247/1"}}},"commit":{"parents":[{"commit":"8517357df793fdd853388b03c75ab5ad426dee8a","subject":"Use tempest-lib\u0027s fake_auth_provider","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8517357df793fdd853388b03c75ab5ad426dee8a"}]}],"author":{"name":"Ken\u0027ichi Ohmichi","email":"ken-oomichi@wx.jp.nec.com","date":"2015-09-14 18:10:24.000000000","tz":0},"committer":{"name":"Ken\u0027ichi Ohmichi","email":"ken-oomichi@wx.jp.nec.com","date":"2015-09-14 18:10:24.000000000","tz":0},"subject":"Use tempest-lib\u0027s tests.base module","message":"Use tempest-lib\u0027s tests.base module\n\ntests.base module is the same between tempest-lib and Tempest.\nAnd we are migrating service client code from Tempest to tempest-lib\nand many service clients unit tests use tests.base module.\nSo this patch replaces tests.base import with tempest-lib\u0027s one for\nsmooth migration.\n\nChange-Id: Iffd6fa21fbba6ffb914f92d4ba1ad989a8c6ff95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/526184b1d31106defcde2b1641fc40c7654914ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/526184b1d31106defcde2b1641fc40c7654914ea"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8517357df793fdd853388b03c75ab5ad426dee8a","is_merged_in_target_branch":false,"change_id":"I3ad6761651cec5e66012d08e6b63322f53286a5c","change_number":223225,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d3c0b3e81f727d2f9f5a5a341f9cd751d0543881":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2015-09-14 18:26:04.000000000","uploader":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"ref":"refs/changes/47/223247/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/47/223247/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/47/223247/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/47/223247/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/47/223247/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/47/223247/2"}}},"commit":{"parents":[{"commit":"9b07ae70111fa133119456e3045e5b4d031155cc","subject":"Remove fake_auth_provider and use tempest-lib\u0027s","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9b07ae70111fa133119456e3045e5b4d031155cc"}]}],"author":{"name":"Ken\u0027ichi Ohmichi","email":"ken-oomichi@wx.jp.nec.com","date":"2015-09-14 18:10:24.000000000","tz":0},"committer":{"name":"Ken\u0027ichi Ohmichi","email":"ken-oomichi@wx.jp.nec.com","date":"2015-09-14 18:25:21.000000000","tz":0},"subject":"Remove tests.base module and use tempest-lib\u0027s","message":"Remove tests.base module and use tempest-lib\u0027s\n\ntests.base module is the same between tempest-lib and Tempest.\nAnd we are migrating service client code from Tempest to tempest-lib\nand many service clients unit tests use tests.base module.\nSo this patch replaces tests.base import with tempest-lib\u0027s one for\nsmooth migration.\n\nChange-Id: Iffd6fa21fbba6ffb914f92d4ba1ad989a8c6ff95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d3c0b3e81f727d2f9f5a5a341f9cd751d0543881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d3c0b3e81f727d2f9f5a5a341f9cd751d0543881"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9b07ae70111fa133119456e3045e5b4d031155cc","is_merged_in_target_branch":false,"change_id":"I3ad6761651cec5e66012d08e6b63322f53286a5c","change_number":223225,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ab753b72bf7a6ce93b219302e388eb6f7f65d0d2":{"kind":"REWORK","_number":3,"created":"2015-09-14 21:08:19.000000000","uploader":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"ref":"refs/changes/47/223247/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/47/223247/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/47/223247/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/47/223247/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/47/223247/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/47/223247/3"}}},"commit":{"parents":[{"commit":"50f8295889126b6693d7b0891e774cf468fb375c","subject":"Use tempest-lib\u0027s fake_auth_provider on services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/50f8295889126b6693d7b0891e774cf468fb375c"}]}],"author":{"name":"Ken\u0027ichi Ohmichi","email":"ken-oomichi@wx.jp.nec.com","date":"2015-09-14 18:10:24.000000000","tz":0},"committer":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2015-09-14 20:44:55.000000000","tz":540},"subject":"Remove tests.base module and use tempest-lib\u0027s","message":"Remove tests.base module and use tempest-lib\u0027s\n\ntests.base module is the same between tempest-lib and Tempest.\nAnd we are migrating service client code from Tempest to tempest-lib\nand many service clients unit tests use tests.base module.\nSo this patch replaces tests.base import with tempest-lib\u0027s one for\nsmooth migration.\n\nChange-Id: Iffd6fa21fbba6ffb914f92d4ba1ad989a8c6ff95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ab753b72bf7a6ce93b219302e388eb6f7f65d0d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ab753b72bf7a6ce93b219302e388eb6f7f65d0d2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"50f8295889126b6693d7b0891e774cf468fb375c","is_merged_in_target_branch":true,"change_id":"I3ad6761651cec5e66012d08e6b63322f53286a5c","change_number":223225,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
