)]}'
{"id":"openstack%2Ftempest~226400","triplet_id":"openstack%2Ftempest~master~Ieb5de54a162714407453552594b5e69097207bb6","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"Ieb5de54a162714407453552594b5e69097207bb6","subject":"Add unit test for test_servers_client","status":"ABANDONED","created":"2015-09-22 16:27:19.000000000","updated":"2015-10-20 18:47:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ebe4fd35fcf4247fc6c79ca0abc77f6777d24450","_number":226400,"virtual_id_number":226400,"owner":{"_account_id":17866,"name":"Truong Le","email":"let@us.ibm.com","username":"truongle"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"date":"2015-09-22 17:27:07.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2015-10-05 15:28:20.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2015-10-18 03:54:29.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-09-22 17:27:07.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-05 15:28:20.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2015-10-18 03:54:29.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3401034b8e253bdf95bef0f21c8198ed590f71a","author":{"_account_id":17866,"name":"Truong Le","email":"let@us.ibm.com","username":"truongle"},"date":"2015-09-22 16:27:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d11163995a24546bf15ffed2ce90011ddd42e9cf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-22 16:30:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-python34 http://logs.openstack.org/00/226400/1/check/gate-tempest-python34/988a4bc/ : SUCCESS in 2m 27s\n- gate-tempest-pep8 http://logs.openstack.org/00/226400/1/check/gate-tempest-pep8/27f9a64/ : SUCCESS in 1m 45s\n- gate-tempest-docs http://docs-draft.openstack.org/00/226400/1/check/gate-tempest-docs/b627887//doc/build/html/ : SUCCESS in 1m 58s\n- gate-tempest-python27 http://logs.openstack.org/00/226400/1/check/gate-tempest-python27/ed1da34/ : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1d73a62c1cc7258bc9540921cdab2cde4b00d315","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-22 17:27:07.000000000","message":"Patch Set 1:\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/00/226400/1/12058/index.html : FAILURE in 57 m 19 s","accounts_in_message":[],"_revision_number":1},{"id":"28972a40c3aff9a034f268434dfeba4b403ff4ca","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-09-22 18:37:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Tempest-Sriov http://144.76.193.39/ci-artifacts/226400/1/Tempest-Sriov : FAILURE in 16m 19s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"39597d9ace0948c5acc2f87c85b89d6bb8f24145","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-05 15:28:20.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"1f6d3ae5d24297ebb4050eda9e0abc7e5341b068","author":{"_account_id":17866,"name":"Truong Le","email":"let@us.ibm.com","username":"truongle"},"date":"2015-10-18 03:52:55.000000000","message":"Abandoned\n\nNo longer needed","accounts_in_message":[],"_revision_number":1},{"id":"5a8b62b6f6e72a858924fc6c618fdba1fad178c4","author":{"_account_id":17866,"name":"Truong Le","email":"let@us.ibm.com","username":"truongle"},"date":"2015-10-18 03:53:57.000000000","message":"Restored\n\nabandon by mistake","accounts_in_message":[],"_revision_number":1},{"id":"e44d0ca4b5b3ecea539074528cf0ead112dab442","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-18 03:54:05.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"57037b252546008531d1facb612e62570719e957","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-10-18 03:54:29.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"ebe4fd35fcf4247fc6c79ca0abc77f6777d24450","author":{"_account_id":17866,"name":"Truong Le","email":"let@us.ibm.com","username":"truongle"},"date":"2015-10-20 18:47:52.000000000","message":"Abandoned\n\nthe original patch was merged to the master branch. This patch is no longer needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e91bfd163d78ebbe6119a522e87629a65a2ff67","revisions":{"4e91bfd163d78ebbe6119a522e87629a65a2ff67":{"kind":"REWORK","_number":1,"created":"2015-09-22 16:27:19.000000000","uploader":{"_account_id":17866,"name":"Truong Le","email":"let@us.ibm.com","username":"truongle"},"ref":"refs/changes/00/226400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/00/226400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/00/226400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/00/226400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/00/226400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/00/226400/1"}}},"commit":{"parents":[{"commit":"de66951533dd2bf1085212f5140c117db723729d","subject":"Add unit test for test_servers_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/de66951533dd2bf1085212f5140c117db723729d"}]}],"author":{"name":"Truong Le","email":"let@us.ibm.com","date":"2015-09-22 16:26:18.000000000","tz":-300},"committer":{"name":"Truong Le","email":"let@us.ibm.com","date":"2015-09-22 16:26:18.000000000","tz":-300},"subject":"Add unit test for test_servers_client","message":"Add unit test for test_servers_client\n\nWe have a plan to migrate service clients to tempest-lib from Tempest.\nHowever, we are not doing enough tests for service clients now.\nThis patch adds unit test for test_servers_client\n\nRemove redundant lines\n\nChange-Id: Ieb5de54a162714407453552594b5e69097207bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4e91bfd163d78ebbe6119a522e87629a65a2ff67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4e91bfd163d78ebbe6119a522e87629a65a2ff67"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"de66951533dd2bf1085212f5140c117db723729d","is_merged_in_target_branch":false,"change_id":"I5c2f73b87b033baa362360b6add6eabd02f54c38","change_number":226391,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
