)]}'
{"id":"openstack%2Ftempest~275652","triplet_id":"openstack%2Ftempest~master~Ib1142245bb1a5c87408978b30dce694c5b520fbb","project":"openstack/tempest","branch":"master","topic":"bug/1537811","hashtags":[],"change_id":"Ib1142245bb1a5c87408978b30dce694c5b520fbb","subject":"Check Content-Length header according to RFC 7230","status":"ABANDONED","created":"2016-02-03 12:00:51.000000000","updated":"2016-11-16 11:26:42.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"583f61b7c15655fc0027dfa6326beee8b4b43d4b","_number":275652,"virtual_id_number":275652,"owner":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"date":"2016-05-12 12:24:46.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":97,"name":"Daryl Walleck","email":"daryl.walleck@rackspace.com","username":"dwalleck"},{"value":-1,"date":"2016-05-12 12:17:52.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7428,"name":"Yaroslav Lobankov","email":"lobankovyv@gmail.com","username":"ylobankov"},{"_account_id":20553,"name":"Brian Ober","email":"bober@us.ibm.com","username":"ober37"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},{"_account_id":7662,"name":"Rob Crittenden","email":"rcritten@redhat.com","username":"rcritten"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7428,"name":"Yaroslav Lobankov","email":"lobankovyv@gmail.com","username":"ylobankov"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":97,"name":"Daryl Walleck","email":"daryl.walleck@rackspace.com","username":"dwalleck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-1,"date":"2016-05-12 10:54:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7428,"name":"Yaroslav Lobankov","email":"lobankovyv@gmail.com","username":"ylobankov"},{"value":-1,"date":"2016-05-12 10:54:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20553,"name":"Brian Ober","email":"bober@us.ibm.com","username":"ober37"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7662,"name":"Rob Crittenden","email":"rcritten@redhat.com","username":"rcritten"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":97,"name":"Daryl Walleck","email":"daryl.walleck@rackspace.com","username":"dwalleck"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7428,"name":"Yaroslav Lobankov","email":"lobankovyv@gmail.com","username":"ylobankov"},{"_account_id":20553,"name":"Brian Ober","email":"bober@us.ibm.com","username":"ober37"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},{"_account_id":7662,"name":"Rob Crittenden","email":"rcritten@redhat.com","username":"rcritten"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":97,"name":"Daryl Walleck","email":"daryl.walleck@rackspace.com","username":"dwalleck"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7428,"name":"Yaroslav Lobankov","email":"lobankovyv@gmail.com","username":"ylobankov"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":20553,"name":"Brian Ober","email":"bober@us.ibm.com","username":"ober37"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7662,"name":"Rob Crittenden","email":"rcritten@redhat.com","username":"rcritten"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":97,"name":"Daryl Walleck","email":"daryl.walleck@rackspace.com","username":"dwalleck"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7428,"name":"Yaroslav Lobankov","email":"lobankovyv@gmail.com","username":"ylobankov"},{"_account_id":7662,"name":"Rob Crittenden","email":"rcritten@redhat.com","username":"rcritten"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},{"_account_id":20553,"name":"Brian Ober","email":"bober@us.ibm.com","username":"ober37"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-02-03 19:53:18.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2016-02-03 20:57:52.000000000","updated_by":{"_account_id":97,"name":"Daryl Walleck","email":"daryl.walleck@rackspace.com","username":"dwalleck"},"reviewer":{"_account_id":97,"name":"Daryl Walleck","email":"daryl.walleck@rackspace.com","username":"dwalleck"},"state":"REVIEWER"},{"updated":"2016-02-03 22:22:42.000000000","updated_by":{"_account_id":7428,"name":"Yaroslav Lobankov","email":"lobankovyv@gmail.com","username":"ylobankov"},"reviewer":{"_account_id":7428,"name":"Yaroslav Lobankov","email":"lobankovyv@gmail.com","username":"ylobankov"},"state":"REVIEWER"},{"updated":"2016-04-15 20:11:31.000000000","updated_by":{"_account_id":7662,"name":"Rob Crittenden","email":"rcritten@redhat.com","username":"rcritten"},"reviewer":{"_account_id":7662,"name":"Rob Crittenden","email":"rcritten@redhat.com","username":"rcritten"},"state":"REVIEWER"},{"updated":"2016-04-16 17:51:56.000000000","updated_by":{"_account_id":20553,"name":"Brian Ober","email":"bober@us.ibm.com","username":"ober37"},"reviewer":{"_account_id":20553,"name":"Brian Ober","email":"bober@us.ibm.com","username":"ober37"},"state":"REVIEWER"},{"updated":"2016-05-12 12:17:52.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-05-12 12:24:46.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9ed2644d947e5519a8317152587cfb53256de68","author":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"date":"2016-02-03 12:00:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b10524c4f72a117a8490e0e87d3601a2401e4baf","author":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"date":"2016-02-03 15:27:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b861976c27ff6c5f62696b7439e78a9ac5bc1f0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-03 17:02:55.000000000","message":"Patch Set 2:\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/52/275652/2/5120/index.html : FAILURE in 01 h 05 m 39 s","accounts_in_message":[],"_revision_number":2},{"id":"e8a003953c06331b6cf0bbe1390c7da8666a2d25","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2016-02-03 19:53:18.000000000","message":"Patch Set 2:\n\nwow, so that looks like a fairly generic check - is it used on all services, and swift is the only wsgi server that will return zero value content-length with a 204?\n\noh, probably just no other services ever have responses with no body...\n\nI\u0027m not sure how you\u0027re going to get this merged w/o the upstream fix - or the upstream fix w/o this tempest change?  Is there a procedure for this?  It might be one of those situations where we can\u0027t change it despite the RFC violation? [1]  Might have to wait for a new swift api version to clean up - we\u0027re supposed to have a list going for these things somewhere...\n\n1. In fairness 2616 doesn\u0027t make such a strong prescription against this behavior as near as I can tell and 7230 was drafted after the swift v1.0 api was released.  Further, the client behavior described in 7230 makes it clear that regardless of the presence of a Content-Length header (zero value or otherwise) the 204 response is terminated after the first blank line - so there\u0027s probably little risk of harm to well behaved clients.","accounts_in_message":[],"_revision_number":2},{"id":"00a13d39f920ee541fa30e9967a2dac702e61b88","author":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"date":"2016-02-03 20:03:44.000000000","message":"Patch Set 2:\n\nIt seems, that yes, only swift is guilty :)\nThe failed CI check has failed on object store tests only. But in one test the response doesn\u0027t have a http status?","accounts_in_message":[],"_revision_number":2},{"id":"1330cdb87fc575c12e348e2ca3bbdef7d447bc6e","author":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"date":"2016-02-03 20:05:47.000000000","message":"Patch Set 2:\n\nI\u0027m not familiar how to resolve the chicken and egg problem here, but in a TDD world, the test should go in first.","accounts_in_message":[],"_revision_number":2},{"id":"d5cc1209abf07a1b4b76097eb0b03edc46b83731","author":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"date":"2016-02-03 20:17:07.000000000","message":"Patch Set 2:\n\nIf I just make it non-fail with 204 and no Content-Length, then the test will mostly useless.","accounts_in_message":[],"_revision_number":2},{"id":"c3d04d8f70a29130cb66864d1ff8d921d2652c5f","author":{"_account_id":97,"name":"Daryl Walleck","email":"daryl.walleck@rackspace.com","username":"dwalleck"},"date":"2016-02-03 20:57:52.000000000","message":"Patch Set 2:\n\nI understand the purpose of this patch, but this essentially adds a shadow test. If the desired test is to verify that a specific response does not contain a field, it would be much clearer to write that as a concrete test","accounts_in_message":[],"_revision_number":2},{"id":"91ffaa7e4e8d821f8b1882fadfbfdf50b8095fae","author":{"_account_id":7428,"name":"Yaroslav Lobankov","email":"lobankovyv@gmail.com","username":"ylobankov"},"date":"2016-02-03 22:22:42.000000000","message":"Patch Set 2: Code-Review-1\n\nIt looks like this patch https://review.openstack.org/#/c/272062/2 does the same thing as you. Or I am missing something...","accounts_in_message":[],"_revision_number":2},{"id":"f5bc5c49bf4710ae0a1fbba75b2b2e392b06d309","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-03 23:13:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-full http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-full/989d931/ : FAILURE in 54m 42s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-postgres-full/c5a286a/ : FAILURE in 52m 42s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-neutron-full/398da9a/ : FAILURE in 1h 07m 21s\n- gate-grenade-dsvm http://logs.openstack.org/52/275652/2/check/gate-grenade-dsvm/1179c35/ : FAILURE in 50m 11s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-large-ops/6438cfe/ : SUCCESS in 49m 32s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-neutron-large-ops/9fdce4e/ : SUCCESS in 30m 52s\n- gate-tempest-python34 http://logs.openstack.org/52/275652/2/check/gate-tempest-python34/19e2cec/ : SUCCESS in 2m 15s\n- gate-tempest-pep8 http://logs.openstack.org/52/275652/2/check/gate-tempest-pep8/99ea761/ : SUCCESS in 1m 42s\n- gate-tempest-docs http://docs-draft.openstack.org/52/275652/2/check/gate-tempest-docs/b8d7876//doc/build/html/ : SUCCESS in 2m 01s\n- gate-tempest-python27 http://logs.openstack.org/52/275652/2/check/gate-tempest-python27/7698938/ : SUCCESS in 2m 07s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/52/275652/2/check/gate-grenade-dsvm-multinode/8eec65b/ : FAILURE in 51m 47s (non-voting)\n- gate-grenade-dsvm-neutron http://logs.openstack.org/52/275652/2/check/gate-grenade-dsvm-neutron/11a67be/ : FAILURE in 30m 32s\n- gate-tempest-dsvm-ironic-agent_ssh-nv http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-ironic-agent_ssh-nv/4967815/ : SUCCESS in 37m 58s (non-voting)\n- gate-tempest-dsvm-neutron-full-ssh http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-neutron-full-ssh/aef18a2/ : FAILURE in 59m 22s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full/976535d/ : FAILURE in 59m 37s (non-voting)\n- gate-tempest-dsvm-layer4 http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-layer4/fdf670e/ : SUCCESS in 56m 31s\n- gate-tempest-dsvm-neutron-dvr http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-neutron-dvr/40624e3/ : FAILURE in 1h 04m 58s\n- gate-tempest-dsvm-full-liberty http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-full-liberty/808a8b0/ : FAILURE in 1h 11m 18s\n- gate-tempest-dsvm-postgres-full-liberty http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-postgres-full-liberty/18db8f6/ : FAILURE in 43m 48s\n- gate-tempest-dsvm-neutron-full-liberty http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-neutron-full-liberty/0b6e024/ : FAILURE in 1h 15m 15s\n- gate-tempest-dsvm-full-kilo http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-full-kilo/901d6ca/ : FAILURE in 51m 21s\n- gate-tempest-dsvm-postgres-full-kilo http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-postgres-full-kilo/5c47770/ : FAILURE in 48m 02s\n- gate-tempest-dsvm-neutron-full-kilo http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-neutron-full-kilo/e0252a3/ : FAILURE in 1h 05m 24s\n- gate-tempest-dsvm-ceilometer-mysql-neutron-full http://logs.openstack.org/52/275652/2/check/gate-tempest-dsvm-ceilometer-mysql-neutron-full/ebbf104/ : FAILURE in 1h 08m 48s","accounts_in_message":[],"_revision_number":2},{"id":"f4b4dbca55f2f04aa767137a02f32af850355972","author":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"date":"2016-02-04 07:37:42.000000000","message":"Patch Set 2:\n\nIt doesn\u0027t do the same, it just for the same bug :)\nHowever, I missed it. Actually I don\u0027t know if it is a good idea to only relax the tests to make them pass. That\u0027s kind of manager thinking, who wants only to see green lights, but nothing is fixed.","accounts_in_message":[],"_revision_number":2},{"id":"7e409c17515929114192606578bdde570e872f4d","author":{"_account_id":7428,"name":"Yaroslav Lobankov","email":"lobankovyv@gmail.com","username":"ylobankov"},"date":"2016-02-05 15:37:29.000000000","message":"Patch Set 2:\n\nIt looks like your patch doesn\u0027t work. Could you please take a look at test failures in Jenkins checks? Thanks.","accounts_in_message":[],"_revision_number":2},{"id":"56ea20d1a960c2e966d1f8c5870a2cbcff4a36b6","author":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"date":"2016-02-05 15:45:46.000000000","message":"Patch Set 2:\n\nHi,\nThe patch works, because it shows that the current code is not RFC 7230 compliant (as in the related bug report).","accounts_in_message":[],"_revision_number":2},{"id":"a8d7c74aede945396076e466c9f20d161c30d77d","author":{"_account_id":7662,"name":"Rob Crittenden","email":"rcritten@redhat.com","username":"rcritten"},"date":"2016-04-15 20:11:31.000000000","message":"Patch Set 2:\n\nThe server bug is tracked in https://bugs.launchpad.net/swift/+bug/1537811 review https://review.openstack.org/291461","accounts_in_message":[],"_revision_number":2},{"id":"a14d78ac20918f13755da2cf532e327739020bd4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2016-04-15 23:24:47.000000000","message":"Patch Set 2:\n\n(1 comment)\n\noh oh !  a difference in response because apache with mod_ssl is striping headers is *sort* of a demonstrated harm?  If clients are finding that the presence of a content-length header on 204 is *already* unreliable (because of intermediate proxies stripping it) - then maybe there would be some benefit to remove it.\n\nAlthough FWIW any client that wants to *remove* their reliance on the Content-Length header being present for 204 should feel quite free to do so with or without a change to the Swift API?  It\u0027s exactly the clients that *expect* the header to be there that I\u0027m worried will break?  (a test suite is a type of client)\n\nIf you just don\u0027t even *look* for a Content-Length header on 204 you shouldn\u0027t notice if it\u0027s there or not?  If we knew that no one\u0027s test suites would break if we suddenly just stopped adding content-length to 204 responses we could be a lot more confident and sane that this is *not* a part of the API that risks breaking someone [1]\n\n1. https://xkcd.com/1172/","accounts_in_message":[],"_revision_number":2},{"id":"d40440309b3d53bb7ba719ee8a77c4e7883f07ca","author":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"date":"2016-04-16 08:46:38.000000000","message":"Patch Set 2:\n\nIt is true that apache mod_proxy already removes Content-Length for 204. We didn\u0027t have problems with that (except tempest test failures), but we don\u0027t use a lot of clients, only the ones in OpenStack.\nI don\u0027t know if the consesus is allow this header anyway, or go for full RFC 7230 conformance. This patch is about the later.","accounts_in_message":[],"_revision_number":2},{"id":"1a3f20f029f64ad27796048ee80292575ea84d97","author":{"_account_id":20553,"name":"Brian Ober","email":"bober@us.ibm.com","username":"ober37"},"date":"2016-04-16 17:51:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d2a77bd6314603d0baaabc4274689587e6de5e8","author":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"date":"2016-04-18 12:20:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e938f38e5b828d28d09a2715a9d42edb3292ae44","author":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"date":"2016-05-12 10:54:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"717758c7f573dcf93bb2624e98f96b28a386c388","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-12 12:17:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-full http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-full/cd4a9be/ : FAILURE in 39m 44s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-postgres-full/f78b676/ : FAILURE in 46m 47s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-neutron-full/aff5029/ : FAILURE in 54m 18s\n- gate-grenade-dsvm http://logs.openstack.org/52/275652/3/check/gate-grenade-dsvm/36efa36/ : FAILURE in 29m 40s\n- gate-tempest-python34 http://logs.openstack.org/52/275652/3/check/gate-tempest-python34/05242bd/ : SUCCESS in 2m 46s\n- gate-tempest-pep8 http://logs.openstack.org/52/275652/3/check/gate-tempest-pep8/0077f9e/ : SUCCESS in 2m 43s\n- gate-tempest-docs http://docs-draft.openstack.org/52/275652/3/check/gate-tempest-docs/163dec6//doc/build/html/ : SUCCESS in 2m 12s\n- gate-tempest-python27 http://logs.openstack.org/52/275652/3/check/gate-tempest-python27/c479bf4/ : SUCCESS in 2m 11s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/52/275652/3/check/gate-grenade-dsvm-multinode/7c6944c/ : FAILURE in 32m 50s\n- gate-grenade-dsvm-neutron http://logs.openstack.org/52/275652/3/check/gate-grenade-dsvm-neutron/b89971b/ : FAILURE in 32m 43s\n- gate-tempest-dsvm-ironic-agent_ssh-nv http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-ironic-agent_ssh-nv/de5649f/ : FAILURE in 37m 27s (non-voting)\n- gate-tempest-dsvm-neutron-full-ssh http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-neutron-full-ssh/1a7f57a/ : FAILURE in 54m 55s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/28e6e1a/ : FAILURE in 14m 45s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph/de1516b/ : FAILURE in 47m 29s\n- gate-tempest-dsvm-layer4 http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-layer4/03ca220/ : SUCCESS in 29m 03s\n- gate-tempest-dsvm-neutron-dvr http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-neutron-dvr/34060e0/ : FAILURE in 1h 02m 53s\n- gate-tempest-dsvm-full-mitaka http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-full-mitaka/9ce39a7/ : FAILURE in 43m 58s\n- gate-tempest-dsvm-postgres-full-mitaka http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-postgres-full-mitaka/7812b3c/ : FAILURE in 43m 00s\n- gate-tempest-dsvm-neutron-full-mitaka http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-neutron-full-mitaka/485a7de/ : FAILURE in 1h 00m 46s\n- gate-tempest-dsvm-full-liberty http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-full-liberty/4cd47bf/ : FAILURE in 43m 34s\n- gate-tempest-dsvm-postgres-full-liberty http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-postgres-full-liberty/425225e/ : FAILURE in 45m 46s\n- gate-tempest-dsvm-neutron-full-liberty http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-neutron-full-liberty/8b65dbd/ : FAILURE in 1h 03m 51s\n- gate-tempest-dsvm-ceilometer-mysql-neutron-full http://logs.openstack.org/52/275652/3/check/gate-tempest-dsvm-ceilometer-mysql-neutron-full/7d305e7/ : FAILURE in 56m 30s","accounts_in_message":[],"_revision_number":3},{"id":"86f6c8878f4f8cfcd5b365011d53b0d3ea28dbed","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-12 12:24:46.000000000","message":"Patch Set 3:\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/52/275652/3/14836/index.html : FAILURE in 01 h 18 m 18 s","accounts_in_message":[],"_revision_number":3},{"id":"c57dc3afa66a66a84e0b3236e32eb67855132219","author":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"date":"2016-11-16 11:26:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"87300578be41f29f90593da4965cac8ccadcecc4","revisions":{"5cffcbcd09175fa15b83409af7a003b863738f5b":{"kind":"REWORK","_number":1,"created":"2016-02-03 12:00:51.000000000","uploader":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"ref":"refs/changes/52/275652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/52/275652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/275652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/275652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/275652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/52/275652/1"}}},"commit":{"parents":[{"commit":"781f2b7a1ce71a041cfd81a946f9acc05084feff","subject":"Merge \"skip test_list_server_filters_by_regex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/781f2b7a1ce71a041cfd81a946f9acc05084feff"}]}],"author":{"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","date":"2016-02-03 11:57:16.000000000","tz":60},"committer":{"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","date":"2016-02-03 12:00:49.000000000","tz":60},"subject":"Check Content-Length header according to RFC 7230","message":"Check Content-Length header according to RFC 7230\n\n\"A server MUST NOT send a Content-Length header field in any\nresponse with a status code of 1xx (Informational) or 204 (No\nContent).\"\n\nChange-Id: Ib1142245bb1a5c87408978b30dce694c5b520fbb\nRelated-Bug: 1537811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5cffcbcd09175fa15b83409af7a003b863738f5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5cffcbcd09175fa15b83409af7a003b863738f5b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"781f2b7a1ce71a041cfd81a946f9acc05084feff","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e19177590f29581c04456f243c63b24be9b58d6e":{"kind":"REWORK","_number":2,"created":"2016-02-03 15:27:33.000000000","uploader":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"ref":"refs/changes/52/275652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/52/275652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/275652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/275652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/275652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/52/275652/2"}}},"commit":{"parents":[{"commit":"781f2b7a1ce71a041cfd81a946f9acc05084feff","subject":"Merge \"skip test_list_server_filters_by_regex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/781f2b7a1ce71a041cfd81a946f9acc05084feff"}]}],"author":{"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","date":"2016-02-03 11:57:16.000000000","tz":60},"committer":{"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","date":"2016-02-03 15:27:21.000000000","tz":60},"subject":"Check Content-Length header according to RFC 7230","message":"Check Content-Length header according to RFC 7230\n\n\"A server MUST NOT send a Content-Length header field in any\nresponse with a status code of 1xx (Informational) or 204 (No\nContent).\"\n\nChange-Id: Ib1142245bb1a5c87408978b30dce694c5b520fbb\nRelated-Bug: 1537811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e19177590f29581c04456f243c63b24be9b58d6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e19177590f29581c04456f243c63b24be9b58d6e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"781f2b7a1ce71a041cfd81a946f9acc05084feff","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"87300578be41f29f90593da4965cac8ccadcecc4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2016-05-12 10:54:09.000000000","uploader":{"_account_id":18940,"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","username":"gyurco78"},"ref":"refs/changes/52/275652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/52/275652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/275652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/275652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/275652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/52/275652/3"}}},"commit":{"parents":[{"commit":"9d0424960291f070d31ef122ff6350937ddd98aa","subject":"Skip test_resize_volume_backed_server_confirm for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9d0424960291f070d31ef122ff6350937ddd98aa"}]}],"author":{"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","date":"2016-02-03 11:57:16.000000000","tz":60},"committer":{"name":"Gyorgy Szombathelyi","email":"gyurco@freemail.hu","date":"2016-05-12 10:54:08.000000000","tz":0},"subject":"Check Content-Length header according to RFC 7230","message":"Check Content-Length header according to RFC 7230\n\n\"A server MUST NOT send a Content-Length header field in any\nresponse with a status code of 1xx (Informational) or 204 (No\nContent).\"\n\nChange-Id: Ib1142245bb1a5c87408978b30dce694c5b520fbb\nRelated-Bug: 1537811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/87300578be41f29f90593da4965cac8ccadcecc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/87300578be41f29f90593da4965cac8ccadcecc4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9d0424960291f070d31ef122ff6350937ddd98aa","is_merged_in_target_branch":true,"change_id":"I1f334f43175fe56bdbe6bd35655ce671df7600a7","change_number":315106,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
