)]}'
{"id":"openstack%2Ftempest~28416","triplet_id":"openstack%2Ftempest~master~I81cc39fed392fce9dfbda5683ffba29eeba936be","project":"openstack/tempest","branch":"master","topic":"master","hashtags":[],"change_id":"I81cc39fed392fce9dfbda5683ffba29eeba936be","subject":"Adding new test for iSCSI live block migration","status":"MERGED","created":"2013-05-07 14:50:14.000000000","updated":"2013-05-18 18:54:11.000000000","submitted":"2013-05-18 18:54:11.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f63eb0e61e46a1b2e5d2d53470245c60dc170375","_number":28416,"virtual_id_number":28416,"owner":{"_account_id":6735,"name":"Bob Ball","email":"bob.ball@citrix.com","username":"bob-ball"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-05-18 18:54:11.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":6735,"name":"Bob Ball","email":"bob.ball@citrix.com","username":"bob-ball"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-05-18 16:24:35.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":2,"date":"2013-05-14 19:49:11.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":1,"date":"2013-05-14 09:42:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":6735,"name":"Bob Ball","email":"bob.ball@citrix.com","username":"bob-ball"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-05-18 16:24:35.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"date":"2013-05-14 19:49:11.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":6735,"name":"Bob Ball","email":"bob.ball@citrix.com","username":"bob-ball"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":6735,"name":"Bob Ball","email":"bob.ball@citrix.com","username":"bob-ball"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6735,"name":"Bob Ball","email":"bob.ball@citrix.com","username":"bob-ball"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-05-07 14:54:57.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2013-05-07 14:56:38.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2013-05-12 10:48:29.000000000","updated_by":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"reviewer":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"state":"REVIEWER"},{"updated":"2013-05-14 09:42:30.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2013-05-14 19:49:11.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2013-05-18 16:24:35.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2013-05-18 18:54:11.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"bf572db71ff4cac1d875a05e9e2da7d887b10449","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-07 15:36:47.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/28416/1/check/gate-tempest-pep8/3162 : FAILURE in 00:04:13\n- http://logs.openstack.org/28416/1/check/gate-tempest-devstack-vm-full/14838 : SUCCESS in 00:45:07\n- http://logs.openstack.org/28416/1/check/gate-tempest-devstack-vm-postgres-full/14553 : FAILURE in 00:45:04 (non-voting)\n- http://logs.openstack.org/28416/1/check/gate-tempest-devstack-vm-quantum/21053 : SUCCESS in 00:20:24\n- http://logs.openstack.org/28416/1/check/gate-tempest-devstack-vm-quantum-full/1311 : FAILURE in 00:44:44 (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"196b44f340415cc3af671db63172f40c9fef1383","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-05-07 15:39:53.000000000","message":"Patch Set 1: (2 inline comments)\n\nSee the comments in-line.","accounts_in_message":[],"_revision_number":1},{"id":"91767ee58e3b9762daae36328d3b3dd5fe602d69","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-05-07 15:40:05.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n","accounts_in_message":[],"_revision_number":1},{"id":"474a50c8e9a2e62a7c3d7a9222f51326a33185f4","author":{"_account_id":6735,"name":"Bob Ball","email":"bob.ball@citrix.com","username":"bob-ball"},"date":"2013-05-07 16:07:44.000000000","message":"Patch Set 1: (2 inline comments)\n\nForgot to run pep8 - new commit on it\u0027s way.","accounts_in_message":[],"_revision_number":1},{"id":"00445632514ba8ae96138d956c3ea776ec4046a3","author":{"_account_id":6735,"name":"Bob Ball","email":"bob.ball@citrix.com","username":"bob-ball"},"date":"2013-05-08 08:51:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b6559af4fc66cf3222e0630a0f6c616f372769b","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-05-08 09:23:29.000000000","message":"Patch Set 2: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"140b45b541d4077928cba325a63f32f43a043544","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-08 09:40:06.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28416/2/check/gate-tempest-pep8/3175 : SUCCESS in 00:03:09\n- http://logs.openstack.org/28416/2/check/gate-tempest-devstack-vm-full/14979 : SUCCESS in 00:44:21\n- http://logs.openstack.org/28416/2/check/gate-tempest-devstack-vm-postgres-full/14694 : FAILURE in 00:44:45 (non-voting)\n- http://logs.openstack.org/28416/2/check/gate-tempest-devstack-vm-quantum/21208 : SUCCESS in 00:20:58\n- http://logs.openstack.org/28416/2/check/gate-tempest-devstack-vm-quantum-full/1324 : FAILURE in 00:48:18 (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"c2aa7467818ef66031bf8a7c93ab1515d6dfc94f","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2013-05-11 09:34:22.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n(2 inline comments)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"8fe945927d011c6ef8b827a4c7061e0a3c68acd4","author":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"date":"2013-05-12 10:48:29.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"adb184b2acbc49746b31cce28154aa209a1f9de8","author":{"_account_id":6735,"name":"Bob Ball","email":"bob.ball@citrix.com","username":"bob-ball"},"date":"2013-05-14 08:55:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e41a3d155dc9a5178d5f54a94e53b4909d617d10","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-14 09:42:07.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28416/3/check/gate-tempest-pep8/3294 : SUCCESS in 00:03:28\n- http://logs.openstack.org/28416/3/check/gate-tempest-devstack-vm-full/15842 : SUCCESS in 00:45:21\n- http://logs.openstack.org/28416/3/check/gate-tempest-devstack-vm-postgres-full/15558 : FAILURE in 00:45:40 (non-voting)\n- http://logs.openstack.org/28416/3/check/gate-tempest-devstack-vm-quantum/22277 : SUCCESS in 00:21:33\n- http://logs.openstack.org/28416/3/check/gate-tempest-devstack-vm-quantum-full/1429 : FAILURE in 00:46:05 (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"e295dac2b2d264237ea1d0c64b9b54c6b784fd9d","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2013-05-14 09:42:30.000000000","message":"Patch Set 3: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":3},{"id":"405054a59a854ed53a201766f8962206f3ee0654","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-05-14 19:49:11.000000000","message":"Patch Set 3: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"86ceee007f51e11e80641c94c20f18664531b2f8","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2013-05-18 16:24:35.000000000","message":"Patch Set 3: Looks good to me (core reviewer); Approved\n\nCan\u0027t find anything wrong with this patch at all, and I went over it with a pretty fine-toothed comb.\n\nNice work, Bob.\n\n-jay","accounts_in_message":[],"_revision_number":3},{"id":"fb3fee24e2dc4c33bbee8c2ed41b87d77c352da1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-18 16:52:53.000000000","message":"Patch Set 3: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"b187766e2a9c13a06b700199644d72fb4e2dc68d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-18 18:54:11.000000000","message":"Patch Set 3: Verified\n\nBuild succeeded.\n\n- http://logs.openstack.org/28416/3/gate/gate-tempest-pep8/3373 : SUCCESS in 3m 02s\n- http://logs.openstack.org/28416/3/gate/gate-tempest-devstack-vm-full/16716 : SUCCESS in 49m 42s\n- http://logs.openstack.org/28416/3/gate/gate-tempest-devstack-vm-postgres-full/16431 : FAILURE in 41m 39s (non-voting)\n- http://logs.openstack.org/28416/3/gate/gate-tempest-devstack-vm-quantum/23294 : SUCCESS in 21m 47s\n- http://logs.openstack.org/28416/3/gate/gate-tempest-devstack-vm-quantum-full/1508 : FAILURE in 38m 44s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"8b1c156b783428ca0aed128237191c15ff064298","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-18 18:54:11.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c078be9baba5df8b598f9905634627e64acdd6b1","revisions":{"11c20e0d37dfa244480491cf6b608c69000bc835":{"kind":"REWORK","_number":1,"created":"2013-05-07 14:50:14.000000000","uploader":{"_account_id":6735,"name":"Bob Ball","email":"bob.ball@citrix.com","username":"bob-ball"},"ref":"refs/changes/16/28416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/16/28416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/28416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/28416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/28416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/16/28416/1"}}},"commit":{"parents":[{"commit":"ad979a97f064c93a28547491cb3b16b2ab0de181","subject":"Merge \"Adds Cinder Multi-Backend Test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad979a97f064c93a28547491cb3b16b2ab0de181"}]}],"author":{"name":"Bob Ball","email":"bob.ball@citrix.com","date":"2013-04-09 13:25:00.000000000","tz":60},"committer":{"name":"Bob Ball","email":"bob.ball@citrix.com","date":"2013-05-07 14:49:43.000000000","tz":60},"subject":"Adding new test for iSCSI live block migration","message":"Adding new test for iSCSI live block migration\n\nRequrires a new config value as the test depends on a\nspecific patch being made to XAPI to support the\nrelax-xsm-sr-check behaviour\n\nChange-Id: I81cc39fed392fce9dfbda5683ffba29eeba936be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/11c20e0d37dfa244480491cf6b608c69000bc835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/11c20e0d37dfa244480491cf6b608c69000bc835"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ad979a97f064c93a28547491cb3b16b2ab0de181","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6146f64c759347ea324bc089988dfecde1be0d8e":{"kind":"REWORK","_number":2,"created":"2013-05-08 08:51:33.000000000","uploader":{"_account_id":6735,"name":"Bob Ball","email":"bob.ball@citrix.com","username":"bob-ball"},"ref":"refs/changes/16/28416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/16/28416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/28416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/28416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/28416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/16/28416/2"}}},"commit":{"parents":[{"commit":"96e9e785b545de3e7fe740d9130d04fd37c25eb7","subject":"Merge \"Update service test case - V3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/96e9e785b545de3e7fe740d9130d04fd37c25eb7"}]}],"author":{"name":"Bob Ball","email":"bob.ball@citrix.com","date":"2013-04-09 13:25:00.000000000","tz":60},"committer":{"name":"Bob Ball","email":"bob.ball@citrix.com","date":"2013-05-08 08:51:30.000000000","tz":60},"subject":"Adding new test for iSCSI live block migration","message":"Adding new test for iSCSI live block migration\n\nRequrires a new config value as the test depends on a\nspecific patch being made to XAPI to support the\nrelax-xsm-sr-check behaviour\n\nChange-Id: I81cc39fed392fce9dfbda5683ffba29eeba936be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6146f64c759347ea324bc089988dfecde1be0d8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6146f64c759347ea324bc089988dfecde1be0d8e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"96e9e785b545de3e7fe740d9130d04fd37c25eb7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c078be9baba5df8b598f9905634627e64acdd6b1":{"kind":"REWORK","_number":3,"created":"2013-05-14 08:55:45.000000000","uploader":{"_account_id":6735,"name":"Bob Ball","email":"bob.ball@citrix.com","username":"bob-ball"},"ref":"refs/changes/16/28416/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/16/28416/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/28416/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/28416/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/28416/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/16/28416/3"}}},"commit":{"parents":[{"commit":"96e9e785b545de3e7fe740d9130d04fd37c25eb7","subject":"Merge \"Update service test case - V3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/96e9e785b545de3e7fe740d9130d04fd37c25eb7"}]}],"author":{"name":"Bob Ball","email":"bob.ball@citrix.com","date":"2013-04-09 13:25:00.000000000","tz":60},"committer":{"name":"Bob Ball","email":"bob.ball@citrix.com","date":"2013-05-14 08:55:11.000000000","tz":60},"subject":"Adding new test for iSCSI live block migration","message":"Adding new test for iSCSI live block migration\n\nRequrires a new config value as the test depends on a\nspecific patch being made to XAPI to support the\nrelax-xsm-sr-check behaviour\n\nChange-Id: I81cc39fed392fce9dfbda5683ffba29eeba936be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c078be9baba5df8b598f9905634627e64acdd6b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c078be9baba5df8b598f9905634627e64acdd6b1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"96e9e785b545de3e7fe740d9130d04fd37c25eb7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
