)]}'
{"id":"openstack%2Ftempest~28893","triplet_id":"openstack%2Ftempest~stable%2Fgrizzly~Idb4a5a0b5eeca266603db8737d65120864de1c6a","project":"openstack/tempest","branch":"stable/grizzly","hashtags":[],"change_id":"Idb4a5a0b5eeca266603db8737d65120864de1c6a","subject":"Put the logic from devstack gate into tox.","status":"MERGED","created":"2013-05-11 20:00:04.000000000","updated":"2013-05-11 23:10:33.000000000","submitted":"2013-05-11 23:10:33.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f94b9e7349c6bb42f812c1762c6ccc6fd9c362e8","_number":28893,"virtual_id_number":28893,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2013-05-11 23:10:33.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-05-11 22:06:00.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-05-11 22:06:00.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-05-11 22:06:00.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2013-05-11 23:10:33.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ec17ead81c593f70ed0de0b4cc44c03745e2c2c6","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2013-05-11 20:02:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ddb343c6b978b6355e7cdb38dcb08d7e7992fc99","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-11 20:44:02.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/28893/2/check/gate-tempest-pep8/3244 : FAILURE in 00:01:54\n- http://logs.openstack.org/28893/2/check/gate-tempest-devstack-vm-full/15557 : SUCCESS in 00:39:56\n- http://logs.openstack.org/28893/2/check/gate-tempest-devstack-vm-postgres-full/15273 : SUCCESS in 00:41:14 (non-voting)\n- http://logs.openstack.org/28893/2/check/gate-tempest-devstack-vm-quantum/21903 : SUCCESS in 00:19:06\n","accounts_in_message":[],"_revision_number":2},{"id":"a6a2ba80f8b377704726f7719ae790e30d2d1d30","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2013-05-11 21:01:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"65e9dd23efa80f9ed838d9d0d8256fadad9ceb19","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-11 21:38:29.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28893/3/check/gate-tempest-pep8/3246 : SUCCESS in 00:02:06\n- http://logs.openstack.org/28893/3/check/gate-tempest-devstack-vm-full/15564 : SUCCESS in 00:36:38\n- http://logs.openstack.org/28893/3/check/gate-tempest-devstack-vm-postgres-full/15280 : SUCCESS in 00:36:15 (non-voting)\n- http://logs.openstack.org/28893/3/check/gate-tempest-devstack-vm-quantum/21913 : SUCCESS in 00:19:09\n","accounts_in_message":[],"_revision_number":3},{"id":"22052262d473704eea3b23b994bf07275eb86d80","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-05-11 22:06:00.000000000","message":"Patch Set 3: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":3},{"id":"ab91ff2ec5cebe18613b7e83356b529c2badba33","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-11 22:33:43.000000000","message":"Patch Set 3: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"f5001d489cc36c589b6aaa270b2205808803b7fa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-11 23:10:33.000000000","message":"Patch Set 3: Verified\n\nBuild succeeded.\n\n- http://logs.openstack.org/28893/3/gate/gate-tempest-pep8/3248 : SUCCESS in 00:03:20\n- http://logs.openstack.org/28893/3/gate/gate-tempest-devstack-vm-full/15570 : SUCCESS in 00:33:06\n- http://logs.openstack.org/28893/3/gate/gate-tempest-devstack-vm-postgres-full/15286 : SUCCESS in 00:36:40 (non-voting)\n- http://logs.openstack.org/28893/3/gate/gate-tempest-devstack-vm-quantum/21919 : SUCCESS in 00:19:09\n","accounts_in_message":[],"_revision_number":3},{"id":"f94b9e7349c6bb42f812c1762c6ccc6fd9c362e8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-11 23:10:33.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b3f318082ac850408ee178fabf485b161a8012cc","revisions":{"4069476271162204f70a2b30089375f8c228d4ab":{"kind":"REWORK","_number":1,"created":"2013-05-11 20:00:04.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/93/28893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/28893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/28893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/28893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/28893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/28893/1"}}},"commit":{"parents":[{"commit":"d36ae895ba73bd9e31dc884ea2fb631de63fc91e","subject":"delete servers in setUpClass if ecxeption raised","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d36ae895ba73bd9e31dc884ea2fb631de63fc91e"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-05-02 18:33:18.000000000","tz":-240},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-05-11 19:59:35.000000000","tz":-240},"subject":"Put the logic from devstack gate into tox.","message":"Put the logic from devstack gate into tox.\n\nChange-Id: Idb4a5a0b5eeca266603db8737d65120864de1c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4069476271162204f70a2b30089375f8c228d4ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4069476271162204f70a2b30089375f8c228d4ab"}]},"branch":"refs/heads/stable/grizzly"},"d6c2aeab272ffb1be45e44111c7512815df01005":{"kind":"REWORK","_number":2,"created":"2013-05-11 20:02:35.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/93/28893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/28893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/28893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/28893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/28893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/28893/2"}}},"commit":{"parents":[{"commit":"d36ae895ba73bd9e31dc884ea2fb631de63fc91e","subject":"delete servers in setUpClass if ecxeption raised","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d36ae895ba73bd9e31dc884ea2fb631de63fc91e"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-05-02 18:33:18.000000000","tz":-240},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-05-11 20:02:22.000000000","tz":-240},"subject":"Put the logic from devstack gate into tox.","message":"Put the logic from devstack gate into tox.\n\nChange-Id: Idb4a5a0b5eeca266603db8737d65120864de1c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d6c2aeab272ffb1be45e44111c7512815df01005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d6c2aeab272ffb1be45e44111c7512815df01005"}]},"branch":"refs/heads/stable/grizzly"},"b3f318082ac850408ee178fabf485b161a8012cc":{"kind":"REWORK","_number":3,"created":"2013-05-11 21:01:40.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/93/28893/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/28893/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/28893/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/28893/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/28893/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/28893/3"}}},"commit":{"parents":[{"commit":"d36ae895ba73bd9e31dc884ea2fb631de63fc91e","subject":"delete servers in setUpClass if ecxeption raised","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d36ae895ba73bd9e31dc884ea2fb631de63fc91e"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-05-02 18:33:18.000000000","tz":-240},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-05-11 21:01:30.000000000","tz":-240},"subject":"Put the logic from devstack gate into tox.","message":"Put the logic from devstack gate into tox.\n\nChange-Id: Idb4a5a0b5eeca266603db8737d65120864de1c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b3f318082ac850408ee178fabf485b161a8012cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b3f318082ac850408ee178fabf485b161a8012cc"}]},"branch":"refs/heads/stable/grizzly"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
