)]}'
{"id":"openstack%2Ftempest~28894","triplet_id":"openstack%2Ftempest~stable%2Ffolsom~Idb4a5a0b5eeca266603db8737d65120864de1c6a","project":"openstack/tempest","branch":"stable/folsom","hashtags":[],"change_id":"Idb4a5a0b5eeca266603db8737d65120864de1c6a","subject":"Put the logic from devstack gate into tox.","status":"MERGED","created":"2013-05-11 20:01:22.000000000","updated":"2013-05-11 22:48:12.000000000","submitted":"2013-05-11 22:48:12.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6dd9dd4855adb407c569e50a4a2a92ad2ab782cf","_number":28894,"virtual_id_number":28894,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2013-05-11 22:48:12.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-05-11 22:05:51.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"all":[{"value":0,"date":"2013-05-11 20:07:59.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-05-11 22:05:51.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-05-11 22:05:51.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2013-05-11 22:48:12.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"2d96584115d575ad0e313e4e81a8356472b667d1","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2013-05-11 20:07:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2d1b00ef3a4ac3b85b462595366698237c782bb2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-11 20:20:11.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28894/2/check/gate-tempest-pep8/3245 : SUCCESS in 00:00:09\n- http://logs.openstack.org/28894/2/check/gate-tempest-devstack-vm-smoke/562 : SUCCESS in 00:09:52\n- http://logs.openstack.org/28894/2/check/gate-tempest-devstack-vm-cinder/6383 : SUCCESS in 00:11:59\n","accounts_in_message":[],"_revision_number":2},{"id":"ee6a326a538093b2fcf7acda57085c9c2d68e794","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-05-11 22:05:51.000000000","message":"Patch Set 2: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":2},{"id":"d9f507fd92d781b091b20dd318ba0940909001b1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-11 22:33:39.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"584e4432a2d9611c576167d8eafa16d30dfc9d90","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-11 22:48:12.000000000","message":"Patch Set 2: Verified\n\nBuild succeeded.\n\n- http://logs.openstack.org/28894/2/gate/gate-tempest-pep8/3247 : SUCCESS in 00:00:10\n- http://logs.openstack.org/28894/2/gate/gate-tempest-devstack-vm-smoke/563 : SUCCESS in 00:13:59\n- http://logs.openstack.org/28894/2/gate/gate-tempest-devstack-vm-cinder/6384 : SUCCESS in 00:14:25\n","accounts_in_message":[],"_revision_number":2},{"id":"6dd9dd4855adb407c569e50a4a2a92ad2ab782cf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-11 22:48:12.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"17dfb1b0ef2eff686dcefc46ed0f8405a45d2cc3","revisions":{"bcdb16409c7386d2dc7a9fc79db009c4b45491fd":{"kind":"REWORK","_number":1,"created":"2013-05-11 20:01:22.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/94/28894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/28894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/28894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/28894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/28894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/28894/1"}}},"commit":{"parents":[{"commit":"481116d19e3f81a8ef144baf8e2868f569a13e7e","subject":"Do super()\u0027s tearDowns last (bug 1178337)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/481116d19e3f81a8ef144baf8e2868f569a13e7e"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-05-02 18:33:18.000000000","tz":-240},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-05-11 20:01:12.000000000","tz":-240},"subject":"Put the logic from devstack gate into tox.","message":"Put the logic from devstack gate into tox.\n\nChange-Id: Idb4a5a0b5eeca266603db8737d65120864de1c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bcdb16409c7386d2dc7a9fc79db009c4b45491fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bcdb16409c7386d2dc7a9fc79db009c4b45491fd"}]},"branch":"refs/heads/stable/folsom"},"17dfb1b0ef2eff686dcefc46ed0f8405a45d2cc3":{"kind":"REWORK","_number":2,"created":"2013-05-11 20:07:59.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/94/28894/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/28894/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/28894/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/28894/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/28894/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/28894/2"}}},"commit":{"parents":[{"commit":"481116d19e3f81a8ef144baf8e2868f569a13e7e","subject":"Do super()\u0027s tearDowns last (bug 1178337)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/481116d19e3f81a8ef144baf8e2868f569a13e7e"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-05-02 18:33:18.000000000","tz":-240},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-05-11 20:02:51.000000000","tz":-240},"subject":"Put the logic from devstack gate into tox.","message":"Put the logic from devstack gate into tox.\n\nChange-Id: Idb4a5a0b5eeca266603db8737d65120864de1c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/17dfb1b0ef2eff686dcefc46ed0f8405a45d2cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/17dfb1b0ef2eff686dcefc46ed0f8405a45d2cc3"}]},"branch":"refs/heads/stable/folsom"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
