)]}'
{"id":"openstack%2Ftempest~29240","triplet_id":"openstack%2Ftempest~master~I4bae858dfc5fba2c45e869a58e89fe7c14c3b30b","project":"openstack/tempest","branch":"master","topic":"bp/set-gate-attribute","hashtags":[],"change_id":"I4bae858dfc5fba2c45e869a58e89fe7c14c3b30b","subject":"Set smoke/gate attributes for tests in \"image\"","status":"MERGED","created":"2013-05-15 13:46:20.000000000","updated":"2013-05-20 11:34:09.000000000","submitted":"2013-05-20 11:34:09.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bcb4c0bb3576a3a8d91473207272bf5d24725e20","_number":29240,"virtual_id_number":29240,"owner":{"_account_id":4598,"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","username":"lauria"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-05-20 11:34:09.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4598,"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","username":"lauria"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4598,"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","username":"lauria"},{"value":2,"date":"2013-05-19 07:40:53.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":1,"date":"2013-05-15 16:26:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":2,"date":"2013-05-20 10:42:54.000000000","_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":1,"date":"2013-05-15 16:36:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-05-15 15:17:12.000000000","_account_id":4598,"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","username":"lauria"},{"value":0,"date":"2013-05-19 07:40:53.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":1,"date":"2013-05-20 10:42:54.000000000","_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4598,"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","username":"lauria"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4598,"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","username":"lauria"},{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-05-15 16:26:57.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2013-05-15 16:36:00.000000000","updated_by":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"reviewer":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"state":"REVIEWER"},{"updated":"2013-05-19 07:40:53.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2013-05-20 10:42:54.000000000","updated_by":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"reviewer":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"state":"REVIEWER"},{"updated":"2013-05-20 11:34:09.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"e7efc033e4f71472b1036c14336dafed870556a8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-15 14:39:26.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/29240/1/check/gate-tempest-pep8/3320 : FAILURE in 4m 03s\n- http://logs.openstack.org/29240/1/check/gate-tempest-devstack-vm-full/16074 : FAILURE in 42m 36s\n- http://logs.openstack.org/29240/1/check/gate-tempest-devstack-vm-postgres-full/15790 : FAILURE in 52m 52s (non-voting)\n- http://logs.openstack.org/29240/1/check/gate-tempest-devstack-vm-quantum/22558 : FAILURE in 23m 27s\n- http://logs.openstack.org/29240/1/check/gate-tempest-devstack-vm-quantum-full/1455 : FAILURE in 48m 02s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"9b6cf4931edf5e639b7c43ca5bd2d398c3532b0f","author":{"_account_id":4598,"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","username":"lauria"},"date":"2013-05-15 15:17:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb8b220ce5a98a95ff159e134703e2bd0347e297","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-15 16:10:10.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/29240/2/check/gate-tempest-pep8/3328 : SUCCESS in 4m 35s\n- http://logs.openstack.org/29240/2/check/gate-tempest-devstack-vm-full/16102 : SUCCESS in 44m 12s\n- http://logs.openstack.org/29240/2/check/gate-tempest-devstack-vm-postgres-full/15818 : FAILURE in 52m 44s (non-voting)\n- http://logs.openstack.org/29240/2/check/gate-tempest-devstack-vm-quantum/22590 : SUCCESS in 24m 17s\n- http://logs.openstack.org/29240/2/check/gate-tempest-devstack-vm-quantum-full/1463 : FAILURE in 46m 55s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"a032b22991a4e96213cfac3f0b0e6ebb644978bb","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2013-05-15 16:26:57.000000000","message":"Patch Set 2: Looks good to me, but someone else must approve\n\nI think these will all be subject to a future review to mark properly which tests are smoke and which aren\u0027t (if we really want to get \u003c10mins)","accounts_in_message":[],"_revision_number":2},{"id":"80e5300e734048f6abcf144cfd4dc94d3eab96da","author":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"date":"2013-05-15 16:36:00.000000000","message":"Patch Set 2: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":2},{"id":"302735eb11ea88664b6f10574b37641ed733b6b1","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-05-19 07:40:53.000000000","message":"Patch Set 2: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"e176911ee13e91ea83694b090535f6c1998e6ad7","author":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"date":"2013-05-20 10:42:54.000000000","message":"Patch Set 2: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":2},{"id":"1a8c23b3d84401dc7d25bc299cf527d9a2f91d7c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-20 10:42:55.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"671260e97071b8e1c8d3e0f2e8bd707fc6bba058","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-20 11:34:09.000000000","message":"Patch Set 2: Verified\n\nBuild succeeded.\n\n- http://logs.openstack.org/29240/2/gate/gate-tempest-pep8/3386 : SUCCESS in 2m 37s\n- http://logs.openstack.org/29240/2/gate/gate-tempest-devstack-vm-full/16863 : SUCCESS in 44m 55s\n- http://logs.openstack.org/29240/2/gate/gate-tempest-devstack-vm-postgres-full/16578 : FAILURE in 51m 05s (non-voting)\n- http://logs.openstack.org/29240/2/gate/gate-tempest-devstack-vm-quantum/23470 : SUCCESS in 21m 23s\n- http://logs.openstack.org/29240/2/gate/gate-tempest-devstack-vm-quantum-full/1520 : FAILURE in 37m 45s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"d82dbcc9255beac29d7b82c1f77cd1488d631745","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-20 11:34:09.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fd5f5953ca320ff621b245ad6f9038bd6e258aae","revisions":{"45859608072f3c632a3f301402f3b21cd315c899":{"kind":"REWORK","_number":1,"created":"2013-05-15 13:46:20.000000000","uploader":{"_account_id":4598,"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","username":"lauria"},"ref":"refs/changes/40/29240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/40/29240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/40/29240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/40/29240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/40/29240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/40/29240/1"}}},"commit":{"parents":[{"commit":"0e98675c1751b7543b66b21d0397c0ba43feef35","subject":"Merge \"Permits a list of values for the \"type\u003d\" tests attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e98675c1751b7543b66b21d0397c0ba43feef35"}]}],"author":{"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","date":"2013-05-15 13:44:24.000000000","tz":-240},"committer":{"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","date":"2013-05-15 13:45:55.000000000","tz":-240},"subject":"Set smoke/gate attributes for tests in \"image\"","message":"Set smoke/gate attributes for tests in \"image\"\n\nChange-Id: I4bae858dfc5fba2c45e869a58e89fe7c14c3b30b\nImplements: blueprint set-gate-attribute\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/45859608072f3c632a3f301402f3b21cd315c899"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/45859608072f3c632a3f301402f3b21cd315c899"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0e98675c1751b7543b66b21d0397c0ba43feef35","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"fd5f5953ca320ff621b245ad6f9038bd6e258aae":{"kind":"REWORK","_number":2,"created":"2013-05-15 15:17:12.000000000","uploader":{"_account_id":4598,"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","username":"lauria"},"ref":"refs/changes/40/29240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/40/29240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/40/29240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/40/29240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/40/29240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/40/29240/2"}}},"commit":{"parents":[{"commit":"0e98675c1751b7543b66b21d0397c0ba43feef35","subject":"Merge \"Permits a list of values for the \"type\u003d\" tests attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e98675c1751b7543b66b21d0397c0ba43feef35"}]}],"author":{"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","date":"2013-05-15 13:44:24.000000000","tz":-240},"committer":{"name":"Giampaolo Lauria","email":"lauria@us.ibm.com","date":"2013-05-15 15:16:42.000000000","tz":-240},"subject":"Set smoke/gate attributes for tests in \"image\"","message":"Set smoke/gate attributes for tests in \"image\"\n\nChange-Id: I4bae858dfc5fba2c45e869a58e89fe7c14c3b30b\nImplements: blueprint set-gate-attribute\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fd5f5953ca320ff621b245ad6f9038bd6e258aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fd5f5953ca320ff621b245ad6f9038bd6e258aae"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0e98675c1751b7543b66b21d0397c0ba43feef35","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
