)]}'
{"id":"openstack%2Ftempest~29771","triplet_id":"openstack%2Ftempest~master~I173af3f46fc2f6771237ab29b472249b00a9ed2d","project":"openstack/tempest","branch":"master","topic":"deletecont","hashtags":[],"change_id":"I173af3f46fc2f6771237ab29b472249b00a9ed2d","subject":"Moved swift container cleanup to a class method.","status":"MERGED","created":"2013-05-20 13:42:35.000000000","updated":"2013-05-26 01:15:05.000000000","submitted":"2013-05-26 01:15:04.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7010083e335871dbec555f99572f1a477c57dc00","_number":29771,"virtual_id_number":29771,"owner":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-05-26 01:15:04.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":6683,"name":"Pavel Sedlák","email":"psedlak@redhat.com","username":"psedlak"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-05-24 16:43:40.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":6683,"name":"Pavel Sedlák","email":"psedlak@redhat.com","username":"psedlak"},{"value":1,"date":"2013-05-24 14:05:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":2,"date":"2013-05-26 00:26:14.000000000","_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"date":"2013-05-24 13:56:26.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-05-24 16:43:40.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":6683,"name":"Pavel Sedlák","email":"psedlak@redhat.com","username":"psedlak"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":1,"date":"2013-05-26 00:26:14.000000000","_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":6683,"name":"Pavel Sedlák","email":"psedlak@redhat.com","username":"psedlak"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6683,"name":"Pavel Sedlák","email":"psedlak@redhat.com","username":"psedlak"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-05-22 08:48:24.000000000","updated_by":{"_account_id":6683,"name":"Pavel Sedlák","email":"psedlak@redhat.com","username":"psedlak"},"reviewer":{"_account_id":6683,"name":"Pavel Sedlák","email":"psedlak@redhat.com","username":"psedlak"},"state":"REVIEWER"},{"updated":"2013-05-24 14:05:50.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2013-05-24 16:43:40.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2013-05-26 00:26:14.000000000","updated_by":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"reviewer":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"state":"REVIEWER"},{"updated":"2013-05-26 01:15:04.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"8ca4a916932a7215ad351262fe667972c51121e8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-20 14:34:24.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/29771/1/check/gate-tempest-pep8/3391 : SUCCESS in 2m 24s\n- http://logs.openstack.org/29771/1/check/gate-tempest-devstack-vm-full/16895 : SUCCESS in 44m 05s\n- http://logs.openstack.org/29771/1/check/gate-tempest-devstack-vm-postgres-full/16610 : FAILURE in 51m 32s (non-voting)\n- http://logs.openstack.org/29771/1/check/gate-tempest-devstack-vm-quantum/23504 : FAILURE in 26m 45s\n- http://logs.openstack.org/29771/1/check/gate-tempest-devstack-vm-quantum-full/1525 : FAILURE in 38m 33s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"3c8d3928428af6a2ec4660a72b1d843b3d7cc818","author":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"date":"2013-05-21 09:22:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"13b11c49bbcbf6519952ad9b636e35968e779bf9","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2013-05-21 09:34:33.000000000","message":"Patch Set 2:\n\nwhile you\u0027re at it, maybe you can remove the copyright notice from all the files, as per suggestion from the ML","accounts_in_message":[],"_revision_number":2},{"id":"090f902ed9f972216c3b050239d468269a595fb2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-21 10:07:56.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/29771/2/check/gate-tempest-pep8/3400 : SUCCESS in 2m 50s\n- http://logs.openstack.org/29771/2/check/gate-tempest-devstack-vm-full/17251 : SUCCESS in 44m 53s\n- http://logs.openstack.org/29771/2/check/gate-tempest-devstack-vm-postgres-full/16961 : FAILURE in 45m 06s (non-voting)\n- http://logs.openstack.org/29771/2/check/gate-tempest-devstack-vm-quantum/23910 : SUCCESS in 22m 05s\n- http://logs.openstack.org/29771/2/check/gate-tempest-devstack-vm-quantum-full/1534 : FAILURE in 38m 50s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"c36a6b21109bc7df41a1d09fa5814bec994adf0c","author":{"_account_id":6683,"name":"Pavel Sedlák","email":"psedlak@redhat.com","username":"psedlak"},"date":"2013-05-21 10:35:29.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"7396e216b5092149bd1fac9c470aae232a3c2da5","author":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"date":"2013-05-21 12:08:03.000000000","message":"Patch Set 2: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"67c88207902a69e420d720e53723c285970d2f5c","author":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"date":"2013-05-21 12:14:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cfe2c7c80c1ad1744f1d092ecc83d5863506c3a1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-21 13:05:59.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/29771/3/check/gate-tempest-pep8/3403 : SUCCESS in 3m 39s\n- http://logs.openstack.org/29771/3/check/gate-tempest-devstack-vm-full/17279 : SUCCESS in 40m 16s\n- http://logs.openstack.org/29771/3/check/gate-tempest-devstack-vm-postgres-full/16989 : FAILURE in 50m 55s (non-voting)\n- http://logs.openstack.org/29771/3/check/gate-tempest-devstack-vm-quantum/23946 : SUCCESS in 23m 27s\n- http://logs.openstack.org/29771/3/check/gate-tempest-devstack-vm-quantum-full/1537 : FAILURE in 41m 59s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"a7b6649f1866633e9cbe91ffa914301dd67dc5d8","author":{"_account_id":6683,"name":"Pavel Sedlák","email":"psedlak@redhat.com","username":"psedlak"},"date":"2013-05-22 08:48:24.000000000","message":"Patch Set 3: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":3},{"id":"1b583abe0eaf78104b07c2ce7556aa5410b82639","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-05-24 05:18:40.000000000","message":"Patch Set 3: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"e0003e076bff4ee33e782161ed21f2ade2a7b18c","author":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"date":"2013-05-24 09:03:08.000000000","message":"Patch Set 3: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":3},{"id":"90747a4fcbe407515dc40de5dd7d7ebd8b86ada2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-24 09:03:09.000000000","message":"Patch Set 3: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"21e60fd14df360a54fde37712dae78aa1ec46cc8","date":"2013-05-24 09:49:19.000000000","message":"Your change could not be merged due to a path conflict.\n\nPlease merge (or rebase) the change locally and upload the resolution for review.","accounts_in_message":[],"_revision_number":3},{"id":"996571953e903ee6d51640ccd17c7de2c125d7d6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-24 09:49:19.000000000","message":"Patch Set 3: Verified\n\nBuild succeeded.\n\n- http://logs.openstack.org/29771/3/gate/gate-tempest-pep8/3475 : SUCCESS in 1m 14s\n- http://docs-draft.openstack.org/29771/3/gate/gate-tempest-docs/32/doc/build/html/ : SUCCESS in 1m 20s\n- http://logs.openstack.org/29771/3/gate/gate-tempest-devstack-vm-full/18267 : SUCCESS in 43m 40s\n- http://logs.openstack.org/29771/3/gate/gate-tempest-devstack-vm-postgres-full/17977 : FAILURE in 46m 00s (non-voting)\n- http://logs.openstack.org/29771/3/gate/gate-tempest-devstack-vm-quantum/25032 : SUCCESS in 20m 00s\n- http://logs.openstack.org/29771/3/gate/gate-tempest-devstack-vm-quantum-full/1608 : FAILURE in 39m 22s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"fc0ae49e0c8277b2e15bcdd419f1a3c823dad149","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2013-05-24 12:27:24.000000000","message":"Patch Set 3:\n\nTo follow up on your question in #openstack-infra (since you seem to have dropped off IRC for the moment),the Jenkins tests happen using cgit to merge your change while Gerrit\u0027s merge is implemented with jgit instead. Rarely one will be able to merge something the other sees as an unresolvable conflict. Your best bet is to manually rebase this and then get it approved again.","accounts_in_message":[],"_revision_number":3},{"id":"646193bfa433e56456820e023c13d6377bd9aadb","author":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"date":"2013-05-24 13:55:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"07882af31c00ac116d7a292bd1ad9c5519743729","author":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"date":"2013-05-24 13:56:26.000000000","message":"Patch Set 4:\n\nThe newest patch is just a rebase, because the tests/ folder was moved to api/.","accounts_in_message":[],"_revision_number":4},{"id":"9c345e2c3f264c1e7028b639e89039332b0e798a","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2013-05-24 14:05:50.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":4},{"id":"4bad5055ab5925009f3c96f51c6bf7fb535fcd20","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-24 14:46:27.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/29771/4/check/gate-tempest-pep8/3485 : SUCCESS in 1m 19s\n- http://docs-draft.openstack.org/29771/4/check/gate-tempest-docs/42/doc/build/html/ : SUCCESS in 1m 27s\n- http://logs.openstack.org/29771/4/check/gate-tempest-devstack-vm-full/18329 : SUCCESS in 49m 48s\n- http://logs.openstack.org/29771/4/check/gate-tempest-devstack-vm-postgres-full/18039 : FAILURE in 50m 57s (non-voting)\n- http://logs.openstack.org/29771/4/check/gate-tempest-devstack-vm-quantum/25095 : SUCCESS in 20m 30s\n- http://logs.openstack.org/29771/4/check/gate-tempest-devstack-vm-quantum-full/1617 : FAILURE in 42m 27s (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"d973ed11614f0a4e9f78fa8efe1dc1116ebc638b","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-05-24 16:43:40.000000000","message":"Patch Set 4: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":4},{"id":"d0ed0e010b485125acfa4daa314753e55da30ee2","author":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"date":"2013-05-26 00:26:14.000000000","message":"Patch Set 4: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":4},{"id":"1e73a435b3c894f65529e173247af13a16bd3e4c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-26 00:26:16.000000000","message":"Patch Set 4: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"813ab42d9385a12d3e3e2e2fc429200214eb584a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-26 01:15:04.000000000","message":"Patch Set 4: Verified\n\nBuild succeeded.\n\n- http://logs.openstack.org/29771/4/gate/gate-tempest-pep8/3494 : SUCCESS in 1m 17s\n- http://docs-draft.openstack.org/29771/4/gate/gate-tempest-docs/51/doc/build/html/ : SUCCESS in 1m 16s\n- http://logs.openstack.org/29771/4/gate/gate-tempest-devstack-vm-full/18532 : SUCCESS in 44m 11s\n- http://logs.openstack.org/29771/4/gate/gate-tempest-devstack-vm-postgres-full/18242 : FAILURE in 44m 48s (non-voting)\n- http://logs.openstack.org/29771/4/gate/gate-tempest-devstack-vm-quantum/25319 : SUCCESS in 20m 26s\n- http://logs.openstack.org/29771/4/gate/gate-tempest-devstack-vm-quantum-full/1625 : FAILURE in 42m 36s (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"b6c3dcc2ca67f0142e174dd74399cf486e0ab01e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-26 01:15:05.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fd850b9e3cb7816d8afeda22ff9c26719d91af0a","revisions":{"024577874f3384a2f380e96a865086410a7e0293":{"kind":"REWORK","_number":1,"created":"2013-05-20 13:42:35.000000000","uploader":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"ref":"refs/changes/71/29771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/29771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/29771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/29771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/29771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/29771/1"}}},"commit":{"parents":[{"commit":"b8a8ce9fc4f9780abdeb8f8b3b7f6b988dce3967","subject":"Merge \"Add a sleep with back-off to retries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b8a8ce9fc4f9780abdeb8f8b3b7f6b988dce3967"}]}],"author":{"name":"Martina Kollarova","email":"mkollaro@redhat.com","date":"2013-05-20 13:40:13.000000000","tz":120},"committer":{"name":"Martina Kollarova","email":"mkollaro@redhat.com","date":"2013-05-20 13:40:13.000000000","tz":120},"subject":"Moved swift container cleanup to a class method.","message":"Moved swift container cleanup to a class method.\n\nThe same piece of code was used in almost every tearDown method, moved it into\na separate method to remove code redundancy.\n\nChange-Id: I173af3f46fc2f6771237ab29b472249b00a9ed2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/024577874f3384a2f380e96a865086410a7e0293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/024577874f3384a2f380e96a865086410a7e0293"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b8a8ce9fc4f9780abdeb8f8b3b7f6b988dce3967","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ad8c6cb35e4bd64dbfa50ed0c94a69cd62fd0e41":{"kind":"REWORK","_number":2,"created":"2013-05-21 09:22:38.000000000","uploader":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"ref":"refs/changes/71/29771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/29771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/29771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/29771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/29771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/29771/2"}}},"commit":{"parents":[{"commit":"b8a8ce9fc4f9780abdeb8f8b3b7f6b988dce3967","subject":"Merge \"Add a sleep with back-off to retries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b8a8ce9fc4f9780abdeb8f8b3b7f6b988dce3967"}]}],"author":{"name":"Martina Kollarova","email":"mkollaro@redhat.com","date":"2013-05-20 13:40:13.000000000","tz":120},"committer":{"name":"Martina Kollarova","email":"mkollaro@redhat.com","date":"2013-05-21 09:18:06.000000000","tz":120},"subject":"Moved swift container cleanup to a class method.","message":"Moved swift container cleanup to a class method.\n\nThe same piece of code was used in almost every tearDownClass method, moved it\ninto a separate method to remove code redundancy.\n\nChange-Id: I173af3f46fc2f6771237ab29b472249b00a9ed2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad8c6cb35e4bd64dbfa50ed0c94a69cd62fd0e41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad8c6cb35e4bd64dbfa50ed0c94a69cd62fd0e41"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b8a8ce9fc4f9780abdeb8f8b3b7f6b988dce3967","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"720f8226426c9e67d1fab7172cc4a67e7fc2e511":{"kind":"REWORK","_number":3,"created":"2013-05-21 12:14:53.000000000","uploader":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"ref":"refs/changes/71/29771/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/29771/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/29771/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/29771/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/29771/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/29771/3"}}},"commit":{"parents":[{"commit":"b8a8ce9fc4f9780abdeb8f8b3b7f6b988dce3967","subject":"Merge \"Add a sleep with back-off to retries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b8a8ce9fc4f9780abdeb8f8b3b7f6b988dce3967"}]}],"author":{"name":"Martina Kollarova","email":"mkollaro@redhat.com","date":"2013-05-20 13:40:13.000000000","tz":120},"committer":{"name":"Martina Kollarova","email":"mkollaro@redhat.com","date":"2013-05-21 12:14:27.000000000","tz":120},"subject":"Moved swift container cleanup to a class method.","message":"Moved swift container cleanup to a class method.\n\nThe same piece of code was used in almost every tearDownClass method, moved it\ninto a separate method to remove code redundancy.\n\nChange-Id: I173af3f46fc2f6771237ab29b472249b00a9ed2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/720f8226426c9e67d1fab7172cc4a67e7fc2e511"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/720f8226426c9e67d1fab7172cc4a67e7fc2e511"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b8a8ce9fc4f9780abdeb8f8b3b7f6b988dce3967","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"fd850b9e3cb7816d8afeda22ff9c26719d91af0a":{"kind":"REWORK","_number":4,"created":"2013-05-24 13:55:19.000000000","uploader":{"_account_id":6847,"name":"Martina Kollarova","email":"mkollaro@gmail.com","username":"mkollaro"},"ref":"refs/changes/71/29771/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/29771/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/29771/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/29771/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/29771/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/29771/4"}}},"commit":{"parents":[{"commit":"e2153b7eccb22b593aeb67681c3980084f126d9a","subject":"Merge \"Modify hacking flake8 extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e2153b7eccb22b593aeb67681c3980084f126d9a"}]}],"author":{"name":"Martina Kollarova","email":"mkollaro@redhat.com","date":"2013-05-20 13:40:13.000000000","tz":120},"committer":{"name":"Martina Kollarova","email":"mkollaro@redhat.com","date":"2013-05-24 11:23:54.000000000","tz":120},"subject":"Moved swift container cleanup to a class method.","message":"Moved swift container cleanup to a class method.\n\nThe same piece of code was used in almost every tearDownClass method, moved it\ninto a separate method to remove code redundancy.\n\nChange-Id: I173af3f46fc2f6771237ab29b472249b00a9ed2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fd850b9e3cb7816d8afeda22ff9c26719d91af0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fd850b9e3cb7816d8afeda22ff9c26719d91af0a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e2153b7eccb22b593aeb67681c3980084f126d9a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
