)]}'
{"id":"openstack%2Ftempest~32759","triplet_id":"openstack%2Ftempest~master~I498be0b1e3eeea397e1f164caffe7cc88be0ceda","project":"openstack/tempest","branch":"master","topic":"bug/1180972","hashtags":[],"change_id":"I498be0b1e3eeea397e1f164caffe7cc88be0ceda","subject":"Remove basic_auth strategy","status":"MERGED","created":"2013-06-12 17:54:22.000000000","updated":"2013-06-20 16:33:56.000000000","submitted":"2013-06-20 16:33:55.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ff786e173f4de003c5974ee8c7a383179a7c3c32","_number":32759,"virtual_id_number":32759,"owner":{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":2,"date":"2013-06-20 16:33:55.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"recommended":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"all":[{"value":1,"date":"2013-06-12 22:42:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-06-12 21:44:46.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"date":"2013-06-20 14:47:52.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},{"value":2,"date":"2013-06-15 10:42:20.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-06-12 21:44:46.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"date":"2013-06-20 14:47:52.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},{"value":1,"date":"2013-06-15 10:42:20.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-06-12 21:44:46.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2013-06-12 22:42:41.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2013-06-15 10:42:20.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2013-06-20 14:47:52.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2013-06-20 16:33:55.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"a410634935df564dea3030754e5a30d2fc0f6bfd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-12 18:14:30.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/32759/1/check/gate-tempest-pep8/3708 : FAILURE in 1m 31s\n- http://logs.openstack.org/32759/1/check/gate-tempest-docs/264 : SUCCESS in 1m 19s\n- http://logs.openstack.org/32759/1/check/gate-tempest-devstack-vm-full/22647 : FAILURE in 13m 25s\n- http://logs.openstack.org/32759/1/check/gate-tempest-devstack-vm-postgres-full/22354 : FAILURE in 13m 25s\n- http://logs.openstack.org/32759/1/check/gate-tempest-devstack-vm-quantum/29939 : FAILURE in 12m 22s\n- http://logs.openstack.org/32759/1/check/gate-tempest-devstack-vm-quantum-full/1837 : FAILURE in 19m 51s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"059974168cd1fb15b723f9f904abde265dce5218","author":{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},"date":"2013-06-12 18:30:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e5debac17db4e14c483a74e14380177d3ff6a522","author":{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},"date":"2013-06-12 18:38:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"95f83af4dd89683357f41fad4cd770461ff06f3f","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-06-12 18:38:19.000000000","message":"Patch Set 3:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":3},{"id":"348ac429fa37051808501203992fe00b76e80b01","author":{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},"date":"2013-06-12 18:40:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cf7c22f0d6e6980da4c384df58af348f7f15007c","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-06-12 18:41:07.000000000","message":"Patch Set 4:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":4},{"id":"25d8dd6b5a97400f01cbc8ab8a84c04ed0b87f8c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-12 19:27:01.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/32759/4/check/gate-tempest-pep8/3711 : SUCCESS in 1m 18s\n- http://logs.openstack.org/32759/4/check/gate-tempest-docs/267 : SUCCESS in 1m 17s\n- http://logs.openstack.org/32759/4/check/gate-tempest-devstack-vm-full/22661 : SUCCESS in 40m 21s\n- http://logs.openstack.org/32759/4/check/gate-tempest-devstack-vm-postgres-full/22368 : SUCCESS in 41m 00s\n- http://logs.openstack.org/32759/4/check/gate-tempest-devstack-vm-quantum/29954 : SUCCESS in 19m 44s\n- http://logs.openstack.org/32759/4/check/gate-tempest-devstack-vm-quantum-full/1840 : FAILURE in 45m 48s (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"ea94687d89e64f3319a0535e85bb09570a80f547","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-06-12 21:44:46.000000000","message":"Patch Set 4: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":4},{"id":"e62b12a08b3edb271886866e924e61b7bb6dc822","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2013-06-12 22:42:41.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":4},{"id":"22381e8807fb783d7abc8bf01035d178f0cbc453","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-06-14 09:23:09.000000000","message":"Patch Set 4:\n\nI hope one day we can add back the strategy pattern for keystone v2 and v3. But imho it has to be implemented in different place.","accounts_in_message":[],"_revision_number":4},{"id":"b0298e6c48c2491b4b50d07ecb2325188e61b8ad","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-06-15 10:42:20.000000000","message":"Patch Set 4: Looks good to me (core reviewer); Approved\n\nnice cleanup","accounts_in_message":[],"_revision_number":4},{"id":"3417688601407546bb37e9805260104ff60dac90","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-15 10:42:22.000000000","message":"Patch Set 4: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"19b9e6e12324167e4c01c24c12ddbd1997196dd0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-15 11:32:44.000000000","message":"Patch Set 4: Fails\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/32759/4/gate/gate-tempest-pep8/3724 : SUCCESS in 2m 28s\n- http://logs.openstack.org/32759/4/gate/gate-tempest-docs/279 : SUCCESS in 1m 22s\n- http://logs.openstack.org/32759/4/gate/gate-tempest-devstack-vm-full/23561 : SUCCESS in 48m 40s\n- http://logs.openstack.org/32759/4/gate/gate-tempest-devstack-vm-postgres-full/23267 : SUCCESS in 50m 19s\n- http://logs.openstack.org/32759/4/gate/gate-tempest-devstack-vm-quantum/30927 : FAILURE in 23m 35s\n- http://logs.openstack.org/32759/4/gate/gate-tempest-devstack-vm-quantum-full/1852 : FAILURE in 39m 20s (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"fda9d9149d440e9d40d447f6ecada5d6ef1fbca6","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2013-06-20 14:47:52.000000000","message":"Patch Set 4:\n\nreverify no bug","accounts_in_message":[],"_revision_number":4},{"id":"0c5ea749a6b055df4b5c197df1bba946fd2557eb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-20 14:47:54.000000000","message":"Patch Set 4: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"1181b16f01433b9069e9487e5aaea674e3e0bd4a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-20 16:33:55.000000000","message":"Patch Set 4: Verified\n\nBuild succeeded.\n\n- http://logs.openstack.org/32759/4/gate/gate-tempest-pep8/3826 : SUCCESS in 1m 33s\n- http://logs.openstack.org/32759/4/gate/gate-tempest-docs/375 : SUCCESS in 1m 29s\n- http://logs.openstack.org/32759/4/gate/gate-tempest-devstack-vm-full/25016 : SUCCESS in 45m 22s\n- http://logs.openstack.org/32759/4/gate/gate-tempest-devstack-vm-postgres-full/24722 : SUCCESS in 56m 06s\n- http://logs.openstack.org/32759/4/gate/gate-tempest-devstack-vm-quantum/32530 : SUCCESS in 22m 46s\n- http://logs.openstack.org/32759/4/gate/gate-tempest-devstack-vm-quantum-full/1949 : FAILURE in 48m 54s (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"ce02ba7bb0fdf5e15913d33a9c1c1b8af1389ee4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-20 16:33:56.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"216550f81ac7c2d749946dbf466170febcadfe76","revisions":{"276731bb463b619bff908b6881af06f42d9f4531":{"kind":"REWORK","_number":1,"created":"2013-06-12 17:54:22.000000000","uploader":{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},"ref":"refs/changes/59/32759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/59/32759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/59/32759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/59/32759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/59/32759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/59/32759/1"}}},"commit":{"parents":[{"commit":"aa9ca11919b2fe2f3ea9abe65ea605101668348a","subject":"Merge \"Tempest Coding Guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/aa9ca11919b2fe2f3ea9abe65ea605101668348a"}]}],"author":{"name":"Li Ma","email":"nick.ma@live.cn","date":"2013-06-12 17:39:24.000000000","tz":-420},"committer":{"name":"Li Ma","email":"nick.ma@live.cn","date":"2013-06-12 17:39:24.000000000","tz":-420},"subject":"Remove basic_auth strategy","message":"Remove basic_auth strategy\n\nThe auth strategy selection option is removed from both code\nand conf file, since it is confusing and actually other\nauthentication methods are not supported.\n\nChange-Id: I498be0b1e3eeea397e1f164caffe7cc88be0ceda\nImplements: remove basic_auth strategy\nFixes: bug #1180972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/276731bb463b619bff908b6881af06f42d9f4531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/276731bb463b619bff908b6881af06f42d9f4531"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aa9ca11919b2fe2f3ea9abe65ea605101668348a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"484d2e44ebf45987ac1a1210017cd9fd3f53b80a":{"kind":"REWORK","_number":2,"created":"2013-06-12 18:30:08.000000000","uploader":{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},"ref":"refs/changes/59/32759/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/59/32759/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/59/32759/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/59/32759/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/59/32759/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/59/32759/2"}}},"commit":{"parents":[{"commit":"aa9ca11919b2fe2f3ea9abe65ea605101668348a","subject":"Merge \"Tempest Coding Guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/aa9ca11919b2fe2f3ea9abe65ea605101668348a"}]}],"author":{"name":"Li Ma","email":"nick.ma@live.cn","date":"2013-06-12 18:26:08.000000000","tz":-420},"committer":{"name":"Li Ma","email":"nick.ma@live.cn","date":"2013-06-12 18:26:08.000000000","tz":-420},"subject":"Re-commit the files, because the previous update is not correct.","message":"Re-commit the files, because the previous update is not correct.\n\nChange-Id: I498be0b1e3eeea397e1f164caffe7cc88be0ceda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/484d2e44ebf45987ac1a1210017cd9fd3f53b80a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/484d2e44ebf45987ac1a1210017cd9fd3f53b80a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aa9ca11919b2fe2f3ea9abe65ea605101668348a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a0b2ded7c1472baddf560531cc80eefe6d6d901d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2013-06-12 18:38:12.000000000","uploader":{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},"ref":"refs/changes/59/32759/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/59/32759/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/59/32759/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/59/32759/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/59/32759/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/59/32759/3"}}},"commit":{"parents":[{"commit":"aa9ca11919b2fe2f3ea9abe65ea605101668348a","subject":"Merge \"Tempest Coding Guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/aa9ca11919b2fe2f3ea9abe65ea605101668348a"}]}],"author":{"name":"Li Ma","email":"nick.ma@live.cn","date":"2013-06-12 18:26:08.000000000","tz":-420},"committer":{"name":"Li Ma","email":"nick.ma@live.cn","date":"2013-06-12 18:34:22.000000000","tz":-420},"subject":"Remove basic_auth strategy","message":"Remove basic_auth strategy\n\nThe auth strategy selection is removed since it is confusing and other authentication methods are not supported.\n\nRe-commit the files, because the previous update is not correct.\n\nImplements: remove basic_auth strategy\nFixes: bug #1180972\nChange-Id: I498be0b1e3eeea397e1f164caffe7cc88be0ceda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a0b2ded7c1472baddf560531cc80eefe6d6d901d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a0b2ded7c1472baddf560531cc80eefe6d6d901d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aa9ca11919b2fe2f3ea9abe65ea605101668348a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"216550f81ac7c2d749946dbf466170febcadfe76":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2013-06-12 18:40:59.000000000","uploader":{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},"ref":"refs/changes/59/32759/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/59/32759/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/59/32759/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/59/32759/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/59/32759/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/59/32759/4"}}},"commit":{"parents":[{"commit":"aa9ca11919b2fe2f3ea9abe65ea605101668348a","subject":"Merge \"Tempest Coding Guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/aa9ca11919b2fe2f3ea9abe65ea605101668348a"}]}],"author":{"name":"Li Ma","email":"nick.ma@live.cn","date":"2013-06-12 18:26:08.000000000","tz":-420},"committer":{"name":"Li Ma","email":"nick.ma@live.cn","date":"2013-06-12 18:40:00.000000000","tz":-420},"subject":"Remove basic_auth strategy","message":"Remove basic_auth strategy\n\nThe auth strategy selection is removed\nsince it is confusing and other authentication\nmethods are not supported.\n\nRe-commit the files, because the previous update is not correct.\n\nImplements: remove basic_auth strategy\nFixes: bug #1180972\nChange-Id: I498be0b1e3eeea397e1f164caffe7cc88be0ceda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/216550f81ac7c2d749946dbf466170febcadfe76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/216550f81ac7c2d749946dbf466170febcadfe76"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aa9ca11919b2fe2f3ea9abe65ea605101668348a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
