)]}'
{"id":"openstack%2Ftempest~333074","triplet_id":"openstack%2Ftempest~master~Ie7814bd50f1c57c5f8da8355f665f09c9ac8e2d9","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"Ie7814bd50f1c57c5f8da8355f665f09c9ac8e2d9","subject":"Remove testscenarios usage from test_server_basic_ops","status":"MERGED","created":"2016-06-23 01:18:55.000000000","updated":"2016-06-25 22:06:20.000000000","submitted":"2016-06-23 20:34:28.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"76a07a82b7af417283e594f11f5aa7b2d1f27bef","_number":333074,"virtual_id_number":333074,"owner":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"date":"2016-06-25 22:06:20.000000000","post_submit":true,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-06-23 20:34:28.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2016-06-24 04:35:12.000000000","post_submit":true,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"all":[{"value":2,"date":"2016-06-23 09:42:45.000000000","_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":2,"date":"2016-06-23 18:51:38.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":2,"date":"2016-06-23 10:13:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":1,"date":"2016-06-23 18:51:38.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-23 01:21:47.000000000","updated_by":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"reviewer":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"state":"REVIEWER"},{"updated":"2016-06-23 03:50:06.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-23 09:42:45.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2016-06-23 10:13:48.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2016-06-23 18:51:38.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2016-06-23 20:34:28.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-06-24 04:35:12.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-25 22:06:20.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72fd35ef22c1435c9fc982c01b457d1dcdec7294","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2016-06-23 01:18:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89813c654c3ffabed7dfaddec3e52465162e7769","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2016-06-23 01:21:47.000000000","message":"Patch Set 1: Code-Review+1\n\nI tried removing the line in test_server_basic_ops.py in my environment and it caused test startup to go from minutes to milliseconds.","accounts_in_message":[],"_revision_number":1},{"id":"97467f8a4c0e8c3879a5c63b937faed1febae332","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2016-06-23 01:23:26.000000000","message":"Patch Set 1:\n\nNote that I\u0027m only running keystone so only running identity tests: compute, networking, etc., are all disabled in config; and I set the regex to \"tempest.api.identity\".","accounts_in_message":[],"_revision_number":1},{"id":"80c1f8e8eebcc8c2c3d615dca89659e7beaa033a","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2016-06-23 01:33:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"627639fbd20a7d5710b9ca6634630c9e76871a00","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-23 02:03:32.000000000","message":"Patch Set 2:\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/74/333074/2/20976/index.html : FAILURE in 28 m 17 s","accounts_in_message":[],"_revision_number":2},{"id":"d29bcae4eb4136d814b73fa1ef9ddc2706e2e813","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-23 02:18:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-full http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-full/1df39f2/ : FAILURE in 24m 02s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-postgres-full/7f7b2fb/ : FAILURE in 25m 18s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-neutron-full/3478078/ : FAILURE in 26m 14s\n- gate-grenade-dsvm http://logs.openstack.org/74/333074/2/check/gate-grenade-dsvm/c1ebb5c/ : FAILURE in 21m 47s\n- gate-tempest-python34 http://logs.openstack.org/74/333074/2/check/gate-tempest-python34/ffa5ef0/ : FAILURE in 3m 14s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/74/333074/2/check/gate-tempest-releasenotes/20856cd//releasenotes/build/html/ : SUCCESS in 3m 14s\n- gate-tempest-pep8 http://logs.openstack.org/74/333074/2/check/gate-tempest-pep8/631edbc/ : FAILURE in 3m 29s\n- gate-tempest-docs http://docs-draft.openstack.org/74/333074/2/check/gate-tempest-docs/e93746c//doc/build/html/ : SUCCESS in 2m 07s\n- gate-tempest-python27 http://logs.openstack.org/74/333074/2/check/gate-tempest-python27/b622ccc/ : FAILURE in 3m 20s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/74/333074/2/check/gate-grenade-dsvm-multinode/3204cb0/ : FAILURE in 28m 37s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/74/333074/2/check/gate-grenade-dsvm-neutron-multinode/3eb716e/ : FAILURE in 30m 02s\n- gate-grenade-dsvm-neutron http://logs.openstack.org/74/333074/2/check/gate-grenade-dsvm-neutron/9086614/ : FAILURE in 14m 25s\n- gate-tempest-dsvm-ironic-agent_ssh-nv http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-ironic-agent_ssh-nv/4e986ce/ : FAILURE in 33m 49s (non-voting)\n- gate-tempest-dsvm-neutron-full-ssh http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-neutron-full-ssh/e7c091f/ : FAILURE in 29m 08s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/55b6362/ : FAILURE in 28m 06s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph/3acc10f/ : FAILURE in 24m 04s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7-nv http://logs.openstack.org/74/333074/2/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7-nv/ec6ad5f/ : FAILURE in 21m 11s (non-voting)\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7-nv http://logs.openstack.org/74/333074/2/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7-nv/0b8245b/ : FAILURE in 25m 18s (non-voting)\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7-nv http://logs.openstack.org/74/333074/2/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7-nv/e51d9d2/ : FAILURE in 24m 30s (non-voting)\n- gate-tempest-dsvm-layer4 http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-layer4/d574a44/ : FAILURE in 38m 02s\n- gate-tempest-dsvm-neutron-dvr http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-neutron-dvr/e164aae/ : FAILURE in 27m 57s\n- gate-tempest-dsvm-full-mitaka http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-full-mitaka/f050af5/ : FAILURE in 26m 48s\n- gate-tempest-dsvm-postgres-full-mitaka http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-postgres-full-mitaka/efe2820/ : FAILURE in 26m 05s\n- gate-tempest-dsvm-neutron-full-mitaka http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-neutron-full-mitaka/4a990ea/ : FAILURE in 28m 22s\n- gate-tempest-dsvm-full-liberty http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-full-liberty/c2d6f68/ : FAILURE in 11m 56s\n- gate-tempest-dsvm-postgres-full-liberty http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-postgres-full-liberty/d5593f3/ : FAILURE in 11m 47s\n- gate-tempest-dsvm-neutron-full-liberty http://logs.openstack.org/74/333074/2/check/gate-tempest-dsvm-neutron-full-liberty/e46d401/ : FAILURE in 16m 27s","accounts_in_message":[],"_revision_number":2},{"id":"9e56b4f68ed080e4e9c7f3dbb5c17c7a81446777","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2016-06-23 02:27:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bc5dd118746b059e120ce4ef92f8028016c6efd1","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2016-06-23 02:39:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"08ba3587616300b2bbe4b00352cbef51648f497d","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2016-06-23 03:35:16.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"98ef95e6b1805188c29f12353075807920fddaf8","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-23 03:50:06.000000000","message":"Patch Set 4:\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/74/333074/4/20982/index.html : SUCCESS in 01 h 08 m 35 s","accounts_in_message":[],"_revision_number":4},{"id":"b42cb001fe9502ca7b691050590dde867e05439f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-23 03:59:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-full http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-full/007189c/ : SUCCESS in 44m 31s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-postgres-full/f46e0d1/ : SUCCESS in 45m 39s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-neutron-full/6e85f1d/ : SUCCESS in 55m 43s\n- gate-grenade-dsvm http://logs.openstack.org/74/333074/4/check/gate-grenade-dsvm/9f21b22/ : SUCCESS in 25m 40s\n- gate-tempest-python34 http://logs.openstack.org/74/333074/4/check/gate-tempest-python34/e14347c/ : SUCCESS in 5m 51s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/74/333074/4/check/gate-tempest-releasenotes/78acd3f//releasenotes/build/html/ : SUCCESS in 2m 59s\n- gate-tempest-pep8 http://logs.openstack.org/74/333074/4/check/gate-tempest-pep8/fe34b8a/ : SUCCESS in 2m 59s\n- gate-tempest-docs http://docs-draft.openstack.org/74/333074/4/check/gate-tempest-docs/0634fef//doc/build/html/ : SUCCESS in 3m 05s\n- gate-tempest-python27 http://logs.openstack.org/74/333074/4/check/gate-tempest-python27/52b775c/ : SUCCESS in 2m 28s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/74/333074/4/check/gate-grenade-dsvm-multinode/59deee1/ : SUCCESS in 46m 30s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/74/333074/4/check/gate-grenade-dsvm-neutron-multinode/e0f358b/ : SUCCESS in 54m 23s\n- gate-grenade-dsvm-neutron http://logs.openstack.org/74/333074/4/check/gate-grenade-dsvm-neutron/45a5fe3/ : SUCCESS in 48m 59s\n- gate-tempest-dsvm-ironic-agent_ssh-nv http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-ironic-agent_ssh-nv/e75792e/ : SUCCESS in 33m 56s (non-voting)\n- gate-tempest-dsvm-neutron-full-ssh http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-neutron-full-ssh/a3e1ed0/ : FAILURE in 1h 17m 09s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/b80a8ac/ : SUCCESS in 59m 23s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph/2edbf0d/ : SUCCESS in 49m 55s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7-nv http://logs.openstack.org/74/333074/4/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7-nv/3ca2126/ : SUCCESS in 37m 49s (non-voting)\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7-nv http://logs.openstack.org/74/333074/4/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7-nv/f5132a5/ : SUCCESS in 36m 50s (non-voting)\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7-nv http://logs.openstack.org/74/333074/4/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7-nv/104b986/ : SUCCESS in 31m 56s (non-voting)\n- gate-tempest-dsvm-layer4 http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-layer4/461f39b/ : SUCCESS in 33m 58s\n- gate-tempest-dsvm-neutron-dvr http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-neutron-dvr/85f2a09/ : SUCCESS in 56m 54s\n- gate-tempest-dsvm-full-mitaka http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-full-mitaka/37040a6/ : SUCCESS in 46m 21s\n- gate-tempest-dsvm-postgres-full-mitaka http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-postgres-full-mitaka/3685ab1/ : SUCCESS in 35m 24s\n- gate-tempest-dsvm-neutron-full-mitaka http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-neutron-full-mitaka/28b1ead/ : SUCCESS in 1h 14m 01s\n- gate-tempest-dsvm-full-liberty http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-full-liberty/88affce/ : SUCCESS in 53m 05s\n- gate-tempest-dsvm-postgres-full-liberty http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-postgres-full-liberty/495dac5/ : SUCCESS in 51m 07s\n- gate-tempest-dsvm-neutron-full-liberty http://logs.openstack.org/74/333074/4/check/gate-tempest-dsvm-neutron-full-liberty/501d46e/ : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":4},{"id":"beb20f7f19ca746926d1d5edf246b710738e0542","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2016-06-23 09:41:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f363685fe8c5581ebda0f77f1956ddcb10af7e1","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2016-06-23 09:41:48.000000000","message":"Patch Set 5: Published edit on patch set 4","accounts_in_message":[],"_revision_number":5},{"id":"c36cd2e9142946bbe85ba157fc97d6de2f5fae07","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2016-06-23 09:42:45.000000000","message":"Patch Set 5: Code-Review+2\n\nI agree, this was badly designed. Thanks for the cleanup.","accounts_in_message":[],"_revision_number":5},{"id":"23bbc6f67d3d1b4ea6dcfef0890b4e42977bb6b6","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2016-06-23 10:13:48.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"df9b756dcc2760def0481bcbfe1410b930c29359","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2016-06-23 10:16:18.000000000","message":"Patch Set 5:\n\nLGTM, but I hesitate to approve without non-HPE guys +2.","accounts_in_message":[],"_revision_number":5},{"id":"e86b9b31d2c34b951d438657c39ef614011a8ec9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-23 10:58:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-full http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-full/cb42769/ : SUCCESS in 37m 50s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-postgres-full/8cb4fc4/ : SUCCESS in 50m 29s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-neutron-full/03f5ced/ : SUCCESS in 1h 00m 24s\n- gate-grenade-dsvm http://logs.openstack.org/74/333074/5/check/gate-grenade-dsvm/e571e91/ : SUCCESS in 35m 38s\n- gate-tempest-python34 http://logs.openstack.org/74/333074/5/check/gate-tempest-python34/c91e9ac/ : SUCCESS in 5m 47s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/74/333074/5/check/gate-tempest-releasenotes/5c4894a//releasenotes/build/html/ : SUCCESS in 2m 57s\n- gate-tempest-pep8 http://logs.openstack.org/74/333074/5/check/gate-tempest-pep8/a571c5c/ : SUCCESS in 3m 13s\n- gate-tempest-docs http://docs-draft.openstack.org/74/333074/5/check/gate-tempest-docs/7c5648f//doc/build/html/ : SUCCESS in 4m 36s\n- gate-tempest-python27 http://logs.openstack.org/74/333074/5/check/gate-tempest-python27/12cc362/ : SUCCESS in 5m 15s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/74/333074/5/check/gate-grenade-dsvm-multinode/aa433f4/ : SUCCESS in 46m 20s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/74/333074/5/check/gate-grenade-dsvm-neutron-multinode/734013d/ : SUCCESS in 36m 44s\n- gate-grenade-dsvm-neutron http://logs.openstack.org/74/333074/5/check/gate-grenade-dsvm-neutron/9680594/ : SUCCESS in 52m 07s\n- gate-tempest-dsvm-ironic-agent_ssh-nv http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-ironic-agent_ssh-nv/cb84b89/ : SUCCESS in 35m 37s (non-voting)\n- gate-tempest-dsvm-neutron-full-ssh http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-neutron-full-ssh/d973889/ : FAILURE in 1h 12m 20s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/bcc5722/ : SUCCESS in 57m 40s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph/8851eb8/ : SUCCESS in 46m 30s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7-nv http://logs.openstack.org/74/333074/5/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7-nv/d95320d/ : FAILURE in 34m 47s (non-voting)\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7-nv http://logs.openstack.org/74/333074/5/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7-nv/d7df988/ : SUCCESS in 36m 51s (non-voting)\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7-nv http://logs.openstack.org/74/333074/5/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7-nv/61121cb/ : SUCCESS in 23m 13s (non-voting)\n- gate-tempest-dsvm-layer4 http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-layer4/4a84e0d/ : SUCCESS in 38m 38s\n- gate-tempest-dsvm-neutron-dvr http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-neutron-dvr/922ff78/ : SUCCESS in 1h 14m 59s\n- gate-tempest-dsvm-full-mitaka http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-full-mitaka/fa8bdcd/ : FAILURE in 45m 15s\n- gate-tempest-dsvm-postgres-full-mitaka http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-postgres-full-mitaka/8b2b648/ : SUCCESS in 48m 37s\n- gate-tempest-dsvm-neutron-full-mitaka http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-neutron-full-mitaka/f4a4feb/ : SUCCESS in 1h 04m 30s\n- gate-tempest-dsvm-full-liberty http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-full-liberty/cdac8dd/ : SUCCESS in 43m 01s\n- gate-tempest-dsvm-postgres-full-liberty http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-postgres-full-liberty/79676ed/ : SUCCESS in 41m 24s\n- gate-tempest-dsvm-neutron-full-liberty http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-neutron-full-liberty/5c954c7/ : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":5},{"id":"b9924966727cec91bfe293fd505005582f918ec5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-23 10:59:59.000000000","message":"Patch Set 5:\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/74/333074/5/21031/index.html : SUCCESS in 01 h 15 m 20 s","accounts_in_message":[],"_revision_number":5},{"id":"3c9e82f1172d49158795faf88701199dbcf9cee6","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2016-06-23 14:05:59.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a2fb7b4a6ce220da1ba1d6449c35c31fbeec2fd1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-23 16:01:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-full http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-full/ef3feba/ : SUCCESS in 39m 31s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-postgres-full/dfff2c3/ : SUCCESS in 51m 25s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-neutron-full/228e558/ : SUCCESS in 1h 12m 54s\n- gate-grenade-dsvm http://logs.openstack.org/74/333074/5/check/gate-grenade-dsvm/91440e3/ : SUCCESS in 51m 20s\n- gate-tempest-python34 http://logs.openstack.org/74/333074/5/check/gate-tempest-python34/5882b03/ : SUCCESS in 6m 31s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/74/333074/5/check/gate-tempest-releasenotes/32a4357//releasenotes/build/html/ : SUCCESS in 4m 44s\n- gate-tempest-pep8 http://logs.openstack.org/74/333074/5/check/gate-tempest-pep8/702abae/ : SUCCESS in 2m 43s\n- gate-tempest-docs http://docs-draft.openstack.org/74/333074/5/check/gate-tempest-docs/347b518//doc/build/html/ : SUCCESS in 2m 43s\n- gate-tempest-python27 http://logs.openstack.org/74/333074/5/check/gate-tempest-python27/f6e2148/ : SUCCESS in 3m 07s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/74/333074/5/check/gate-grenade-dsvm-multinode/984b93d/ : SUCCESS in 44m 22s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/74/333074/5/check/gate-grenade-dsvm-neutron-multinode/02aa32e/ : SUCCESS in 52m 16s\n- gate-grenade-dsvm-neutron http://logs.openstack.org/74/333074/5/check/gate-grenade-dsvm-neutron/54c01ab/ : SUCCESS in 33m 42s\n- gate-tempest-dsvm-ironic-agent_ssh-nv http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-ironic-agent_ssh-nv/6410f41/ : SUCCESS in 35m 59s (non-voting)\n- gate-tempest-dsvm-neutron-full-ssh http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-neutron-full-ssh/e9951e5/ : FAILURE in 1h 07m 39s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/4022e3d/ : SUCCESS in 1h 14m 04s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph/fddc055/ : SUCCESS in 47m 49s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7-nv http://logs.openstack.org/74/333074/5/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7-nv/1df4885/ : FAILURE in 36m 01s (non-voting)\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7-nv http://logs.openstack.org/74/333074/5/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7-nv/7a18608/ : SUCCESS in 41m 07s (non-voting)\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7-nv http://logs.openstack.org/74/333074/5/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7-nv/26546e8/ : FAILURE in 35m 20s (non-voting)\n- gate-tempest-dsvm-layer4 http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-layer4/8f857fe/ : SUCCESS in 33m 19s\n- gate-tempest-dsvm-neutron-dvr http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-neutron-dvr/75c76e7/ : SUCCESS in 1h 02m 20s\n- gate-tempest-dsvm-full-mitaka http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-full-mitaka/6e811ea/ : SUCCESS in 50m 41s\n- gate-tempest-dsvm-postgres-full-mitaka http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-postgres-full-mitaka/bb0973f/ : SUCCESS in 40m 20s\n- gate-tempest-dsvm-neutron-full-mitaka http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-neutron-full-mitaka/872aa0f/ : SUCCESS in 1h 03m 28s\n- gate-tempest-dsvm-full-liberty http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-full-liberty/a035517/ : SUCCESS in 41m 24s\n- gate-tempest-dsvm-postgres-full-liberty http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-postgres-full-liberty/a708242/ : SUCCESS in 35m 13s\n- gate-tempest-dsvm-neutron-full-liberty http://logs.openstack.org/74/333074/5/check/gate-tempest-dsvm-neutron-full-liberty/8bec6af/ : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":5},{"id":"67f09e361f6ba582bd190df8d96a59481389e085","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2016-06-23 18:51:38.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThis part was implemented on the commit Ia86f48772ac02d67faa1c7d764cb9dc0938f6452 and the author agrees with its removal now.\nIn addition, the resource qa-spec(https://review.openstack.org/#/c/173334/) will be instead in the future.\n\nSo LGTM","accounts_in_message":[],"_revision_number":5},{"id":"1130d414fa0847b71abbc779bd20c19f824dd3c2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-23 18:51:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":5},{"id":"75f152cad6e7f1f85b20c1650a5ef13495fde2f8","date":"2016-06-23 20:34:28.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":5},{"id":"d7805062a0b136797603184b370e911348de6561","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-23 20:34:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tempest-dsvm-full http://logs.openstack.org/74/333074/5/gate/gate-tempest-dsvm-full/2ee2b19/ : SUCCESS in 38m 37s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/74/333074/5/gate/gate-tempest-dsvm-postgres-full/1e6bf8c/ : SUCCESS in 46m 26s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/74/333074/5/gate/gate-tempest-dsvm-neutron-full/ee53b60/ : SUCCESS in 1h 02m 34s\n- gate-grenade-dsvm http://logs.openstack.org/74/333074/5/gate/gate-grenade-dsvm/66ea19d/ : SUCCESS in 45m 03s\n- gate-tempest-python34 http://logs.openstack.org/74/333074/5/gate/gate-tempest-python34/2cc4536/ : SUCCESS in 3m 44s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/74/333074/5/gate/gate-tempest-releasenotes/167ca90//releasenotes/build/html/ : SUCCESS in 3m 26s\n- gate-tempest-pep8 http://logs.openstack.org/74/333074/5/gate/gate-tempest-pep8/06752bb/ : SUCCESS in 3m 07s\n- gate-tempest-docs http://docs-draft.openstack.org/74/333074/5/gate/gate-tempest-docs/d40cd7a//doc/build/html/ : SUCCESS in 4m 22s\n- gate-tempest-python27 http://logs.openstack.org/74/333074/5/gate/gate-tempest-python27/0c7068e/ : SUCCESS in 4m 49s\n- gate-tempest-dsvm-layer4 http://logs.openstack.org/74/333074/5/gate/gate-tempest-dsvm-layer4/a2996d8/ : SUCCESS in 38m 46s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/74/333074/5/gate/gate-grenade-dsvm-multinode/c0932b7/ : SUCCESS in 46m 01s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/74/333074/5/gate/gate-grenade-dsvm-neutron-multinode/39580ac/ : SUCCESS in 50m 12s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/74/333074/5/gate/gate-tempest-dsvm-full-devstack-plugin-ceph/a7b28b5/ : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":5},{"id":"bcfa5db63e5ff2b5dd20bfa608b8aeb81d375aaf","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-06-24 04:35:12.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Tempest-Sriov http://13.69.151.247/74/333074/5/check-sriov-tempest/Tempest-Sriov/3bcb9d5 : SUCCESS in 32m 36s (non-voting)\n\nTo re-run the job post \u0027recheck tempest-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"76a07a82b7af417283e594f11f5aa7b2d1f27bef","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2016-06-25 22:06:20.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- EMC_VNX_FC http://publiclogs.emc.com/vnx_ostack/EMC_VNX_FC/727 : SUCCESS in 1h 24m 33s\n- EMC_VNX_iSCSI http://publiclogs.emc.com/vnx_ostack/EMC_VNX_iSCSI/705 : SUCCESS in 1h 19m 59s\n\nLeave a comment with \u0027run emc-vnx\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-vnx-ci","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f2c45014bbc16bcc868921d3194a8771aae4e536","revisions":{"1a40b15d6ef38062d4c3a263c06c28e164d0b011":{"kind":"REWORK","_number":1,"created":"2016-06-23 01:18:55.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/74/333074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/74/333074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/74/333074/1"}}},"commit":{"parents":[{"commit":"23d05753380986dec7f890c38ed47ea3cc99d835","subject":"Merge \"Fix un-assignment local variable \u0027returncode\u0027 error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/23d05753380986dec7f890c38ed47ea3cc99d835"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2016-06-23 01:13:42.000000000","tz":-240},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2016-06-23 01:14:20.000000000","tz":-240},"subject":"Remove testscenarios usage from test_server_basic_ops","message":"Remove testscenarios usage from test_server_basic_ops\n\nThis commit removes the scenario test wrapper around\ntest_server_basic_ops. This has always caused problems in tempest as\nit actually tries to do real operations during discovery which is a\nterrible idea. This weighed against a very limited benefit and limited\nuse by most people it\u0027s time to just call the experiment and remove\nit. This is not saying there is no usefulness in testing a full matrix\nof images and flavors on a real deployment, but instead that the place\nfor this kind of testing is not in tempest discovery. If people who\nfound this valuable in the past want to retain the functionality should\nhelp work on an alternative method of doing it that isn\u0027t so invasive.\n\nChange-Id: Ie7814bd50f1c57c5f8da8355f665f09c9ac8e2d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1a40b15d6ef38062d4c3a263c06c28e164d0b011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1a40b15d6ef38062d4c3a263c06c28e164d0b011"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"23d05753380986dec7f890c38ed47ea3cc99d835","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"68cf398a9dac8e04b1746f2d71514e9d76b94ff9":{"kind":"REWORK","_number":2,"created":"2016-06-23 01:33:59.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/74/333074/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/74/333074/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/74/333074/2"}}},"commit":{"parents":[{"commit":"23d05753380986dec7f890c38ed47ea3cc99d835","subject":"Merge \"Fix un-assignment local variable \u0027returncode\u0027 error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/23d05753380986dec7f890c38ed47ea3cc99d835"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2016-06-23 01:13:42.000000000","tz":-240},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2016-06-23 01:33:47.000000000","tz":-240},"subject":"Remove testscenarios usage from test_server_basic_ops","message":"Remove testscenarios usage from test_server_basic_ops\n\nThis commit removes the scenario test wrapper around\ntest_server_basic_ops. This has always caused problems in tempest as\nit actually tries to do real operations during discovery which is a\nterrible idea. This weighed against a very limited benefit and limited\nuse by most people it\u0027s time to just call the experiment and remove\nit. This is not saying there is no usefulness in testing a full matrix\nof images and flavors on a real deployment, but instead that the place\nfor this kind of testing is not in tempest discovery. If people who\nfound this valuable in the past want to retain the functionality should\nhelp work on an alternative method of doing it that isn\u0027t so invasive.\n\nChange-Id: Ie7814bd50f1c57c5f8da8355f665f09c9ac8e2d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/68cf398a9dac8e04b1746f2d71514e9d76b94ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/68cf398a9dac8e04b1746f2d71514e9d76b94ff9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"23d05753380986dec7f890c38ed47ea3cc99d835","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8af738618dd0d277a54b6ec0384190867092e6bf":{"kind":"REWORK","_number":3,"created":"2016-06-23 02:27:19.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/74/333074/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/74/333074/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/74/333074/3"}}},"commit":{"parents":[{"commit":"23d05753380986dec7f890c38ed47ea3cc99d835","subject":"Merge \"Fix un-assignment local variable \u0027returncode\u0027 error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/23d05753380986dec7f890c38ed47ea3cc99d835"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2016-06-23 01:13:42.000000000","tz":-240},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2016-06-23 02:27:08.000000000","tz":-240},"subject":"Remove testscenarios usage from test_server_basic_ops","message":"Remove testscenarios usage from test_server_basic_ops\n\nThis commit removes the scenario test wrapper around\ntest_server_basic_ops. This has always caused problems in tempest as\nit actually tries to do real operations during discovery which is a\nterrible idea. This weighed against a very limited benefit and limited\nuse by most people it\u0027s time to just call the experiment and remove\nit. This is not saying there is no usefulness in testing a full matrix\nof images and flavors on a real deployment, but instead that the place\nfor this kind of testing is not in tempest discovery. If people who\nfound this valuable in the past want to retain the functionality should\nhelp work on an alternative method of doing it that isn\u0027t so invasive.\n\nChange-Id: Ie7814bd50f1c57c5f8da8355f665f09c9ac8e2d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8af738618dd0d277a54b6ec0384190867092e6bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8af738618dd0d277a54b6ec0384190867092e6bf"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"23d05753380986dec7f890c38ed47ea3cc99d835","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d63d82c042751e5ab7130a4cce3352771f37d274":{"kind":"REWORK","_number":4,"created":"2016-06-23 02:39:42.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/74/333074/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/74/333074/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/74/333074/4"}}},"commit":{"parents":[{"commit":"23d05753380986dec7f890c38ed47ea3cc99d835","subject":"Merge \"Fix un-assignment local variable \u0027returncode\u0027 error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/23d05753380986dec7f890c38ed47ea3cc99d835"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2016-06-23 01:13:42.000000000","tz":-240},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2016-06-23 02:39:29.000000000","tz":-240},"subject":"Remove testscenarios usage from test_server_basic_ops","message":"Remove testscenarios usage from test_server_basic_ops\n\nThis commit removes the scenario test wrapper around\ntest_server_basic_ops. This has always caused problems in tempest as\nit actually tries to do real operations during discovery which is a\nterrible idea. This weighed against a very limited benefit and limited\nuse by most people it\u0027s time to just call the experiment and remove\nit. This is not saying there is no usefulness in testing a full matrix\nof images and flavors on a real deployment, but instead that the place\nfor this kind of testing is not in tempest discovery. If people who\nfound this valuable in the past want to retain the functionality should\nhelp work on an alternative method of doing it that isn\u0027t so invasive.\n\nChange-Id: Ie7814bd50f1c57c5f8da8355f665f09c9ac8e2d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d63d82c042751e5ab7130a4cce3352771f37d274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d63d82c042751e5ab7130a4cce3352771f37d274"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"23d05753380986dec7f890c38ed47ea3cc99d835","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f2c45014bbc16bcc868921d3194a8771aae4e536":{"kind":"REWORK","_number":5,"created":"2016-06-23 09:41:48.000000000","uploader":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"ref":"refs/changes/74/333074/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/74/333074/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/74/333074/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/74/333074/5"}}},"commit":{"parents":[{"commit":"23d05753380986dec7f890c38ed47ea3cc99d835","subject":"Merge \"Fix un-assignment local variable \u0027returncode\u0027 error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/23d05753380986dec7f890c38ed47ea3cc99d835"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2016-06-23 01:13:42.000000000","tz":-240},"committer":{"name":"Andrea Frittoli","email":"andrea.frittoli@hpe.com","date":"2016-06-23 09:41:38.000000000","tz":0},"subject":"Remove testscenarios usage from test_server_basic_ops","message":"Remove testscenarios usage from test_server_basic_ops\n\nThis commit removes the scenario test wrapper around\ntest_server_basic_ops. This has always caused problems in tempest as\nit actually tries to do real operations during discovery which is a\nterrible idea. This weighed against a very limited benefit and limited\nuse by most people it\u0027s time to just call the experiment and remove\nit. This is not saying there is no usefulness in testing a full matrix\nof images and flavors on a real deployment, but instead that the place\nfor this kind of testing is not in tempest discovery. If people who\nfound this valuable in the past want to retain the functionality should\nhelp work on an alternative method of doing it that isn\u0027t so invasive.\n\nChange-Id: Ie7814bd50f1c57c5f8da8355f665f09c9ac8e2d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f2c45014bbc16bcc868921d3194a8771aae4e536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f2c45014bbc16bcc868921d3194a8771aae4e536"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"23d05753380986dec7f890c38ed47ea3cc99d835","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
