)]}'
{"id":"openstack%2Ftempest~34406","triplet_id":"openstack%2Ftempest~master~Ic54e5b4b7f7d95a0a6db568fb9a66a566a101fc9","project":"openstack/tempest","branch":"master","topic":"bug/1194468","hashtags":[],"change_id":"Ic54e5b4b7f7d95a0a6db568fb9a66a566a101fc9","subject":"Skip test that is not relevant to Quantum","status":"MERGED","created":"2013-06-25 14:16:19.000000000","updated":"2013-06-26 13:35:29.000000000","submitted":"2013-06-26 13:35:29.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"190bdd89b58236d3f4826a956d416ca8865e7c7f","_number":34406,"virtual_id_number":34406,"owner":{"_account_id":7781,"name":"Ala Rezmerita","email":"ala.rezmerita@orange.com","username":"alla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-06-26 13:35:29.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-06-26 06:32:20.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":2,"date":"2013-06-26 12:40:40.000000000","_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-06-26 06:32:20.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":1,"date":"2013-06-26 12:40:40.000000000","_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"date":"2013-06-25 14:26:47.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-06-25 14:26:47.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2013-06-26 06:32:20.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2013-06-26 12:40:40.000000000","updated_by":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"reviewer":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"state":"REVIEWER"},{"updated":"2013-06-26 13:35:29.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"67dd4bee28c2446d5bc7e2ec30ece5931b693506","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-25 15:06:44.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34406/1/check/gate-tempest-pep8/3879 : SUCCESS in 2m 15s\n- http://logs.openstack.org/34406/1/check/gate-tempest-docs/428 : SUCCESS in 1m 59s\n- http://logs.openstack.org/34406/1/check/gate-tempest-devstack-vm-full/26066 : SUCCESS in 41m 34s\n- http://logs.openstack.org/34406/1/check/gate-tempest-devstack-vm-postgres-full/25772 : SUCCESS in 41m 54s\n- http://logs.openstack.org/34406/1/check/gate-tempest-devstack-vm-quantum/33696 : SUCCESS in 17m 55s\n- http://logs.openstack.org/34406/1/check/gate-tempest-devstack-vm-quantum-full/2002 : FAILURE in 48m 00s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"b71601019b04a404d6d147616138db665bd0b663","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-06-26 06:32:20.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"f7e5258e86f3d9aa0139349b6cddd4b8bddd357b","author":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"date":"2013-06-26 12:40:40.000000000","message":"Patch Set 1: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":1},{"id":"0bed2877c125765fd0a6b04824d367c9fc9f9782","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-26 12:40:42.000000000","message":"Patch Set 1: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"03c5dbaad18a5762a26bf1944d5c841107b07f98","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-26 13:35:29.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1},{"id":"ce25d160fa16a3f4977b0bc682798550bff59ee2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-26 13:35:29.000000000","message":"Patch Set 1: Verified\n\nBuild succeeded.\n\n- http://logs.openstack.org/34406/1/gate/gate-tempest-pep8/3896 : SUCCESS in 1m 37s\n- http://logs.openstack.org/34406/1/gate/gate-tempest-docs/445 : SUCCESS in 2m 50s\n- http://logs.openstack.org/34406/1/gate/gate-tempest-devstack-vm-full/26407 : SUCCESS in 52m 17s\n- http://logs.openstack.org/34406/1/gate/gate-tempest-devstack-vm-postgres-full/26113 : SUCCESS in 53m 36s\n- http://logs.openstack.org/34406/1/gate/gate-tempest-devstack-vm-quantum/34049 : SUCCESS in 21m 37s\n- http://logs.openstack.org/34406/1/gate/gate-tempest-devstack-vm-quantum-full/2019 : FAILURE in 54m 44s (non-voting)\n","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30d6ae7a317a38a78c7db88e33834f5541175221","revisions":{"30d6ae7a317a38a78c7db88e33834f5541175221":{"kind":"REWORK","_number":1,"created":"2013-06-25 14:16:19.000000000","uploader":{"_account_id":7781,"name":"Ala Rezmerita","email":"ala.rezmerita@orange.com","username":"alla"},"ref":"refs/changes/06/34406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/06/34406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/06/34406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/06/34406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/06/34406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/06/34406/1"}}},"commit":{"parents":[{"commit":"74af42c714de3eec4657a9cfd7ae7d0570b73828","subject":"Oslo Merge","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/74af42c714de3eec4657a9cfd7ae7d0570b73828"}]}],"author":{"name":"Ala Rezmerita","email":"ala.rezmerita@cloudwatt.com","date":"2013-06-25 14:13:06.000000000","tz":120},"committer":{"name":"Ala Rezmerita","email":"ala.rezmerita@cloudwatt.com","date":"2013-06-25 14:13:06.000000000","tz":120},"subject":"Skip test that is not relevant to Quantum","message":"Skip test that is not relevant to Quantum\n\nQuantum allows duplicate names for security groups,\nthus this test must be skipped.\n\nbug #1194468\n\nChange-Id: Ic54e5b4b7f7d95a0a6db568fb9a66a566a101fc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/30d6ae7a317a38a78c7db88e33834f5541175221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/30d6ae7a317a38a78c7db88e33834f5541175221"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"74af42c714de3eec4657a9cfd7ae7d0570b73828","is_merged_in_target_branch":true,"change_id":"Iff6af9298cfa571b70729ea242a3b9dfc111a890","change_number":34148,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
