)]}'
{"id":"openstack%2Ftempest~362048","triplet_id":"openstack%2Ftempest~master~Id744e25331e4ff60f1367c603f50683961f75313","project":"openstack/tempest","branch":"master","topic":"backups_quota","hashtags":[],"change_id":"Id744e25331e4ff60f1367c603f50683961f75313","subject":"Adding update backups quota","status":"MERGED","created":"2016-08-29 11:54:32.000000000","updated":"2016-10-06 09:35:09.000000000","submitted":"2016-10-06 09:35:09.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ca7de077ed3d415fbdef6e2b6ccef257c0d15264","_number":362048,"virtual_id_number":362048,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":21187,"name":"Tzach Shefi","email":"tshefi@redhat.com","username":"tshefi"},{"value":2,"date":"2016-10-06 09:35:09.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"recommended":{"_account_id":21187,"name":"Tzach Shefi","email":"tshefi@redhat.com","username":"tshefi"},"all":[{"value":1,"date":"2016-08-30 07:58:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21187,"name":"Tzach Shefi","email":"tshefi@redhat.com","username":"tshefi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-10-06 08:38:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2016-08-31 13:14:58.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":0,"_account_id":21187,"name":"Tzach Shefi","email":"tshefi@redhat.com","username":"tshefi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-10-06 08:38:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21187,"name":"Tzach Shefi","email":"tshefi@redhat.com","username":"tshefi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":21187,"name":"Tzach Shefi","email":"tshefi@redhat.com","username":"tshefi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-08-30 07:58:44.000000000","updated_by":{"_account_id":21187,"name":"Tzach Shefi","email":"tshefi@redhat.com","username":"tshefi"},"reviewer":{"_account_id":21187,"name":"Tzach Shefi","email":"tshefi@redhat.com","username":"tshefi"},"state":"REVIEWER"},{"updated":"2016-08-31 13:14:58.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2016-10-06 08:38:51.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2016-10-06 09:35:09.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"f119fd94ae154f13a13dd715b65bed34b7ab3a3a","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2016-08-29 11:54:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3034e83e247966bd262c3b46fde2996f543c0ffb","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2016-08-29 12:35:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7b16a67b3b919897ddc21aa56456f0f14e64eb56","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-08-29 13:55:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/48/362048/2/check/dsvm-tempest-nova-network/d3910f4 : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":2},{"id":"46905e65a5ea263512ccfbc7b2279852d3b38a25","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-29 14:08:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/48/362048/2/check/gate-grenade-dsvm-neutron-ubuntu-trusty/678a8f2/ : SUCCESS in 44m 18s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/48/362048/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/05512cb/ : SUCCESS in 1h 00m 57s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/48/362048/2/check/gate-tempest-dsvm-full-ubuntu-xenial/b505b2f/ : SUCCESS in 47m 50s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/48/362048/2/check/gate-grenade-dsvm-ubuntu-trusty/4622e1c/ : SUCCESS in 35m 34s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/48/362048/2/check/gate-tempest-docs-ubuntu-xenial/59b4361//doc/build/html/ : SUCCESS in 4m 13s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/48/362048/2/check/gate-tempest-pep8-ubuntu-xenial/410caa5/ : SUCCESS in 2m 56s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/48/362048/2/check/gate-tempest-python27-ubuntu-xenial/2f02351/ : SUCCESS in 2m 44s\n- gate-tempest-python34 http://logs.openstack.org/48/362048/2/check/gate-tempest-python34/6cecb3f/ : SUCCESS in 2m 41s\n- gate-tempest-python35 http://logs.openstack.org/48/362048/2/check/gate-tempest-python35/e4fa42c/ : SUCCESS in 2m 47s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/48/362048/2/check/gate-grenade-dsvm-multinode/bb16a80/ : SUCCESS in 41m 11s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/48/362048/2/check/gate-grenade-dsvm-neutron-multinode/8ecbfc4/ : SUCCESS in 48m 19s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/48/362048/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/f3cef4a/ : SUCCESS in 54m 05s (non-voting)\n- gate-tempest-dsvm-neutron-full-ssh http://logs.openstack.org/48/362048/2/check/gate-tempest-dsvm-neutron-full-ssh/0d64aaf/ : FAILURE in 1h 27m 37s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/48/362048/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/d331728/ : FAILURE in 22m 41s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/48/362048/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph/db618dc/ : SUCCESS in 57m 25s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7-nv http://logs.openstack.org/48/362048/2/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7-nv/9671c3e/ : SUCCESS in 27m 26s (non-voting)\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7-nv http://logs.openstack.org/48/362048/2/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7-nv/19dcb89/ : SUCCESS in 41m 27s (non-voting)\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7-nv http://logs.openstack.org/48/362048/2/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7-nv/bc6b5f3/ : SUCCESS in 35m 59s (non-voting)\n- gate-tempest-dsvm-layer4 http://logs.openstack.org/48/362048/2/check/gate-tempest-dsvm-layer4/67f3e68/ : SUCCESS in 29m 32s\n- gate-tempest-dsvm-neutron-dvr http://logs.openstack.org/48/362048/2/check/gate-tempest-dsvm-neutron-dvr/b598851/ : SUCCESS in 1h 10m 18s\n- gate-tempest-dsvm-full-ubuntu-trusty-mitaka http://logs.openstack.org/48/362048/2/check/gate-tempest-dsvm-full-ubuntu-trusty-mitaka/b2aedd7/ : SUCCESS in 1h 02m 02s\n- gate-tempest-dsvm-neutron-full-ubuntu-trusty-mitaka http://logs.openstack.org/48/362048/2/check/gate-tempest-dsvm-neutron-full-ubuntu-trusty-mitaka/69a9be4/ : SUCCESS in 1h 07m 04s\n- gate-tempest-dsvm-full-ubuntu-trusty-liberty http://logs.openstack.org/48/362048/2/check/gate-tempest-dsvm-full-ubuntu-trusty-liberty/ee459e4/ : SUCCESS in 57m 49s\n- gate-tempest-dsvm-neutron-full-ubuntu-trusty-liberty http://logs.openstack.org/48/362048/2/check/gate-tempest-dsvm-neutron-full-ubuntu-trusty-liberty/ea057f9/ : SUCCESS in 1h 02m 40s","accounts_in_message":[],"_revision_number":2},{"id":"81ba3b6601436dceb1e14127cf070aca5c52d644","author":{"_account_id":21187,"name":"Tzach Shefi","email":"tshefi@redhat.com","username":"tshefi"},"date":"2016-08-30 07:58:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3bdc82a1ef8f9bd5ab2fac8838922c52d69cffa8","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2016-08-31 13:14:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"371503bf2b166069e601073f0b040c3ca7563729","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2016-10-06 08:38:51.000000000","message":"Patch Set 2: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"da4e0ee7c168158d4c299549c5942c0c58aa9186","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-06 08:39:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"9e469e3d6ae3353f873ee005f1784c7906fe0094","date":"2016-10-06 09:35:09.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"31602157e4f7d8006d1bf66e13f696887b207c28","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-06 09:35:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/48/362048/2/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/36f0f6b/ : SUCCESS in 46m 24s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/48/362048/2/gate/gate-grenade-dsvm-neutron-ubuntu-xenial/df4cc39/ : SUCCESS in 30m 12s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/48/362048/2/gate/gate-tempest-dsvm-full-ubuntu-xenial/dd63ea3/ : SUCCESS in 55m 50s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/48/362048/2/gate/gate-grenade-dsvm-ubuntu-xenial/bb27035/ : SUCCESS in 36m 31s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/48/362048/2/gate/gate-tempest-docs-ubuntu-xenial/30578d8//doc/build/html/ : SUCCESS in 3m 54s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/48/362048/2/gate/gate-tempest-pep8-ubuntu-xenial/7de964b/ : SUCCESS in 3m 41s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/48/362048/2/gate/gate-tempest-python27-ubuntu-xenial/364714d/ : SUCCESS in 4m 22s\n- gate-tempest-python34 http://logs.openstack.org/48/362048/2/gate/gate-tempest-python34/9e22d6c/ : SUCCESS in 2m 41s\n- gate-tempest-python35 http://logs.openstack.org/48/362048/2/gate/gate-tempest-python35/5a6f70b/ : SUCCESS in 3m 19s\n- gate-tempest-dsvm-layer4 http://logs.openstack.org/48/362048/2/gate/gate-tempest-dsvm-layer4/9986753/ : SUCCESS in 36m 40s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/48/362048/2/gate/gate-grenade-dsvm-multinode/727c77b/ : SUCCESS in 28m 52s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/48/362048/2/gate/gate-grenade-dsvm-neutron-multinode/0a31439/ : SUCCESS in 33m 45s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/48/362048/2/gate/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/761c287/ : SUCCESS in 44m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"070204bb1f5a07278adb9dbfffe8519a4fe756a7","revisions":{"8886056592b8830d44ce40c2de9173589cd77f55":{"kind":"REWORK","_number":1,"created":"2016-08-29 11:54:32.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/362048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/48/362048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/48/362048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/48/362048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/48/362048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/48/362048/1"}}},"commit":{"parents":[{"commit":"1e8a3a416f6d0a0d744c40c47bf3e1875f63efec","subject":"Remove unused config.CONF","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1e8a3a416f6d0a0d744c40c47bf3e1875f63efec"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2016-08-29 11:48:53.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2016-08-29 11:54:29.000000000","tz":180},"subject":"Adding update backups quota","message":"Adding update backups quota\n\nUpdate a testcase \"test_update_all_quota_resources_for_tenant\"\nto verify backups quota updating.\n\nChange-Id: Id744e25331e4ff60f1367c603f50683961f75313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8886056592b8830d44ce40c2de9173589cd77f55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8886056592b8830d44ce40c2de9173589cd77f55"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1e8a3a416f6d0a0d744c40c47bf3e1875f63efec","is_merged_in_target_branch":true,"change_id":"Ibb479c303ebeafdc8584403b0b334a0f47b6924f","change_number":361794,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"070204bb1f5a07278adb9dbfffe8519a4fe756a7":{"kind":"REWORK","_number":2,"created":"2016-08-29 12:35:45.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/362048/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/48/362048/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/48/362048/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/48/362048/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/48/362048/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/48/362048/2"}}},"commit":{"parents":[{"commit":"1e8a3a416f6d0a0d744c40c47bf3e1875f63efec","subject":"Remove unused config.CONF","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1e8a3a416f6d0a0d744c40c47bf3e1875f63efec"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2016-08-29 11:48:53.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2016-08-29 12:35:32.000000000","tz":180},"subject":"Adding update backups quota","message":"Adding update backups quota\n\nUpdate a testcase \"test_update_all_quota_resources_for_tenant\"\nto verify backups quota updating.\n\nChange-Id: Id744e25331e4ff60f1367c603f50683961f75313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/070204bb1f5a07278adb9dbfffe8519a4fe756a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/070204bb1f5a07278adb9dbfffe8519a4fe756a7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1e8a3a416f6d0a0d744c40c47bf3e1875f63efec","is_merged_in_target_branch":true,"change_id":"Ibb479c303ebeafdc8584403b0b334a0f47b6924f","change_number":361794,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
