)]}'
{"id":"openstack%2Ftempest~383399","triplet_id":"openstack%2Ftempest~master~I428d583eb3f3bbced321348e2a71035fbb282806","project":"openstack/tempest","branch":"master","topic":"releasenote-translation","hashtags":[],"change_id":"I428d583eb3f3bbced321348e2a71035fbb282806","subject":"Enable release notes translation","status":"MERGED","created":"2016-10-06 18:55:02.000000000","updated":"2016-10-13 20:35:58.000000000","submitted":"2016-10-13 20:35:58.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aaaa0aab7b2fb71f1687eb7cf00ece4ce360f10c","_number":383399,"virtual_id_number":383399,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-10-13 20:35:58.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-10-13 19:26:16.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":2,"date":"2016-10-07 02:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-10-13 19:26:16.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-10-07 02:34:49.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2016-10-13 19:26:16.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2016-10-13 20:35:58.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"fef8669bd194f2e0e67c9e5dd4ee9217af9f45f0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2016-10-06 18:55:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9291ca2af1a5ea86755a9824e71ed2759a642070","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-06 20:22:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/99/383399/1/check/gate-tempest-docs-ubuntu-xenial/c6aa4ea//doc/build/html/ : SUCCESS in 3m 53s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/99/383399/1/check/gate-tempest-pep8-ubuntu-xenial/aeea6d0/ : SUCCESS in 2m 36s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/99/383399/1/check/gate-tempest-python27-ubuntu-xenial/bbce9a1/ : SUCCESS in 3m 22s\n- gate-tempest-python34 http://logs.openstack.org/99/383399/1/check/gate-tempest-python34/2a66115/ : SUCCESS in 2m 31s\n- gate-tempest-python35 http://logs.openstack.org/99/383399/1/check/gate-tempest-python35/cff46f3/ : SUCCESS in 3m 19s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/99/383399/1/check/gate-tempest-releasenotes/0b79f8f//releasenotes/build/html/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e2eb2e2cf2c055750a4f0a8d874de65f4c91f7d9","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2016-10-07 02:34:49.000000000","message":"Patch Set 1: Code-Review+2\n\nMaybe, I should be the first translator for Japanese version, but I don\u0027t have the plan at this moment :-p","accounts_in_message":[],"_revision_number":1},{"id":"6c095f8352e4060246b77228f36deacba2e351c8","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2016-10-13 19:26:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bdb09154e66ed71fd1fef13f0597afc60e995f46","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-13 19:26:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"393e7c57720a343d4c66908c2e48ed1e0e54ac7f","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2016-10-13 19:26:57.000000000","message":"Patch Set 1:\n\nI will not stop Masayuki translate them into Japanase ;)","accounts_in_message":[],"_revision_number":1},{"id":"1b941b30c3fccd0d2fa69136c1e4d5df8426f0b8","date":"2016-10-13 20:35:58.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"5c376b76fb941ef138788045429807cae7f0e176","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-13 20:35:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/99/383399/1/gate/gate-tempest-docs-ubuntu-xenial/6b819c7//doc/build/html/ : SUCCESS in 3m 28s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/99/383399/1/gate/gate-tempest-pep8-ubuntu-xenial/bafeea0/ : SUCCESS in 2m 51s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/99/383399/1/gate/gate-tempest-python27-ubuntu-xenial/0d430a4/ : SUCCESS in 2m 46s\n- gate-tempest-python34 http://logs.openstack.org/99/383399/1/gate/gate-tempest-python34/3b43f51/ : SUCCESS in 3m 32s\n- gate-tempest-python35 http://logs.openstack.org/99/383399/1/gate/gate-tempest-python35/a7820a4/ : SUCCESS in 2m 52s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/99/383399/1/gate/gate-tempest-releasenotes/45df895//releasenotes/build/html/ : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5043f86287f856524c4d1e8a0c4d227d61ecac96","revisions":{"5043f86287f856524c4d1e8a0c4d227d61ecac96":{"kind":"REWORK","_number":1,"created":"2016-10-06 18:55:02.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/99/383399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/99/383399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/99/383399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/99/383399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/99/383399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/99/383399/1"}}},"commit":{"parents":[{"commit":"0235f98f143df20dbd17b658f5f9c35f02099411","subject":"Merge \"Adding a new option CONF.network.dvr_extra_resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0235f98f143df20dbd17b658f5f9c35f02099411"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2016-10-06 18:54:54.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2016-10-06 18:54:54.000000000","tz":120},"subject":"Enable release notes translation","message":"Enable release notes translation\n\nReleasenote translation publishing is being prepared. \u0027locale_dirs\u0027\nneeds to be defined in conf.py to generate translated version of the\nrelease notes.\n\nNote that this repository might not get translated release notes - or\nno translations at all - but we add the entry here nevertheless to\nprepare for it.\n\nChange-Id: I428d583eb3f3bbced321348e2a71035fbb282806\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5043f86287f856524c4d1e8a0c4d227d61ecac96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5043f86287f856524c4d1e8a0c4d227d61ecac96"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0235f98f143df20dbd17b658f5f9c35f02099411","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
