)]}'
{"id":"openstack%2Ftempest~39071","triplet_id":"openstack%2Ftempest~master~I07f2d94bd5652aed9f97241a3272b11a8a75785d","project":"openstack/tempest","branch":"master","topic":"sync-temp-log","hashtags":[],"change_id":"I07f2d94bd5652aed9f97241a3272b11a8a75785d","subject":"Use tempest log with tempest/openstack/common/lockutils.py","status":"ABANDONED","created":"2013-07-29 13:27:06.000000000","updated":"2013-07-30 18:22:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8224354ad99a02608fca7ac5db63304b8f3a903d","_number":39071,"virtual_id_number":39071,"owner":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2013-07-29 14:18:25.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-2,"date":"2013-07-30 18:08:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2013-07-30 18:08:55.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-07-29 14:18:25.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2013-07-30 18:08:55.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"}],"messages":[{"id":"bf608e5e78488909e0b38e951594f5609447d0b6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-29 14:18:25.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/71/39071/1/check/gate-tempest-pep8/4477 : SUCCESS in 3m 52s\n- http://logs.openstack.org/71/39071/1/check/gate-tempest-docs/1023 : SUCCESS in 4m 23s\n- http://logs.openstack.org/71/39071/1/check/gate-tempest-devstack-vm-full/36950 : SUCCESS in 50m 34s\n- http://logs.openstack.org/71/39071/1/check/gate-tempest-devstack-vm-cells-full/3705 : FAILURE in 18m 08s (non-voting)\n- http://logs.openstack.org/71/39071/1/check/gate-tempest-devstack-vm-testr-full/2525 : SUCCESS in 30m 53s (non-voting)\n- http://logs.openstack.org/71/39071/1/check/gate-tempest-devstack-vm-postgres-full/36627 : SUCCESS in 47m 53s\n- http://logs.openstack.org/71/39071/1/check/gate-tempest-devstack-vm-neutron/9005 : FAILURE in 17m 18s (non-voting)\n- http://logs.openstack.org/71/39071/1/check/gate-tempest-devstack-vm-neutron-full/1743 : FAILURE in 16m 13s (non-voting)\n- http://logs.openstack.org/71/39071/1/check/gate-grenade-devstack-vm/22365 : SUCCESS in 24m 11s\n","accounts_in_message":[],"_revision_number":1},{"id":"ad7c1d02419e0ee074fe5fbdd0e3fc250493adea","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2013-07-30 18:08:55.000000000","message":"Patch Set 1: Do not merge\n\nPer are discussions yesterday. You shouldn\u0027t do this with something imported from oslo. Also, I believe the move to oslo logging here: https://review.openstack.org/#/c/38879/ has addressed your concerns with the oslo log module. So there isn\u0027t a need for this anymore (if it were a valid change)","accounts_in_message":[],"_revision_number":1},{"id":"2c9ad106d37558e02dd2fe2cec06afbda2963ad0","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-07-30 18:22:07.000000000","message":"Patch Set 1: Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34efcb5e7f911dbb8bc135b1219de9a8cd252967","revisions":{"34efcb5e7f911dbb8bc135b1219de9a8cd252967":{"kind":"REWORK","_number":1,"created":"2013-07-29 13:27:06.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/71/39071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/39071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/39071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/39071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/39071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/39071/1"}}},"commit":{"parents":[{"commit":"c1fa97e7ee219ce3ecdf3c028390ae0462fbbc99","subject":"Merge \"Add posargs to testr-full tox job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c1fa97e7ee219ce3ecdf3c028390ae0462fbbc99"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2013-07-29 13:25:11.000000000","tz":120},"committer":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2013-07-29 13:25:11.000000000","tz":120},"subject":"Use tempest log with tempest/openstack/common/lockutils.py","message":"Use tempest log with tempest/openstack/common/lockutils.py\n\ntempest.common.log is specilaized for argparse less logging\nconfiguration, so it should be used by unittests.\n\nChange-Id: I07f2d94bd5652aed9f97241a3272b11a8a75785d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/34efcb5e7f911dbb8bc135b1219de9a8cd252967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/34efcb5e7f911dbb8bc135b1219de9a8cd252967"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c1fa97e7ee219ce3ecdf3c028390ae0462fbbc99","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
