)]}'
{"id":"openstack%2Ftempest~42071","triplet_id":"openstack%2Ftempest~master~I663cf1c994307160c9f0ebf51fa4c920f2b6125e","project":"openstack/tempest","branch":"master","topic":"cleanup","hashtags":[],"change_id":"I663cf1c994307160c9f0ebf51fa4c920f2b6125e","subject":"Cleanup: Add common \"create_server\" method","status":"MERGED","created":"2013-08-15 07:22:56.000000000","updated":"2013-08-16 16:47:56.000000000","submitted":"2013-08-16 16:47:55.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2e3c017dde52cf077a32393433ae80bebab6d697","_number":42071,"virtual_id_number":42071,"owner":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},{"value":2,"date":"2013-08-16 16:47:55.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"recommended":{"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},"all":[{"value":1,"date":"2013-08-16 08:32:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2013-08-16 09:26:18.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":2,"date":"2013-08-16 11:51:07.000000000","_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":1,"date":"2013-08-16 14:21:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"all":[{"value":0,"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2013-08-16 14:24:10.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"date":"2013-08-16 11:51:07.000000000","_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-08-16 03:48:28.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2013-08-16 08:32:02.000000000","updated_by":{"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},"reviewer":{"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},"state":"REVIEWER"},{"updated":"2013-08-16 11:51:07.000000000","updated_by":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"reviewer":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"state":"REVIEWER"},{"updated":"2013-08-16 14:21:36.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2013-08-16 14:24:10.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2013-08-16 16:47:55.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"c9df32ff5ef797752d2bb6fee9093bed7655e49b","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2013-08-15 15:14:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"60dba8ce74eae6316d3288825638520aead32919","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2013-08-15 22:46:06.000000000","message":"Patch Set 2:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":2},{"id":"a6ef022cc49b9b42a0ceb6a3189f1a2c798fdc20","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-15 23:36:44.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/71/42071/2/check/gate-tempest-pep8/50617bc : FAILURE in 3m 26s\n- http://logs.openstack.org/71/42071/2/check/gate-tempest-docs/6d4dc50 : SUCCESS in 6m 39s\n- http://logs.openstack.org/71/42071/2/check/gate-tempest-devstack-vm-full/834c202 : FAILURE in 39m 21s\n- http://logs.openstack.org/71/42071/2/check/gate-tempest-devstack-vm-cells-full/a895d94 : FAILURE in 14m 19s (non-voting)\n- http://logs.openstack.org/71/42071/2/check/gate-tempest-devstack-vm-testr-full/bcfee38 : FAILURE in 25m 46s (non-voting)\n- http://logs.openstack.org/71/42071/2/check/gate-tempest-devstack-vm-postgres-full/9ac7719 : FAILURE in 38m 43s\n- http://logs.openstack.org/71/42071/2/check/gate-tempest-devstack-vm-neutron/c9d88f0 : FAILURE in 18m 14s\n- http://logs.openstack.org/71/42071/2/check/gate-tempest-devstack-vm-neutron-full/7168741 : FAILURE in 50m 31s (non-voting)\n- http://logs.openstack.org/71/42071/2/check/gate-grenade-devstack-vm/f44d27d : SUCCESS in 25m 20s\n","accounts_in_message":[],"_revision_number":2},{"id":"0b40eb0b7ec4cd5e90f9eef2dd2c111d74efe7ac","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2013-08-15 23:43:44.000000000","message":"Patch Set 2: Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"a148abf2f6f6e8798fe829ed1082f8c5d510e7c6","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2013-08-16 00:08:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"58c9649aa475535f9dfc9195be368a67e627afc4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-16 00:59:56.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/71/42071/3/check/gate-tempest-pep8/d108eeb : SUCCESS in 3m 13s\n- http://logs.openstack.org/71/42071/3/check/gate-tempest-docs/a241bc2 : SUCCESS in 5m 32s\n- http://logs.openstack.org/71/42071/3/check/gate-tempest-devstack-vm-full/8ece428 : FAILURE in 39m 51s\n- http://logs.openstack.org/71/42071/3/check/gate-tempest-devstack-vm-cells-full/769a7fd : FAILURE in 13m 43s (non-voting)\n- http://logs.openstack.org/71/42071/3/check/gate-tempest-devstack-vm-testr-full/3b4d54c : FAILURE in 27m 35s (non-voting)\n- http://logs.openstack.org/71/42071/3/check/gate-tempest-devstack-vm-postgres-full/8d5c787 : FAILURE in 51m 25s\n- http://logs.openstack.org/71/42071/3/check/gate-tempest-devstack-vm-neutron/8580ca9 : SUCCESS in 17m 40s\n- http://logs.openstack.org/71/42071/3/check/gate-tempest-devstack-vm-neutron-full/a505077 : FAILURE in 45m 39s (non-voting)\n- http://logs.openstack.org/71/42071/3/check/gate-grenade-devstack-vm/ffe6229 : SUCCESS in 21m 43s\n","accounts_in_message":[],"_revision_number":3},{"id":"148c284d13501db5b6a46400fcec7940e4045a39","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2013-08-16 01:14:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4a8049f913be5fdda7cbfb1f4c070a115f9daf80","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-16 02:42:34.000000000","message":"Patch Set 4: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/71/42071/4/check/gate-tempest-pep8/09f667d : SUCCESS in 3m 21s\n- http://logs.openstack.org/71/42071/4/check/gate-tempest-docs/5e90d93 : SUCCESS in 5m 13s\n- http://logs.openstack.org/71/42071/4/check/gate-tempest-devstack-vm-full/ab8704b : SUCCESS in 50m 56s\n- http://logs.openstack.org/71/42071/4/check/gate-tempest-devstack-vm-cells-full/89bf578 : FAILURE in 14m 25s (non-voting)\n- http://logs.openstack.org/71/42071/4/check/gate-tempest-devstack-vm-testr-full/0907f9b : FAILURE in 26m 01s (non-voting)\n- http://logs.openstack.org/71/42071/4/check/gate-tempest-devstack-vm-postgres-full/1d81c89 : FAILURE in 1h 27m 29s\n- http://logs.openstack.org/71/42071/4/check/gate-tempest-devstack-vm-neutron/daa4c42 : FAILURE in 15m 07s\n- http://logs.openstack.org/71/42071/4/check/gate-tempest-devstack-vm-neutron-full/6b06ba2 : FAILURE in 47m 54s (non-voting)\n- http://logs.openstack.org/71/42071/4/check/gate-grenade-devstack-vm/e6dc062 : SUCCESS in 26m 23s\n","accounts_in_message":[],"_revision_number":4},{"id":"d0182c6db32fd2f45334b64dd03008fbf5cc0921","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2013-08-16 02:54:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"63428118faca468548f523179e8ad2bda9cd6d0b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-16 03:41:32.000000000","message":"Patch Set 5: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/71/42071/5/check/gate-tempest-pep8/6a60ba6 : SUCCESS in 3m 26s\n- http://logs.openstack.org/71/42071/5/check/gate-tempest-docs/0c30a2d : SUCCESS in 5m 20s\n- http://logs.openstack.org/71/42071/5/check/gate-tempest-devstack-vm-full/5e5d392 : SUCCESS in 40m 14s\n- http://logs.openstack.org/71/42071/5/check/gate-tempest-devstack-vm-cells-full/1069522 : FAILURE in 12m 22s (non-voting)\n- http://logs.openstack.org/71/42071/5/check/gate-tempest-devstack-vm-testr-full/4889870 : SUCCESS in 25m 59s (non-voting)\n- http://logs.openstack.org/71/42071/5/check/gate-tempest-devstack-vm-postgres-full/63afd70 : SUCCESS in 45m 26s\n- http://logs.openstack.org/71/42071/5/check/gate-tempest-devstack-vm-neutron/3f43a63 : SUCCESS in 17m 21s\n- http://logs.openstack.org/71/42071/5/check/gate-tempest-devstack-vm-neutron-full/554f38b : FAILURE in 47m 20s (non-voting)\n- http://logs.openstack.org/71/42071/5/check/gate-grenade-devstack-vm/b0cf2ac : SUCCESS in 23m 09s\n","accounts_in_message":[],"_revision_number":5},{"id":"6e22e03421246c75b3a70ddaf2eebaef171bf970","author":{"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},"date":"2013-08-16 06:35:39.000000000","message":"Patch Set 5: Looks good to me, but someone else must approve\n\n(1 inline comment)\n\nGood work!","accounts_in_message":[],"_revision_number":5},{"id":"45b93f0f29be686ee2bf452802b63771c40ba32a","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2013-08-16 06:40:11.000000000","message":"Patch Set 5: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":5},{"id":"fffb44e62b5bccad61b2340af36b418011cefd16","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2013-08-16 06:40:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c8b46c43395c346b74a5b82318d4b9d1e8972cc8","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-08-16 06:41:04.000000000","message":"Patch Set 6:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":6},{"id":"aabb318dc02dcbde7c55bd06310957714c5d5756","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-16 07:29:25.000000000","message":"Patch Set 6: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/71/42071/6/check/gate-tempest-pep8/057c5d9 : SUCCESS in 3m 21s\n- http://logs.openstack.org/71/42071/6/check/gate-tempest-docs/d728c41 : SUCCESS in 5m 26s\n- http://logs.openstack.org/71/42071/6/check/gate-tempest-devstack-vm-full/c7986bc : SUCCESS in 41m 13s\n- http://logs.openstack.org/71/42071/6/check/gate-tempest-devstack-vm-cells-full/8ed46d1 : FAILURE in 13m 14s (non-voting)\n- http://logs.openstack.org/71/42071/6/check/gate-tempest-devstack-vm-testr-full/6df199d : FAILURE in 23m 50s (non-voting)\n- http://logs.openstack.org/71/42071/6/check/gate-tempest-devstack-vm-postgres-full/fc76f5d : SUCCESS in 41m 25s\n- http://logs.openstack.org/71/42071/6/check/gate-tempest-devstack-vm-neutron/39fe2ab : SUCCESS in 18m 06s\n- http://logs.openstack.org/71/42071/6/check/gate-tempest-devstack-vm-neutron-full/8f942df : FAILURE in 48m 19s (non-voting)\n- http://logs.openstack.org/71/42071/6/check/gate-grenade-devstack-vm/d0177a5 : SUCCESS in 21m 59s\n","accounts_in_message":[],"_revision_number":6},{"id":"d68a1448bb5e09a634f89a25aadd40a5d9d1d878","author":{"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},"date":"2013-08-16 08:32:02.000000000","message":"Patch Set 6: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":6},{"id":"e6ae15b02ed82da6e7057980a45f9d3af50b3b35","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-08-16 09:26:18.000000000","message":"Patch Set 6: Looks good to me (core reviewer)\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"001df705a88fa87ae9111e00b5a543a450527578","author":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"date":"2013-08-16 11:51:07.000000000","message":"Patch Set 6: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":6},{"id":"4ceaaf3fb2f693c71e8a2bdaa8bb54bb93b25544","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2013-08-16 12:32:04.000000000","message":"Patch Set 6: Looks good to me, but someone else must approve\n\n(1 inline comment)\n\nI think this is a nice change so I\u0027m +1. Only, do we really need to catch the AttributeError exception? See my inline comment.","accounts_in_message":[],"_revision_number":6},{"id":"e177f67a18096a5b912ebd3b6236b4448949e64f","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2013-08-16 12:49:42.000000000","message":"Patch Set 6: I would prefer that you didn\u0027t merge this\n\nI\u0027m now changing this to -1 for two reasons:\n1. my inline question from previous review\n2. I see all the utility methods from NetworkScenarioTest are hidden (as in starting with an underscore); why they are? should this method be too?","accounts_in_message":[],"_revision_number":6},{"id":"485835180251a39289139eac8c950718ddba0242","author":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"date":"2013-08-16 12:54:06.000000000","message":"Patch Set 6: (1 inline comment)\n\nGuilo - I don\u0027t think create_server needs to be _create_server. In other places in tempest where we\u0027ve had common create functions like this we haven\u0027t done that.","accounts_in_message":[],"_revision_number":6},{"id":"d18c5c596191bb86b25a5f075aa1299505e97218","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2013-08-16 12:57:37.000000000","message":"Patch Set 6: (1 inline comment)\n\n@cyeoh: agreed, I think it is the other methods in NetworkScenarioTest which shouldn\u0027t be hidden either","accounts_in_message":[],"_revision_number":6},{"id":"62c8b434b80385c44b8c922fbad57bd953820966","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-08-16 14:14:09.000000000","message":"Patch Set 6: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":6},{"id":"e0aa986c336efcf0c52f9ede77bfb02fc9c65123","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2013-08-16 14:21:36.000000000","message":"Patch Set 6: Looks good to me, but someone else must approve\n\nafter a brief chat with afazekas on IRC I decided to +1 so we start improving things by small steps\n\nmy proposal for the next qa meeting will be some sort of \"cleanup\" blueprint where we try to list as TODOs all sort of inconsistencies and slowly work on them","accounts_in_message":[],"_revision_number":6},{"id":"d35c69259ac4b6bf7391e157af15a7ea8070de4f","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-08-16 14:24:10.000000000","message":"Patch Set 6: Approved\n\nthank you","accounts_in_message":[],"_revision_number":6},{"id":"5130673522f69e253e8ef1b02835b86a52382656","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-16 14:24:12.000000000","message":"Patch Set 6: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"8b28fad32f797c6d9600b3b21987d10cfddd43a2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-16 16:47:55.000000000","message":"Patch Set 6: Verified\n\nBuild succeeded.\n\n- http://logs.openstack.org/71/42071/6/gate/gate-tempest-pep8/3ff674e : SUCCESS in 3m 20s\n- http://logs.openstack.org/71/42071/6/gate/gate-tempest-docs/0f7ee0c : SUCCESS in 5m 16s\n- http://logs.openstack.org/71/42071/6/gate/gate-tempest-devstack-vm-full/3444bde : SUCCESS in 40m 59s\n- http://logs.openstack.org/71/42071/6/gate/gate-tempest-devstack-vm-testr-full/edaf147 : FAILURE in 26m 38s (non-voting)\n- http://logs.openstack.org/71/42071/6/gate/gate-tempest-devstack-vm-postgres-full/e357364 : SUCCESS in 42m 20s\n- http://logs.openstack.org/71/42071/6/gate/gate-tempest-devstack-vm-neutron/a523045 : SUCCESS in 21m 22s\n- http://logs.openstack.org/71/42071/6/gate/gate-tempest-devstack-vm-neutron-full/bcee977 : FAILURE in 55m 33s (non-voting)\n- http://logs.openstack.org/71/42071/6/gate/gate-grenade-devstack-vm/9264c63 : SUCCESS in 22m 40s\n","accounts_in_message":[],"_revision_number":6},{"id":"b6e09c17438e72b5af72b8e1aee6da73c2e63302","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-16 16:47:56.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"61f272b03dc79f97f475af2cb6e3f2cf86bedf84","revisions":{"a96aadd668c2d11965285d98e21d211c2d5d2003":{"kind":"REWORK","_number":1,"created":"2013-08-15 07:22:56.000000000","uploader":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"ref":"refs/changes/71/42071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/42071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/42071/1"}}},"commit":{"parents":[{"commit":"bc8dbd315c5dee6920011894bcec891d97f6f75c","subject":"Cleanup try/except/finally blocks in several tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bc8dbd315c5dee6920011894bcec891d97f6f75c"}]}],"author":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2013-08-15 06:58:53.000000000","tz":540},"committer":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2013-08-15 07:17:53.000000000","tz":540},"subject":"Cleanup: Add common \"create_server\" method","message":"Cleanup: Add common \"create_server\" method\n\nIn scenario tests, there are many \"create server\" methods and their\nbehaviors are almost same. They creates a server with randam name,\ndefault image and default flavor, and waits for server status \"ACTIVE\".\nSo this patch adds a common \"create_server\" method and uses the method\ninstead of each implementations.\n\nChange-Id: I663cf1c994307160c9f0ebf51fa4c920f2b6125e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a96aadd668c2d11965285d98e21d211c2d5d2003"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a96aadd668c2d11965285d98e21d211c2d5d2003"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bc8dbd315c5dee6920011894bcec891d97f6f75c","is_merged_in_target_branch":true,"change_id":"I6e9532e5067bbc0458ed3dc3faac58151aa696a0","change_number":40005,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ea6bc5209b3806884f9048c471896b09e45fc664":{"kind":"REWORK","_number":2,"created":"2013-08-15 15:14:13.000000000","uploader":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"ref":"refs/changes/71/42071/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/42071/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/42071/2"}}},"commit":{"parents":[{"commit":"5e293ccf2809fad87c80dfd9f36447fe31e5b32e","subject":"Merge \"setUpClass/tearDownClass full chain\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5e293ccf2809fad87c80dfd9f36447fe31e5b32e"}]}],"author":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2013-08-15 06:58:53.000000000","tz":540},"committer":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2013-08-15 15:11:31.000000000","tz":540},"subject":"Cleanup: Add common \"create_server\" method","message":"Cleanup: Add common \"create_server\" method\n\nIn scenario tests, there are many \"create server\" methods and their\nbehaviors are almost same. They creates a server with randam name,\ndefault image and default flavor, and waits for server status \"ACTIVE\".\nSo this patch adds a common \"create_server\" method and uses the method\ninstead of each implementations.\n\nChange-Id: I663cf1c994307160c9f0ebf51fa4c920f2b6125e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ea6bc5209b3806884f9048c471896b09e45fc664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ea6bc5209b3806884f9048c471896b09e45fc664"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5e293ccf2809fad87c80dfd9f36447fe31e5b32e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b8c0875dd54ef64451dc3c08c0d04842efa5593b":{"kind":"REWORK","_number":3,"created":"2013-08-16 00:08:25.000000000","uploader":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"ref":"refs/changes/71/42071/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/42071/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/42071/3"}}},"commit":{"parents":[{"commit":"5e293ccf2809fad87c80dfd9f36447fe31e5b32e","subject":"Merge \"setUpClass/tearDownClass full chain\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5e293ccf2809fad87c80dfd9f36447fe31e5b32e"}]}],"author":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2013-08-15 06:58:53.000000000","tz":540},"committer":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2013-08-16 00:03:25.000000000","tz":540},"subject":"Cleanup: Add common \"create_server\" method","message":"Cleanup: Add common \"create_server\" method\n\nIn scenario tests, there are many \"create server\" methods and their\nbehaviors are almost same. They creates a server with randam name,\ndefault image and default flavor, and waits for server status \"ACTIVE\".\nSo this patch adds a common \"create_server\" method and uses the method\ninstead of each implementations.\n\nChange-Id: I663cf1c994307160c9f0ebf51fa4c920f2b6125e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b8c0875dd54ef64451dc3c08c0d04842efa5593b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b8c0875dd54ef64451dc3c08c0d04842efa5593b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5e293ccf2809fad87c80dfd9f36447fe31e5b32e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"19fc4e784bebf08842365b77926fc5b6ffdd1faa":{"kind":"REWORK","_number":4,"created":"2013-08-16 01:14:59.000000000","uploader":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"ref":"refs/changes/71/42071/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/42071/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/42071/4"}}},"commit":{"parents":[{"commit":"5e293ccf2809fad87c80dfd9f36447fe31e5b32e","subject":"Merge \"setUpClass/tearDownClass full chain\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5e293ccf2809fad87c80dfd9f36447fe31e5b32e"}]}],"author":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2013-08-15 06:58:53.000000000","tz":540},"committer":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2013-08-16 01:06:01.000000000","tz":540},"subject":"Cleanup: Add common \"create_server\" method","message":"Cleanup: Add common \"create_server\" method\n\nIn scenario tests, there are many \"create server\" methods and their\nbehaviors are almost same. They creates a server with randam name,\ndefault image and default flavor, and waits for server status \"ACTIVE\".\nSo this patch adds a common \"create_server\" method and uses the method\ninstead of each implementation for code-cleanup.\n\nChange-Id: I663cf1c994307160c9f0ebf51fa4c920f2b6125e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/19fc4e784bebf08842365b77926fc5b6ffdd1faa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/19fc4e784bebf08842365b77926fc5b6ffdd1faa"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5e293ccf2809fad87c80dfd9f36447fe31e5b32e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"049f371d56c7ece0cc7740e313a6ad2f7ae14fc8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2013-08-16 02:54:01.000000000","uploader":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"ref":"refs/changes/71/42071/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/42071/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/42071/5"}}},"commit":{"parents":[{"commit":"4369755ae2c7e11b57a63edefac6841773d34e09","subject":"Merge \"Adding extra_specs to flavor format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4369755ae2c7e11b57a63edefac6841773d34e09"}]}],"author":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2013-08-15 06:58:53.000000000","tz":540},"committer":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2013-08-16 02:47:38.000000000","tz":540},"subject":"Cleanup: Add common \"create_server\" method","message":"Cleanup: Add common \"create_server\" method\n\nIn scenario tests, there are many \"create server\" methods and their\nbehaviors are almost same. They creates a server with randam name,\ndefault image and default flavor, and waits for server status \"ACTIVE\".\nSo this patch adds a common \"create_server\" method and uses the method\ninstead of each implementation for code-cleanup.\n\nChange-Id: I663cf1c994307160c9f0ebf51fa4c920f2b6125e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/049f371d56c7ece0cc7740e313a6ad2f7ae14fc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/049f371d56c7ece0cc7740e313a6ad2f7ae14fc8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4369755ae2c7e11b57a63edefac6841773d34e09","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"61f272b03dc79f97f475af2cb6e3f2cf86bedf84":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2013-08-16 06:40:58.000000000","uploader":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"ref":"refs/changes/71/42071/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/42071/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/42071/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/42071/6"}}},"commit":{"parents":[{"commit":"4369755ae2c7e11b57a63edefac6841773d34e09","subject":"Merge \"Adding extra_specs to flavor format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4369755ae2c7e11b57a63edefac6841773d34e09"}]}],"author":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2013-08-15 06:58:53.000000000","tz":540},"committer":{"name":"Ken\u0027ichi Ohmichi","email":"oomichi@mxs.nes.nec.co.jp","date":"2013-08-16 06:34:43.000000000","tz":540},"subject":"Cleanup: Add common \"create_server\" method","message":"Cleanup: Add common \"create_server\" method\n\nIn scenario tests, there are many \"create server\" methods and their\nbehaviors are almost same. They creates a server with random name,\ndefault image and default flavor, and waits for server status \"ACTIVE\".\nSo this patch adds a common \"create_server\" method and uses the method\ninstead of each implementation for code-cleanup.\n\nChange-Id: I663cf1c994307160c9f0ebf51fa4c920f2b6125e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/61f272b03dc79f97f475af2cb6e3f2cf86bedf84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/61f272b03dc79f97f475af2cb6e3f2cf86bedf84"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4369755ae2c7e11b57a63edefac6841773d34e09","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
