)]}'
{"id":"openstack%2Ftempest~474164","triplet_id":"openstack%2Ftempest~master~If42561bfc1d6dcffbb6537634d7769ef90fab21d","project":"openstack/tempest","branch":"master","topic":"plugin-sanity","hashtags":[],"change_id":"If42561bfc1d6dcffbb6537634d7769ef90fab21d","subject":"Added script for doing tempest plugin sanity","status":"MERGED","created":"2017-06-14 11:41:39.000000000","updated":"2017-07-11 13:24:49.000000000","submitted":"2017-07-11 13:24:49.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":31,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3abfad9c00e827ce1a41e2b76e4aa33ce34c316e","_number":474164,"virtual_id_number":474164,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2017-07-11 13:24:49.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2017-07-11 06:57:29.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":2,"date":"2017-07-11 10:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":1,"date":"2017-07-11 10:00:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-06-14 11:49:44.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2017-06-14 13:30:37.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2017-06-14 14:02:43.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2017-06-15 16:52:19.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-06-21 07:16:09.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2017-06-21 07:17:02.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2017-06-26 11:40:42.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2017-07-06 11:22:47.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2017-07-06 15:50:18.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-07-11 06:57:29.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2017-07-11 10:00:39.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2017-07-11 13:24:49.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"296ab472c70ac7bc2d14c019a762f050e725c8c7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-14 11:41:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee9774b51ee93eec8efa0dc67aa4c7673aa7a83b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-14 11:46:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"679eb2abd0f3740eea3af2da8e222cd837abab7f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-14 11:49:29.000000000","message":"Patch Set 2:\n\nUsing the above in project-config https://review.openstack.org/#/c/474168/","accounts_in_message":[],"_revision_number":2},{"id":"e0203e3b89040d6a2d4e92ad1ae9e08bbd6499ee","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-14 11:58:26.000000000","message":"Patch Set 3: Published edit on patch set 2","accounts_in_message":[],"_revision_number":3},{"id":"974772b1de9d0bf18efff72b4dd3c62452389dd6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-14 13:28:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/70a7db7/ : SUCCESS in 1h 34m 33s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/2453cf1/ : SUCCESS in 37m 38s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/2/check/gate-tempest-dsvm-py35-ubuntu-xenial/09732e7/ : SUCCESS in 1h 14m 42s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/2/check/gate-tempest-docs-ubuntu-xenial/fe567c0//doc/build/html/ : SUCCESS in 3m 52s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/2/check/gate-tempest-pep8-ubuntu-xenial/a42ceb4/ : SUCCESS in 1m 59s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/2/check/gate-tempest-python27-ubuntu-xenial/5217e55/ : SUCCESS in 2m 19s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/2/check/gate-tempest-python35/1956c30/ : SUCCESS in 2m 04s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/2/check/gate-tempest-releasenotes/65c802e//releasenotes/build/html/ : SUCCESS in 3m 58s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/2/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/9992cba/ : SUCCESS in 57m 09s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/38c9b8a/ : SUCCESS in 49m 51s (non-voting)\n- gate-tempest-dsvm-neutron-full-ssh http://logs.openstack.org/64/474164/2/check/gate-tempest-dsvm-neutron-full-ssh/ce3184c/ : FAILURE in 1h 13m 34s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/64/474164/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/8b593e7/ : SUCCESS in 1h 37m 09s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/046b459/ : SUCCESS in 1h 12m 33s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/fc69ca1/ : SUCCESS in 1h 00m 00s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/2a00b6a/ : SUCCESS in 37m 16s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/7090930/ : SUCCESS in 46m 40s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/5ee65a7/ : SUCCESS in 31m 11s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/2/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/07b53aa/ : SUCCESS in 1h 37m 14s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/df049ed/ : SUCCESS in 1h 32m 02s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/fd0bd5e/ : SUCCESS in 1h 26m 07s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/2/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/fb0a5e0/ : FAILURE in 9m 50s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/2/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/151ed6d/ : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d68c23b2b5c1f8f7952587a66504d041ee52805a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2017-06-14 13:30:37.000000000","message":"Patch Set 3:\n\nI would remove the Depends-On (there is a circular dependency, and this change should be merged anyway before the other, even if it can\u0027t be tested).","accounts_in_message":[],"_revision_number":3},{"id":"def12874b7596a4f68d62bb01e7bb86350841572","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-14 13:37:46.000000000","message":"Patch Set 4: Published edit on patch set 3","accounts_in_message":[],"_revision_number":4},{"id":"fc78380f4c7aa34cff9e41aa99dbff686dada93b","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-06-14 14:02:43.000000000","message":"Patch Set 2:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-neutron-full-ssh: https://bugs.launchpad.net/bugs/1688038\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":2},{"id":"1a5bb9e47930a2616ac1bd2981b3a192157c4263","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-14 15:00:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"69aec5a1de9ef9b68552e12f8b25096c31d2546f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-14 16:54:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/5/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/6ccc984/ : SUCCESS in 1h 27m 05s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/5/check/gate-grenade-dsvm-neutron-ubuntu-xenial/ad9b4ff/ : SUCCESS in 37m 35s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/5/check/gate-tempest-dsvm-py35-ubuntu-xenial/0af2276/ : SUCCESS in 1h 21m 12s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/5/check/gate-tempest-docs-ubuntu-xenial/51214ed//doc/build/html/ : SUCCESS in 7m 06s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/5/check/gate-tempest-pep8-ubuntu-xenial/4c814dc/ : SUCCESS in 3m 13s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/5/check/gate-tempest-python27-ubuntu-xenial/3cb4d92/ : SUCCESS in 2m 19s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/5/check/gate-tempest-python35/db6e08a/ : SUCCESS in 2m 27s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/5/check/gate-tempest-releasenotes/7414976//releasenotes/build/html/ : SUCCESS in 2m 46s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/5/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/653a7b5/ : SUCCESS in 40m 57s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/5/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/b6465ab/ : SUCCESS in 49m 28s (non-voting)\n- gate-tempest-dsvm-neutron-full-ssh http://logs.openstack.org/64/474164/5/check/gate-tempest-dsvm-neutron-full-ssh/85dea02/ : SUCCESS in 1h 34m 08s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/64/474164/5/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/1cfff15/ : SUCCESS in 1h 37m 53s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/f830f23/ : SUCCESS in 1h 16m 22s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/5/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/8aa92f1/ : SUCCESS in 48m 14s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/5/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/f4d5bdb/ : SUCCESS in 41m 25s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/5/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/b0d174e/ : SUCCESS in 50m 42s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/5/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/3f9700e/ : SUCCESS in 31m 10s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/5/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/8882868/ : SUCCESS in 1h 50m 04s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/5/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/28468a2/ : SUCCESS in 1h 17m 04s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/5/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/ab6e4ce/ : SUCCESS in 1h 12m 40s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/5/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/afbb2c5/ : FAILURE in 1h 02m 21s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/5/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/39a4346/ : FAILURE in 1m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cbbb8f1169e43971b601aae70814d23a5810bcdd","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-06-14 17:01:28.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/5/check/dsvm-tempest-neutron-network/6b865aa : SUCCESS in 1h 56m 09s","accounts_in_message":[],"_revision_number":5},{"id":"66d697b4c6d43e701bd22bdcf83df04b3ce4c8c8","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-15 08:03:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3eb60c6325c59d67abddb9733d808fb1dd9a3135","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-15 09:43:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/f75b5e9/ : SUCCESS in 1h 13m 12s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/6/check/gate-grenade-dsvm-neutron-ubuntu-xenial/e431563/ : SUCCESS in 46m 57s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/6/check/gate-tempest-dsvm-py35-ubuntu-xenial/42e141c/ : SUCCESS in 1h 22m 26s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/6/check/gate-tempest-docs-ubuntu-xenial/b6d2433//doc/build/html/ : SUCCESS in 7m 59s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/6/check/gate-tempest-pep8-ubuntu-xenial/959c82b/ : SUCCESS in 2m 56s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/6/check/gate-tempest-python27-ubuntu-xenial/2a2dd5b/ : SUCCESS in 1m 44s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/6/check/gate-tempest-python35/cc329bb/ : SUCCESS in 2m 05s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/6/check/gate-tempest-releasenotes/936029f//releasenotes/build/html/ : SUCCESS in 3m 16s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/6/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/b56d8da/ : SUCCESS in 48m 07s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/6/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/966b975/ : SUCCESS in 37m 27s (non-voting)\n- gate-tempest-dsvm-neutron-full-ssh http://logs.openstack.org/64/474164/6/check/gate-tempest-dsvm-neutron-full-ssh/230399e/ : SUCCESS in 1h 22m 47s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/64/474164/6/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/51ed1e2/ : SUCCESS in 1h 13m 50s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/5ac6ae5/ : SUCCESS in 1h 16m 22s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/6/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/182449c/ : SUCCESS in 56m 27s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/6/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/31b0f2d/ : SUCCESS in 42m 03s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/6/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/975fa04/ : SUCCESS in 45m 05s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/6/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/af8e520/ : SUCCESS in 31m 27s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/6/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/e19eb82/ : SUCCESS in 1h 37m 46s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/4fb13d9/ : SUCCESS in 1h 22m 13s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/3c866d7/ : SUCCESS in 1h 16m 18s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/6/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/e8226c3/ : SUCCESS in 1h 18m 06s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/6/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/4f42b2d/ : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c2ac0b2638ed4fb7ac1c4b8d638d8c9fc9f6a07b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-06-15 11:16:43.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/6/check/dsvm-tempest-neutron-network/1caf71c : FAILURE in 3h 09m 31s","accounts_in_message":[],"_revision_number":6},{"id":"d85992520e06e671292b0812e56bf9c552924e87","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2017-06-15 16:52:19.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/xio_ostack/EMC_XtremIO_FC/1231 : FAILURE in 15m 32s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/xio_ostack/EMC_XtremIO_iSCSI/1412 : FAILURE in 13m 49s\n\nLeave a comment with \u0027run emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":6},{"id":"cf2829dd5d1f92a40525d5374712b1f53b42da81","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2017-06-20 08:34:18.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)\n\nIt looks mostly good, thank you - a couple of comments inline.\nI would like to make sure this runs against all plugins, even if one in the middle fails.","accounts_in_message":[],"_revision_number":6},{"id":"e10bc9a6cd6c6dc48b8eaf8544462a36290a3c30","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-20 18:28:21.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"85a80b5ac998b48d88e0fbc920776d1a813ab6c1","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-20 18:29:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0ebcf09cb48b2f1ce7bb9e4f234fac6137bdcba8","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-06-20 20:05:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/7/check/dsvm-tempest-neutron-network/023976d : SUCCESS in 1h 35m 20s","accounts_in_message":[],"_revision_number":7},{"id":"5f5f8488f36d9f4b5ee08d2d94fcfc56f1e3504f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-20 20:07:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a6116a3/ : SUCCESS in 1h 14m 10s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-grenade-dsvm-neutron-ubuntu-xenial/2f542e7/ : SUCCESS in 52m 49s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-py35-ubuntu-xenial/c990515/ : SUCCESS in 1h 26m 52s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/7/check/gate-tempest-docs-ubuntu-xenial/ca06128//doc/build/html/ : SUCCESS in 2m 03s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-pep8-ubuntu-xenial/96a2d72/ : SUCCESS in 2m 10s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-python27-ubuntu-xenial/8c0e1ca/ : SUCCESS in 2m 34s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/7/check/gate-tempest-python35/dc0b179/ : SUCCESS in 2m 42s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/7/check/gate-tempest-releasenotes/021b781//releasenotes/build/html/ : SUCCESS in 3m 00s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/58f669a/ : SUCCESS in 50m 01s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/0d1c197/ : SUCCESS in 35m 23s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/79f5271/ : SUCCESS in 1h 13m 15s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/b40f786/ : SUCCESS in 50m 03s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/ebd55a6/ : SUCCESS in 51m 54s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/1e33f96/ : SUCCESS in 1h 00m 49s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/9846509/ : SUCCESS in 40m 04s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/3777a91/ : SUCCESS in 1h 35m 32s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/93fe2cd/ : SUCCESS in 1h 32m 20s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/df09d6a/ : FAILURE in 1h 02m 25s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/226daac/ : SUCCESS in 1h 00m 27s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/e2e9be0/ : SUCCESS in 1m 19s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/tempest-coverage-ubuntu-xenial-nv/6b5251c/ : FAILURE in 2m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f9a7810f6adecbb63de96573b9b5347fa25dde38","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-20 20:21:43.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"90fe9a1cacde090e3ee8764fffc68c9f77eb8c46","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-20 21:49:42.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/9db673b/ : SUCCESS in 1h 26m 54s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-grenade-dsvm-neutron-ubuntu-xenial/59ea4fa/ : SUCCESS in 47m 42s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-py35-ubuntu-xenial/0fadd80/ : FAILURE in 59m 38s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/7/check/gate-tempest-docs-ubuntu-xenial/dc91734//doc/build/html/ : SUCCESS in 1m 55s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-pep8-ubuntu-xenial/bb5e59d/ : SUCCESS in 1m 51s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-python27-ubuntu-xenial/9ab8f48/ : SUCCESS in 1m 50s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/7/check/gate-tempest-python35/144bfca/ : SUCCESS in 1m 07s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/7/check/gate-tempest-releasenotes/aa81092//releasenotes/build/html/ : SUCCESS in 3m 19s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/0a4bc99/ : SUCCESS in 38m 54s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/159fc46/ : SUCCESS in 44m 01s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/1462b14/ : SUCCESS in 1h 13m 05s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/88d3f35/ : SUCCESS in 53m 06s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/aeb7b5a/ : FAILURE in 42m 20s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/8cdc076/ : SUCCESS in 58m 56s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/38d51b0/ : SUCCESS in 29m 50s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/3c2b94a/ : SUCCESS in 1h 20m 57s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/09a09e4/ : SUCCESS in 1h 17m 22s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/11ea5c0/ : FAILURE in 1h 09m 32s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/4109a4b/ : SUCCESS in 1h 00m 03s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/e1ba26d/ : SUCCESS in 1m 22s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/tempest-coverage-ubuntu-xenial-nv/07503df/ : SUCCESS in 1m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"eb83caef20cb8328c45f896fdeb3ce5234eb17a7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-21 13:22:45.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"5c3394b148aaed6e7331971912369cd5b5a49e25","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-21 15:09:19.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d9c05b5/ : SUCCESS in 1h 21m 28s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-grenade-dsvm-neutron-ubuntu-xenial/efc6942/ : SUCCESS in 36m 09s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-py35-ubuntu-xenial/0eb9da3/ : SUCCESS in 1h 02m 44s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/7/check/gate-tempest-docs-ubuntu-xenial/cb44389//doc/build/html/ : SUCCESS in 2m 45s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-pep8-ubuntu-xenial/cdcbc75/ : SUCCESS in 4m 26s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-python27-ubuntu-xenial/a6a70db/ : SUCCESS in 1m 48s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/7/check/gate-tempest-python35/a0f7e1c/ : SUCCESS in 2m 21s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/7/check/gate-tempest-releasenotes/5bd5858//releasenotes/build/html/ : SUCCESS in 2m 53s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/ff06be5/ : SUCCESS in 52m 19s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/803e58b/ : FAILURE in 22m 07s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/68c6580/ : SUCCESS in 1h 02m 43s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/c89bd84/ : SUCCESS in 47m 26s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/3d82054/ : SUCCESS in 41m 30s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/691f114/ : SUCCESS in 50m 15s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/52163d6/ : SUCCESS in 30m 23s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/8151b86/ : SUCCESS in 1h 29m 13s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/77943be/ : SUCCESS in 1h 16m 17s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/01683a4/ : FAILURE in 56m 02s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/2ece9ed/ : SUCCESS in 1h 15m 30s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/ef887fe/ : SUCCESS in 2m 17s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/tempest-coverage-ubuntu-xenial-nv/dba134b/ : FAILURE in 2m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"27272bab76b8214cc48b9617e87cede8c0e3ec2e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-22 07:22:55.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"45e450ecc3d33198987f04554873e509560f66ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-22 08:56:25.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/402e4ba/ : SUCCESS in 1h 30m 36s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-grenade-dsvm-neutron-ubuntu-xenial/deae38f/ : SUCCESS in 51m 14s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-py35-ubuntu-xenial/2f67e0c/ : SUCCESS in 1h 20m 02s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/7/check/gate-tempest-docs-ubuntu-xenial/e18dac0//doc/build/html/ : SUCCESS in 13m 09s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-pep8-ubuntu-xenial/7813942/ : SUCCESS in 3m 04s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-python27-ubuntu-xenial/004cb95/ : FAILURE in 1m 56s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/7/check/gate-tempest-python35/c926ed9/ : FAILURE in 1m 58s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/7/check/gate-tempest-releasenotes/a4ebb14//releasenotes/build/html/ : SUCCESS in 4m 03s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/8d628ce/ : SUCCESS in 47m 22s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/686c2a4/ : FAILURE in 33m 54s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/fa9445a/ : SUCCESS in 1h 17m 23s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/f315a72/ : SUCCESS in 1h 00m 30s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/027f028/ : SUCCESS in 54m 08s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/9daa6d3/ : SUCCESS in 44m 37s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/7/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/914fa37/ : SUCCESS in 34m 01s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/205f297/ : SUCCESS in 1h 30m 41s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/20dcadf/ : SUCCESS in 1h 24m 06s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/3cb40f8/ : FAILURE in 56m 45s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/e08c81d/ : SUCCESS in 56m 43s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/132b412/ : SUCCESS in 3m 37s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/7/check/tempest-coverage-ubuntu-xenial-nv/c4db359/ : SUCCESS in 1m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e20076c8c5ae29d98bf7a001cbbd2af066696f82","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-23 05:33:11.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"84e679270646b663c0c797821056659a2ad8b5fe","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-06-23 05:55:58.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9da5cea8acfcb448602238db52d4e21b943c39df","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-06-23 06:01:22.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/8/check/dsvm-tempest-neutron-network/c7d6384 : FAILURE in 27m 42s","accounts_in_message":[],"_revision_number":8},{"id":"66f148eb0ab102ffaefc20b7363843c3ccf11a15","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-23 06:27:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/96a0d51/ : FAILURE in 13m 35s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/8/check/gate-grenade-dsvm-neutron-ubuntu-xenial/11d12dd/ : FAILURE in 11m 29s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/8/check/gate-tempest-dsvm-py35-ubuntu-xenial/e73ce8d/ : FAILURE in 13m 37s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/8/check/gate-tempest-docs-ubuntu-xenial/4e3d960//doc/build/html/ : SUCCESS in 13m 08s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/8/check/gate-tempest-pep8-ubuntu-xenial/9135c4f/ : SUCCESS in 3m 22s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/8/check/gate-tempest-python27-ubuntu-xenial/ad56ddf/ : SUCCESS in 1m 04s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/8/check/gate-tempest-python35/a987674/ : SUCCESS in 1m 07s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/8/check/gate-tempest-releasenotes/85fcd2f//releasenotes/build/html/ : SUCCESS in 2m 03s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/8/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/ff77864/ : FAILURE in 15m 46s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/8/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/c279550/ : FAILURE in 6m 26s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/8/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/5871fe7/ : FAILURE in 7m 05s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/8/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/59ea3b5/ : SUCCESS in 53m 30s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/8/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/d2d9855/ : SUCCESS in 38m 40s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/8/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/39d7410/ : SUCCESS in 45m 10s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/8/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/9b9bb11/ : SUCCESS in 32m 49s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/8/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/c9bcd5b/ : FAILURE in 10m 57s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/56ebce4/ : FAILURE in 12m 20s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/0555bb4/ : FAILURE in 12m 19s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/8/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/bbe6f37/ : FAILURE in 12m 23s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/8/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/cf94c8e/ : SUCCESS in 2m 16s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/8/check/tempest-coverage-ubuntu-xenial-nv/89fc359/ : SUCCESS in 1m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ed5009c22ff0388409e3eaffff06194d4c9f445d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-23 13:48:57.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f0774a7ca3d1ef1ef8c1531b66426d1d9ad7b16e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-23 13:53:45.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e7be7f3116e023955b47648581b05b52af9fa6ae","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-06-23 15:26:20.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/9/check/dsvm-tempest-neutron-network/b9ff7b7 : SUCCESS in 1h 32m 02s","accounts_in_message":[],"_revision_number":9},{"id":"d32a9b52c36574d469495275681ac78c0e7d60e9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-23 15:31:55.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/9/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/e9fb8d6/ : SUCCESS in 1h 16m 45s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/9/check/gate-grenade-dsvm-neutron-ubuntu-xenial/a139b2a/ : SUCCESS in 37m 43s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/9/check/gate-tempest-dsvm-py35-ubuntu-xenial/de3279b/ : SUCCESS in 1h 07m 35s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/9/check/gate-tempest-docs-ubuntu-xenial/47c420a//doc/build/html/ : SUCCESS in 1m 31s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/9/check/gate-tempest-pep8-ubuntu-xenial/f2a6d63/ : SUCCESS in 1m 30s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/9/check/gate-tempest-python27-ubuntu-xenial/05a69b3/ : SUCCESS in 3m 24s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/9/check/gate-tempest-python35/378bb08/ : SUCCESS in 3m 13s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/9/check/gate-tempest-releasenotes/aa4dc58//releasenotes/build/html/ : SUCCESS in 3m 51s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/9/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/9b98ed4/ : SUCCESS in 41m 22s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/9/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/e7eb4dc/ : SUCCESS in 37m 21s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/9/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/e77c82b/ : SUCCESS in 1h 08m 57s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/9/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/fc44f9d/ : SUCCESS in 44m 18s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/9/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/bcc7e76/ : SUCCESS in 36m 19s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/9/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/e6c2147/ : SUCCESS in 47m 13s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/9/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/8388617/ : SUCCESS in 29m 07s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/9/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/1f990bf/ : SUCCESS in 1h 36m 46s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/9/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/49c44e4/ : SUCCESS in 1h 18m 49s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/9/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/245e56f/ : SUCCESS in 1h 09m 12s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/9/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/79f203b/ : FAILURE in 39m 29s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/9/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/418a4f2/ : SUCCESS in 3m 51s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/9/check/tempest-coverage-ubuntu-xenial-nv/c4a0b98/ : SUCCESS in 1m 26s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"559b597b2237854a1a9f73513424458348572bf8","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-06-26 07:54:07.000000000","message":"Patch Set 9:\n\n(1 comment)\n\noverall lgtm. one question inline","accounts_in_message":[],"_revision_number":9},{"id":"690490287055c92c7d618b7574d41e8d1c3932c1","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-26 08:36:30.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2666f8cd1f88bfe3c29d6f837a5a356a3dc0d428","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2017-06-26 11:40:42.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2a930fc71ea2ae17ad06756ed771df7b5fca3272","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-26 13:06:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"14b9f63f34f96946fca6b1e3a1a122c20112f900","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-26 13:07:19.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"06295149631ae76b01e38b16b21dc23e24ac5589","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-06-26 13:54:42.000000000","message":"Patch Set 10:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/10/check/dsvm-tempest-neutron-network/ea8abff : FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"a40347954f9c5bccfb5a4c630115d75d9de9f5aa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-26 14:47:58.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/10/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/143acfd/ : SUCCESS in 1h 35m 29s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/10/check/gate-grenade-dsvm-neutron-ubuntu-xenial/9dd8213/ : SUCCESS in 47m 39s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/10/check/gate-tempest-dsvm-py35-ubuntu-xenial/784d74b/ : SUCCESS in 1h 14m 33s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/10/check/gate-tempest-docs-ubuntu-xenial/ccd3cf4//doc/build/html/ : SUCCESS in 6m 18s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/10/check/gate-tempest-pep8-ubuntu-xenial/3a38955/ : SUCCESS in 2m 18s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/10/check/gate-tempest-python27-ubuntu-xenial/89b6bbb/ : FAILURE in 2m 51s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/10/check/gate-tempest-python35/efa372a/ : SUCCESS in 3m 45s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/10/check/gate-tempest-releasenotes/857b239//releasenotes/build/html/ : SUCCESS in 4m 17s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/10/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/5cfbfd3/ : SUCCESS in 54m 16s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/10/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/7adf209/ : SUCCESS in 46m 22s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/10/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/055ff21/ : SUCCESS in 1h 08m 53s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/10/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/1d5635a/ : SUCCESS in 54m 06s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/10/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/f1d8745/ : SUCCESS in 51m 46s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/10/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/4ad420f/ : FAILURE in 1h 05m 29s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/10/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/5707e09/ : SUCCESS in 39m 53s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/10/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/a12b5ac/ : SUCCESS in 1h 15m 07s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/10/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/672afeb/ : SUCCESS in 1h 23m 50s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/10/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/0d612a6/ : SUCCESS in 1h 09m 56s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/10/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/3fe17ad/ : FAILURE in 38m 17s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/10/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/29f9141/ : SUCCESS in 12m 03s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/10/check/tempest-coverage-ubuntu-xenial-nv/c9ae518/ : SUCCESS in 2m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ebe87bafc7053a537dba49a11e91bed342fbef0e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-29 08:27:27.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0a1d64c034c57155a997d1330697ef2a4e6cce4d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-29 08:31:30.000000000","message":"Patch Set 11:\n\nPlease merge this first https://review.openstack.org/#/c/477549/ then this.","accounts_in_message":[],"_revision_number":11},{"id":"6adf1e53f5f75fc33a22ec14e06fdf6b716787ea","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-29 10:22:25.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/11/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/6950632/ : SUCCESS in 1h 17m 08s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/11/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3fe5b00/ : SUCCESS in 39m 08s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/11/check/gate-tempest-dsvm-py35-ubuntu-xenial/3b1151f/ : SUCCESS in 1h 06m 21s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/11/check/gate-tempest-docs-ubuntu-xenial/0a3bdc6//doc/build/html/ : SUCCESS in 2m 03s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/11/check/gate-tempest-pep8-ubuntu-xenial/4c831ef/ : SUCCESS in 4m 22s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/11/check/gate-tempest-python27-ubuntu-xenial/d44269c/ : SUCCESS in 1m 33s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/11/check/gate-tempest-python35/f7e5f51/ : SUCCESS in 1m 44s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/11/check/gate-tempest-releasenotes/cc11178//releasenotes/build/html/ : SUCCESS in 2m 04s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/11/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/6c8b03a/ : SUCCESS in 42m 58s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/11/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/f0b12f4/ : SUCCESS in 38m 26s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/11/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/880c0c3/ : SUCCESS in 1h 20m 10s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/11/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/f847e06/ : SUCCESS in 40m 04s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/11/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/c8a4275/ : SUCCESS in 39m 08s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/11/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/4546e0b/ : SUCCESS in 46m 31s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/11/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/7289ed2/ : SUCCESS in 32m 29s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/11/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/ff5b0bc/ : SUCCESS in 1h 32m 59s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/11/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/259e1b8/ : SUCCESS in 1h 27m 41s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/11/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/9785433/ : SUCCESS in 1h 25m 02s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/11/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/b3fdc1f/ : SUCCESS in 1h 08m 33s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/11/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/3b88597/ : FAILURE in 1m 08s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/11/check/tempest-coverage-ubuntu-xenial-nv/7a73941/ : SUCCESS in 2m 18s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0d0a1b60bcdb4476b1475739a17afdac8872c777","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-06-29 10:58:00.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/11/check/dsvm-tempest-neutron-network/cd33645 : SUCCESS in 1h 29m 43s","accounts_in_message":[],"_revision_number":11},{"id":"91e45a2d169fd8c5d90109adf1fc19d95c9bc24c","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-06-30 03:31:46.000000000","message":"Patch Set 11:\n\nChandan Kumar wrote:\n\n\u003e Please merge this first https://review.openstack.org/#/c/477549/\n\u003e then this.\n\nyeah, this has a dependency. So, I recommend that you should put this on that patch.\nhttps://docs.openstack.org/infra/manual/developers.html#adding-a-dependency","accounts_in_message":[],"_revision_number":11},{"id":"fa70435e4d9c1d7eb307b4494ac64df664d8707d","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2017-06-30 06:31:50.000000000","message":"Patch Set 11: Code-Review-1\n\nplease quote (\")  the variables which contains dir/file names , because they might have space.","accounts_in_message":[],"_revision_number":11},{"id":"7b6ad86bbcf3b88394c738cfa4ac1bfe37a79a11","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-30 09:04:05.000000000","message":"Patch Set 11:\n\n\u003e please quote (\")  the variables which contains dir/file names ,\n \u003e because they might have space.\nSorry i did not get, you mean about putting test-requirements.txt with in quotes?","accounts_in_message":[],"_revision_number":11},{"id":"30e1eea8ca5cbad62c251b7f3113e06f1a2ccd99","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-30 09:04:45.000000000","message":"Uploaded patch set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"3aa84f7b90bc769d72825e6ffb75bd2c7a57c8f0","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-30 09:51:03.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8132b9a41521d746bc1e83f07ee6a2b276130d99","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-30 11:31:52.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/34d0962/ : SUCCESS in 1h 31m 11s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-grenade-dsvm-neutron-ubuntu-xenial/07a6c37/ : SUCCESS in 41m 54s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-py35-ubuntu-xenial/5b40f6a/ : SUCCESS in 1h 25m 03s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/13/check/gate-tempest-docs-ubuntu-xenial/6c05011//doc/build/html/ : SUCCESS in 2m 23s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-tempest-pep8-ubuntu-xenial/4e6111d/ : SUCCESS in 3m 20s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-tempest-python27-ubuntu-xenial/05e925f/ : SUCCESS in 3m 59s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/13/check/gate-tempest-python35/a80c521/ : SUCCESS in 2m 21s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/13/check/gate-tempest-releasenotes/9446b8a//releasenotes/build/html/ : SUCCESS in 2m 33s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/f4c2ce0/ : SUCCESS in 42m 51s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/6265bc2/ : SUCCESS in 37m 30s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/9d4d1c7/ : SUCCESS in 1h 07m 47s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/13/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/0eeb226/ : SUCCESS in 46m 48s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/13/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/b4d1523/ : SUCCESS in 54m 51s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/13/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/e1f641f/ : SUCCESS in 48m 30s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/13/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/94f0d8e/ : SUCCESS in 39m 54s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/230dafe/ : SUCCESS in 1h 33m 26s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/2e61b92/ : SUCCESS in 1h 39m 33s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/8b008e0/ : FAILURE in 1h 15m 21s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/f69b768/ : SUCCESS in 59m 37s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/13/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/c796a8b/ : FAILURE in 1m 03s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/13/check/tempest-coverage-ubuntu-xenial-nv/16aae2e/ : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"26d53771a39b91070040d85ade0378c093d2d413","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-06-30 11:34:35.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/13/check/dsvm-tempest-neutron-network/a598138 : SUCCESS in 1h 40m 28s","accounts_in_message":[],"_revision_number":13},{"id":"e5aa9902c1eb5c8e16805bca563e88394548dd65","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-06-30 11:35:46.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"222ce2c38ad94f9f68512850a0bb1357e8bd15c1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-30 13:11:32.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d5b9d6d/ : SUCCESS in 1h 23m 36s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-grenade-dsvm-neutron-ubuntu-xenial/af34f84/ : SUCCESS in 46m 56s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-py35-ubuntu-xenial/31e3aa1/ : SUCCESS in 1h 19m 03s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/13/check/gate-tempest-docs-ubuntu-xenial/97fa5f2//doc/build/html/ : SUCCESS in 3m 03s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-tempest-pep8-ubuntu-xenial/be128d8/ : SUCCESS in 2m 18s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-tempest-python27-ubuntu-xenial/71088a8/ : SUCCESS in 2m 56s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/13/check/gate-tempest-python35/e7add5a/ : SUCCESS in 1m 40s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/13/check/gate-tempest-releasenotes/39f24ee//releasenotes/build/html/ : SUCCESS in 2m 18s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/859df37/ : SUCCESS in 43m 49s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/2489a51/ : SUCCESS in 26m 00s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/3acb239/ : SUCCESS in 1h 09m 35s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/13/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/c85ff37/ : SUCCESS in 40m 36s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/13/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/348a0c8/ : SUCCESS in 54m 10s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/13/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/5df6789/ : SUCCESS in 45m 55s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/13/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/2351bfa/ : SUCCESS in 28m 05s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/07c512f/ : SUCCESS in 1h 28m 36s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/19090b6/ : SUCCESS in 1h 25m 13s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/4666ece/ : SUCCESS in 1h 19m 11s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/13/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/28d1340/ : SUCCESS in 55m 57s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/13/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/5b940d5/ : FAILURE in 30s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/13/check/tempest-coverage-ubuntu-xenial-nv/7f3bd2d/ : SUCCESS in 2m 55s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f31f1bf339a977349b3e99e68b820ab5ce76d31f","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-07-03 07:39:11.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)\n\nlgtm, but I think the \"Depends-on\" directive should be removed.","accounts_in_message":[],"_revision_number":13},{"id":"7f13d69cf5f351a3e641950488631483e7d38e6c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-03 07:45:27.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"4f47d38c6e0b536d7aa159acf057de0600d5201a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-03 07:45:50.000000000","message":"Patch Set 13:\n\n(1 comment)\n\nThanks for the pointer :-)","accounts_in_message":[],"_revision_number":13},{"id":"6bc962fdc1247d7e181d299a79601b92b3683551","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-03 09:20:11.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/14/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/3b46f89/ : SUCCESS in 1h 33m 42s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/14/check/gate-grenade-dsvm-neutron-ubuntu-xenial/20317cc/ : SUCCESS in 53m 54s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/14/check/gate-tempest-dsvm-py35-ubuntu-xenial/4862ca9/ : SUCCESS in 1h 22m 57s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/14/check/gate-tempest-docs-ubuntu-xenial/c0a91fd//doc/build/html/ : SUCCESS in 4m 46s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/14/check/gate-tempest-pep8-ubuntu-xenial/d2e1993/ : SUCCESS in 2m 43s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/14/check/gate-tempest-python27-ubuntu-xenial/e3e3974/ : SUCCESS in 1m 52s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/14/check/gate-tempest-python35/cc857d8/ : SUCCESS in 1m 53s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/14/check/gate-tempest-releasenotes/48c978c//releasenotes/build/html/ : SUCCESS in 3m 29s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/14/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/c7102b9/ : SUCCESS in 45m 24s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/14/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/72dc7c2/ : SUCCESS in 39m 36s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/14/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/ffaf492/ : SUCCESS in 1h 07m 35s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/14/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/7d49af6/ : FAILURE in 3m 04s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/14/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/1eb0530/ : FAILURE in 1m 45s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/14/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/55a8e51/ : FAILURE in 2m 01s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/14/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/4541848/ : FAILURE in 3m 21s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/14/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/165a226/ : SUCCESS in 1h 20m 22s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/14/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/119f321/ : SUCCESS in 1h 21m 11s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/14/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/475507b/ : SUCCESS in 1h 15m 32s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/14/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/fb0104e/ : SUCCESS in 53m 41s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/14/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/4a1e59a/ : FAILURE in 22s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/14/check/tempest-coverage-ubuntu-xenial-nv/a625e99/ : SUCCESS in 1m 57s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2b704a0c77cc5097e9b218fbfb876eb15ff5d5c5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-03 09:26:57.000000000","message":"Patch Set 14:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/14/check/dsvm-tempest-neutron-network/8eb35ab : FAILURE in 1h 41m 03s","accounts_in_message":[],"_revision_number":14},{"id":"c2ce19a7519b5da142f19b3a74a2189964996d06","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2017-07-03 20:41:03.000000000","message":"Patch Set 14:\n\n(1 comment)\n\nThe (L)expressions with contains SANITY_DIR needs be quoted.","accounts_in_message":[],"_revision_number":14},{"id":"35575cbed1218e5c7db7d62bf3be086a7650a787","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-07-04 03:34:46.000000000","message":"Patch Set 14:\n\nafazekas wrote:\n\n\u003e (1 comment)\n\u003e\n\u003e The (L)expressions with contains SANITY_DIR needs be quoted.\n\nAh, I see the point. Indeed.","accounts_in_message":[],"_revision_number":14},{"id":"be6980e3a053b7d6815c70acbd7eea8f31740fe4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-04 05:30:03.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d764663abd0c02c0f16e2a0457a1ec875450407a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-04 05:30:05.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"4d16f3c4fa9b748260885e127f069adc1f342bf7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-04 06:52:42.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/15/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8eb8636/ : SUCCESS in 1h 22m 16s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/15/check/gate-grenade-dsvm-neutron-ubuntu-xenial/04e4396/ : SUCCESS in 34m 27s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/15/check/gate-tempest-dsvm-py35-ubuntu-xenial/a22c39d/ : SUCCESS in 1h 06m 44s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/15/check/gate-tempest-docs-ubuntu-xenial/e3ed097//doc/build/html/ : SUCCESS in 12m 04s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/15/check/gate-tempest-pep8-ubuntu-xenial/04d620a/ : SUCCESS in 1m 50s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/15/check/gate-tempest-python27-ubuntu-xenial/1452205/ : SUCCESS in 2m 21s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/15/check/gate-tempest-python35/ac91208/ : SUCCESS in 1m 58s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/15/check/gate-tempest-releasenotes/294578c//releasenotes/build/html/ : SUCCESS in 2m 10s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/15/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/a347ed0/ : SUCCESS in 1h 01m 26s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/15/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/813cf8a/ : SUCCESS in 26m 09s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/15/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/f5ba89c/ : SUCCESS in 1h 10m 16s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/15/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/8bc7521/ : SUCCESS in 40m 04s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/15/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/6886199/ : SUCCESS in 37m 20s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/15/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/88dd19a/ : SUCCESS in 43m 25s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/15/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/ef9b760/ : SUCCESS in 27m 01s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/15/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/fc5cb63/ : SUCCESS in 1h 20m 37s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/15/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/384b8fa/ : SUCCESS in 1h 14m 14s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/15/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/f5adbe0/ : SUCCESS in 1h 07m 43s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/15/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/63a21a4/ : SUCCESS in 1h 03m 45s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/15/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/795d66d/ : FAILURE in 19s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/15/check/tempest-coverage-ubuntu-xenial-nv/88b3106/ : SUCCESS in 1m 09s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5a4fdb2f37c46d11465785454867827dd6fed6f4","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-04 07:30:34.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/15/check/dsvm-tempest-neutron-network/3be1d8d : SUCCESS in 1h 59m 55s","accounts_in_message":[],"_revision_number":15},{"id":"13141709734a82dba2d76750efa2a5b7ec14714f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-04 08:20:19.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"aa808a1624bdec760ffc05d821bd2c53b1fb506c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-04 09:01:11.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"887bcd025c0e175f39c5d12112b8330351f7cbc8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-04 10:37:45.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/17/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/45746a0/ : SUCCESS in 1h 21m 47s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/17/check/gate-grenade-dsvm-neutron-ubuntu-xenial/56c6f35/ : SUCCESS in 41m 15s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/17/check/gate-tempest-dsvm-py35-ubuntu-xenial/c2c80a8/ : SUCCESS in 1h 07m 14s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/17/check/gate-tempest-docs-ubuntu-xenial/d06779c//doc/build/html/ : SUCCESS in 13m 13s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/17/check/gate-tempest-pep8-ubuntu-xenial/38d48e0/ : SUCCESS in 2m 48s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/17/check/gate-tempest-python27-ubuntu-xenial/50ab15c/ : SUCCESS in 3m 24s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/17/check/gate-tempest-python35/9970827/ : SUCCESS in 2m 00s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/17/check/gate-tempest-releasenotes/4807a20//releasenotes/build/html/ : SUCCESS in 2m 59s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/17/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/588b385/ : SUCCESS in 1h 02m 47s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/17/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/125125c/ : SUCCESS in 43m 59s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/17/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/088230d/ : SUCCESS in 1h 09m 50s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/17/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/1f65c1a/ : SUCCESS in 50m 47s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/17/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/7f85183/ : SUCCESS in 38m 45s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/17/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/9b149c6/ : SUCCESS in 55m 49s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/17/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/6afdfb4/ : SUCCESS in 39m 24s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/17/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/cf78c30/ : SUCCESS in 1h 23m 46s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/17/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/c19531b/ : SUCCESS in 1h 23m 12s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/17/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/ae43a63/ : SUCCESS in 1h 14m 37s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/17/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/fcfce54/ : SUCCESS in 1h 24m 12s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/17/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/35f16cd/ : FAILURE in 1m 45s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/17/check/tempest-coverage-ubuntu-xenial-nv/74bc560/ : SUCCESS in 3m 11s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"55aeab3a391c1a6655bc8af9376a658276d5635d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-04 12:36:53.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/17/check/dsvm-tempest-neutron-network/09f6f5e : SUCCESS in 2h 57m 40s","accounts_in_message":[],"_revision_number":17},{"id":"194c398d839fab32196bb6003f33409a194f089f","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-07-05 02:04:22.000000000","message":"Patch Set 17:\n\n(1 comment)\n\nWhat do you think?","accounts_in_message":[],"_revision_number":17},{"id":"63c8eb816b5bfa8aa68f654480ca535c44ed8a1a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-05 05:54:01.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"398b3a715102d3deb9247126111f7684ea1f460b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-05 07:29:32.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/18/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/3c87f0e/ : SUCCESS in 1h 11m 06s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/18/check/gate-grenade-dsvm-neutron-ubuntu-xenial/e25a713/ : SUCCESS in 34m 20s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/18/check/gate-tempest-dsvm-py35-ubuntu-xenial/22d97dd/ : SUCCESS in 1h 05m 01s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/18/check/gate-tempest-docs-ubuntu-xenial/6156ac3//doc/build/html/ : SUCCESS in 1m 31s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/18/check/gate-tempest-pep8-ubuntu-xenial/0c91fc0/ : SUCCESS in 3m 52s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/18/check/gate-tempest-python27-ubuntu-xenial/31fb13a/ : SUCCESS in 1m 51s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/18/check/gate-tempest-python35/cfadd99/ : SUCCESS in 1m 58s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/18/check/gate-tempest-releasenotes/621c7c5//releasenotes/build/html/ : SUCCESS in 2m 03s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/18/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/e291cb4/ : SUCCESS in 41m 32s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/18/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/82c345e/ : SUCCESS in 28m 39s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/18/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/8097b8c/ : SUCCESS in 1h 10m 22s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/18/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/bdba9d1/ : SUCCESS in 41m 32s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/18/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/65b03c0/ : SUCCESS in 38m 27s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/18/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/7914c94/ : SUCCESS in 54m 11s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/18/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/8936c2c/ : SUCCESS in 30m 21s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/18/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/711d50e/ : SUCCESS in 1h 35m 11s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/18/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/0d45024/ : SUCCESS in 1h 15m 24s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/18/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/a087b4f/ : SUCCESS in 1h 10m 45s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/18/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/614626f/ : SUCCESS in 52m 43s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/18/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/fa706b6/ : SUCCESS in 1m 00s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/18/check/tempest-coverage-ubuntu-xenial-nv/a902a9a/ : SUCCESS in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"51b9eab472919f5566c761e1fbbe29d58a3bd96c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-05 07:31:32.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/18/check/dsvm-tempest-neutron-network/24908fd : SUCCESS in 1h 37m 06s","accounts_in_message":[],"_revision_number":18},{"id":"f1fb2bbfd07c2e746c919dd24dec4a52e3d25cf6","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-07-05 08:28:06.000000000","message":"Patch Set 18: Code-Review+2\n\nThanks :)","accounts_in_message":[],"_revision_number":18},{"id":"458fdb5012fff6582330e631d7331372b2f9ab38","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2017-07-06 11:22:47.000000000","message":"Patch Set 18: Code-Review-1\n\n(3 comments)\n\nThanks for the update!\nI\u0027m still -1 since the return code is not honoured","accounts_in_message":[],"_revision_number":18},{"id":"47ea5a5441f566d6308878ce01d47b9380837810","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-06 12:24:34.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"1aa07224bb94f5d36e9b85bf5146feab3d17c2d5","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-06 13:10:07.000000000","message":"Patch Set 18:\n\n(3 comments)\n\nThanks andreaf for the review. I have updated the comments.","accounts_in_message":[],"_revision_number":18},{"id":"fe8becc999a71543e1a341d25e7a5118b73fdf6c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-06 13:57:43.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/19/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/508bf85/ : SUCCESS in 1h 18m 36s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/19/check/gate-grenade-dsvm-neutron-ubuntu-xenial/1c7e41c/ : SUCCESS in 36m 49s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/19/check/gate-tempest-dsvm-py35-ubuntu-xenial/02c6fbf/ : SUCCESS in 1h 19m 35s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/19/check/gate-tempest-docs-ubuntu-xenial/ca64975//doc/build/html/ : SUCCESS in 14m 49s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/19/check/gate-tempest-pep8-ubuntu-xenial/c1fd57d/ : SUCCESS in 3m 22s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/19/check/gate-tempest-python27-ubuntu-xenial/a1f1dad/ : SUCCESS in 3m 14s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/19/check/gate-tempest-python35/8fa892f/ : SUCCESS in 3m 13s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/19/check/gate-tempest-releasenotes/a0ef19f//releasenotes/build/html/ : SUCCESS in 3m 54s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/19/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/51318b1/ : SUCCESS in 1h 01m 26s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/19/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/836ed5b/ : SUCCESS in 42m 47s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/19/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/0b09002/ : SUCCESS in 1h 03m 48s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/19/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/a6da960/ : SUCCESS in 37m 51s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/19/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/b0e2220/ : SUCCESS in 40m 08s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/19/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/9566f46/ : SUCCESS in 43m 21s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/19/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/9b0d7ec/ : SUCCESS in 38m 36s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/19/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/166ee55/ : SUCCESS in 1h 18m 43s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/19/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/bc4700d/ : SUCCESS in 1h 30m 01s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/19/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/db9b9c5/ : SUCCESS in 1h 05m 28s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/19/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/47283a2/ : SUCCESS in 1h 08m 10s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/19/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/d233f47/ : SUCCESS in 6m 09s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/19/check/tempest-coverage-ubuntu-xenial-nv/f3583d7/ : SUCCESS in 1m 34s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"931201fdd4d5e10ffa0fa48967740d6e03c20a7b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-06 15:50:18.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/19/check/dsvm-tempest-neutron-network/3db288b : SUCCESS in 1h 41m 26s","accounts_in_message":[],"_revision_number":19},{"id":"9a5c7d5f6687e94770319dbeb905c564d6c83d06","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-07-10 06:51:34.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"a60918677425ba74d67ab2d704655b0d0971bdb5","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2017-07-10 23:31:41.000000000","message":"Patch Set 19: Code-Review-1\n\n(2 comments)\n\nthe code contains a variable which is never red.","accounts_in_message":[],"_revision_number":19},{"id":"5a6034b26dff2a808868218214efdb6ec4aed663","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-11 02:47:57.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"ea22453303b86cc90240bb4e7ea14fc2f2e444a0","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-11 02:55:56.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"f9c74994726e37a8b360794a3bbc963bf583cb51","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-11 04:52:20.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/474164/20/check/dsvm-tempest-neutron-network/ee1f27e : SUCCESS in 1h 54m 44s","accounts_in_message":[],"_revision_number":20},{"id":"4fdd1114b24b65d1837501e3206d0d3f8804d6ff","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-11 05:49:37.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/20/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/7921432/ : SUCCESS in 1h 25m 51s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/20/check/gate-grenade-dsvm-neutron-ubuntu-xenial/a498474/ : SUCCESS in 45m 50s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/20/check/gate-tempest-dsvm-py35-ubuntu-xenial/35d2d23/ : SUCCESS in 1h 18m 06s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/20/check/gate-tempest-docs-ubuntu-xenial/03e4553//doc/build/html/ : SUCCESS in 6m 54s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/20/check/gate-tempest-pep8-ubuntu-xenial/746d30c/ : SUCCESS in 2m 51s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/20/check/gate-tempest-python27-ubuntu-xenial/8162bbf/ : SUCCESS in 2m 05s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/20/check/gate-tempest-python35/6a723ff/ : SUCCESS in 2m 03s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/20/check/gate-tempest-releasenotes/5e9b23d//releasenotes/build/html/ : SUCCESS in 2m 15s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/20/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/914e934/ : SUCCESS in 38m 08s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/64/474164/20/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/92c284e/ : SUCCESS in 28m 52s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/64/474164/20/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7c64f50/ : SUCCESS in 1h 07m 58s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/64/474164/20/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/0e2c0af/ : SUCCESS in 39m 07s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/64/474164/20/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/dae7be3/ : SUCCESS in 50m 44s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/64/474164/20/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/0864562/ : SUCCESS in 47m 02s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/64/474164/20/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/3700924/ : SUCCESS in 29m 57s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/64/474164/20/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/2e26a8e/ : SUCCESS in 1h 26m 44s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/64/474164/20/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/48cb121/ : SUCCESS in 1h 25m 25s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/64/474164/20/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/574c7e4/ : SUCCESS in 1h 27m 21s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/64/474164/20/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/8690771/ : SUCCESS in 1h 06m 58s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/64/474164/20/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/2125eba/ : SUCCESS in 4m 31s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/64/474164/20/check/tempest-coverage-ubuntu-xenial-nv/d1eb81d/ : SUCCESS in 2m 15s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"568a7ad9c0dee68a6ff29f428b04def2013fad51","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2017-07-11 06:57:29.000000000","message":"Patch Set 20: Code-Review+2\n\ngood enough to start with","accounts_in_message":[],"_revision_number":20},{"id":"e4f49fdea5d3c1219db4ba4bab141935bb79edcf","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-07-11 10:00:39.000000000","message":"Patch Set 20: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"9e4e9a2e6ae8e5119d5a82205529996f5f98249b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-11 10:00:50.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":20},{"id":"503060b7396770014f0efd6cb0e25cf739f66847","date":"2017-07-11 13:24:49.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":20},{"id":"1fc63c4896239bcc04bc38c3eb4158980f19ad78","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-11 13:24:49.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/64/474164/20/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/bd463f2/ : SUCCESS in 1h 23m 47s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/64/474164/20/gate/gate-grenade-dsvm-neutron-ubuntu-xenial/2ef693f/ : SUCCESS in 52m 38s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/64/474164/20/gate/gate-tempest-dsvm-py35-ubuntu-xenial/34127dc/ : SUCCESS in 1h 20m 42s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/64/474164/20/gate/gate-tempest-docs-ubuntu-xenial/9f47668//doc/build/html/ : SUCCESS in 4m 24s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/64/474164/20/gate/gate-tempest-pep8-ubuntu-xenial/0740f1e/ : SUCCESS in 4m 23s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/64/474164/20/gate/gate-tempest-python27-ubuntu-xenial/23b0966/ : SUCCESS in 2m 51s\n- gate-tempest-python35 http://logs.openstack.org/64/474164/20/gate/gate-tempest-python35/b925453/ : SUCCESS in 2m 46s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/64/474164/20/gate/gate-tempest-releasenotes/2b3e4b6//releasenotes/build/html/ : SUCCESS in 3m 42s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/64/474164/20/gate/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/8d87fa8/ : SUCCESS in 42m 16s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"b7affe4c0339083d3870ad498143d1b1fbdd1c1b","revisions":{"521843b6e35e3e760381a6c5311923622cd80d00":{"kind":"REWORK","_number":1,"created":"2017-06-14 11:41:39.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/1"}}},"commit":{"parents":[{"commit":"5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0","subject":"Merge \"Prevent error in _parse_resp when nullable list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:40:31.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:40:31.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/521843b6e35e3e760381a6c5311923622cd80d00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/521843b6e35e3e760381a6c5311923622cd80d00"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"dcfb99953f8dbf916a38f1392634b9800c6fcd05":{"kind":"REWORK","_number":2,"created":"2017-06-14 11:46:22.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/2"}}},"commit":{"parents":[{"commit":"5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0","subject":"Merge \"Prevent error in _parse_resp when nullable list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:40:31.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:46:00.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dcfb99953f8dbf916a38f1392634b9800c6fcd05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dcfb99953f8dbf916a38f1392634b9800c6fcd05"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ff849dbbdbfa66f2d3e3079b003f17e5aa23e92a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2017-06-14 11:58:26.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/3"}}},"commit":{"parents":[{"commit":"5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0","subject":"Merge \"Prevent error in _parse_resp when nullable list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:40:31.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:58:23.000000000","tz":0},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nDepends-on: I86a534043d2c0ba2d26ef14b75ac06659a017f2d\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ff849dbbdbfa66f2d3e3079b003f17e5aa23e92a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ff849dbbdbfa66f2d3e3079b003f17e5aa23e92a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"06e04f895b0c47ac95dbcf89041e8a82756493e7":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2017-06-14 13:37:46.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/4"}}},"commit":{"parents":[{"commit":"5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0","subject":"Merge \"Prevent error in _parse_resp when nullable list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:40:31.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 13:37:42.000000000","tz":0},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/06e04f895b0c47ac95dbcf89041e8a82756493e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/06e04f895b0c47ac95dbcf89041e8a82756493e7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"201907bb5be6f7f032b79f7db36b06a89d87771f":{"kind":"REWORK","_number":5,"created":"2017-06-14 15:00:04.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/5"}}},"commit":{"parents":[{"commit":"5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0","subject":"Merge \"Prevent error in _parse_resp when nullable list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:40:31.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 14:59:39.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/201907bb5be6f7f032b79f7db36b06a89d87771f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/201907bb5be6f7f032b79f7db36b06a89d87771f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5aad8e753a31cc243e6bcfe1cc5dc5c04812ecb0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c3f711fc636534985748eec180a31a9d9626ad7b":{"kind":"REWORK","_number":6,"created":"2017-06-15 08:03:10.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/6"}}},"commit":{"parents":[{"commit":"1f170f6a2290043b850002a9262b0f0d3ae6fdc5","subject":"Merge \"Use urllib2 instead of requests in tempest generate plugin list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1f170f6a2290043b850002a9262b0f0d3ae6fdc5"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:40:31.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-15 08:01:13.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or tox -e sanity will\ninvoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c3f711fc636534985748eec180a31a9d9626ad7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c3f711fc636534985748eec180a31a9d9626ad7b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1f170f6a2290043b850002a9262b0f0d3ae6fdc5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"08bc5e565bdc5664befa6a5ceedb4f2f4a20c49a":{"kind":"REWORK","_number":7,"created":"2017-06-20 18:29:44.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/7"}}},"commit":{"parents":[{"commit":"7172548f3041a2ce977c011a7b28391bfa19b887","subject":"Merge \"Test coverage for network v2 security group rules client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7172548f3041a2ce977c011a7b28391bfa19b887"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:40:31.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-20 18:29:19.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/08bc5e565bdc5664befa6a5ceedb4f2f4a20c49a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/08bc5e565bdc5664befa6a5ceedb4f2f4a20c49a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7172548f3041a2ce977c011a7b28391bfa19b887","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d3f2edddcda10ea28afbe81cd5836b119423e9ed":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2017-06-23 05:33:11.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/8"}}},"commit":{"parents":[{"commit":"7226611b00ef4fca8e74df3c697929eb167ba49c","subject":"Merge \"Fixed project lists for retrieving tempest plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7226611b00ef4fca8e74df3c697929eb167ba49c"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:40:31.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-23 05:32:58.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d3f2edddcda10ea28afbe81cd5836b119423e9ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d3f2edddcda10ea28afbe81cd5836b119423e9ed"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7226611b00ef4fca8e74df3c697929eb167ba49c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"602b15a484ffe1f2fad59f2012290ebe8071472b":{"kind":"REWORK","_number":9,"created":"2017-06-23 13:53:45.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/9"}}},"commit":{"parents":[{"commit":"7226611b00ef4fca8e74df3c697929eb167ba49c","subject":"Merge \"Fixed project lists for retrieving tempest plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7226611b00ef4fca8e74df3c697929eb167ba49c"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:40:31.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-23 13:47:47.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/602b15a484ffe1f2fad59f2012290ebe8071472b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/602b15a484ffe1f2fad59f2012290ebe8071472b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7226611b00ef4fca8e74df3c697929eb167ba49c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e3b7c45a532d5f79d058d912c9475f44d25e735a":{"kind":"REWORK","_number":10,"created":"2017-06-26 13:06:58.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/10"}}},"commit":{"parents":[{"commit":"9ef109d9998a6c4db189e44ac640872acac54ba7","subject":"Merge \"Add api-ref urls for some volume v2 APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9ef109d9998a6c4db189e44ac640872acac54ba7"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-14 11:40:31.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-26 13:06:40.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e3b7c45a532d5f79d058d912c9475f44d25e735a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e3b7c45a532d5f79d058d912c9475f44d25e735a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9ef109d9998a6c4db189e44ac640872acac54ba7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"260d29329c9d442ec45215438559d9923a241229":{"kind":"REWORK","_number":11,"created":"2017-06-29 08:27:27.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/11"}}},"commit":{"parents":[{"commit":"1b786f1c6d4ed605be0f09647b44e97f8bd40664","subject":"Merge \"Fix no attribute \u0027urlopen\u0027 error in python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1b786f1c6d4ed605be0f09647b44e97f8bd40664"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-29 08:27:01.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-29 08:27:24.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/260d29329c9d442ec45215438559d9923a241229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/260d29329c9d442ec45215438559d9923a241229"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1b786f1c6d4ed605be0f09647b44e97f8bd40664","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"fb70b3933073ebf59c1b3beaedbf309c545c69bb":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2017-06-30 09:04:45.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/12"}}},"commit":{"parents":[{"commit":"1b786f1c6d4ed605be0f09647b44e97f8bd40664","subject":"Merge \"Fix no attribute \u0027urlopen\u0027 error in python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1b786f1c6d4ed605be0f09647b44e97f8bd40664"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-29 08:27:01.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-30 09:02:40.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nDepends-on: I4528a23ca4933fdb7a3168f8dc99bbf0497ae5cc\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fb70b3933073ebf59c1b3beaedbf309c545c69bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fb70b3933073ebf59c1b3beaedbf309c545c69bb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1b786f1c6d4ed605be0f09647b44e97f8bd40664","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ff02f52171e01a3febbe2aba48f6e439bb089f54":{"kind":"REWORK","_number":13,"created":"2017-06-30 09:51:03.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/13"}}},"commit":{"parents":[{"commit":"1b786f1c6d4ed605be0f09647b44e97f8bd40664","subject":"Merge \"Fix no attribute \u0027urlopen\u0027 error in python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1b786f1c6d4ed605be0f09647b44e97f8bd40664"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-29 08:27:01.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-30 09:50:49.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nDepends-on: I4528a23ca4933fdb7a3168f8dc99bbf0497ae5cc\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ff02f52171e01a3febbe2aba48f6e439bb089f54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ff02f52171e01a3febbe2aba48f6e439bb089f54"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1b786f1c6d4ed605be0f09647b44e97f8bd40664","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"bba42762687c1899ec926f4ca83a83269224989b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":14,"created":"2017-07-03 07:45:27.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/14"}}},"commit":{"parents":[{"commit":"1a3b3dc50b418d3a15839830d7d1ff88c8c76cff","subject":"Merge \"Fix image deletion checks after unshelve server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1a3b3dc50b418d3a15839830d7d1ff88c8c76cff"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-29 08:27:01.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-03 07:45:12.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bba42762687c1899ec926f4ca83a83269224989b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bba42762687c1899ec926f4ca83a83269224989b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1a3b3dc50b418d3a15839830d7d1ff88c8c76cff","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f278a0596b2317a8a682832020a7532106256545":{"kind":"REWORK","_number":15,"created":"2017-07-04 05:30:05.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/15"}}},"commit":{"parents":[{"commit":"479863dd47beeb35aef5fd5541956658940c48cc","subject":"Merge \"Add response assertions and remove unused variables in volume tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/479863dd47beeb35aef5fd5541956658940c48cc"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-29 08:27:01.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-04 05:29:50.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f278a0596b2317a8a682832020a7532106256545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f278a0596b2317a8a682832020a7532106256545"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"479863dd47beeb35aef5fd5541956658940c48cc","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"37d0b321df7468819e500ea14cc21473d6f23600":{"kind":"REWORK","_number":16,"created":"2017-07-04 08:20:19.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/16"}}},"commit":{"parents":[{"commit":"479863dd47beeb35aef5fd5541956658940c48cc","subject":"Merge \"Add response assertions and remove unused variables in volume tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/479863dd47beeb35aef5fd5541956658940c48cc"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-29 08:27:01.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-04 08:20:04.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/37d0b321df7468819e500ea14cc21473d6f23600"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/37d0b321df7468819e500ea14cc21473d6f23600"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"479863dd47beeb35aef5fd5541956658940c48cc","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f1e7c7b306652b03f5f1e3e2c65244bf2a9d2020":{"kind":"REWORK","_number":17,"created":"2017-07-04 09:01:11.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/17"}}},"commit":{"parents":[{"commit":"479863dd47beeb35aef5fd5541956658940c48cc","subject":"Merge \"Add response assertions and remove unused variables in volume tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/479863dd47beeb35aef5fd5541956658940c48cc"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-29 08:27:01.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-04 09:00:41.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f1e7c7b306652b03f5f1e3e2c65244bf2a9d2020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f1e7c7b306652b03f5f1e3e2c65244bf2a9d2020"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"479863dd47beeb35aef5fd5541956658940c48cc","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b09c556d233116145e9071f0eef85b03d3d662e3":{"kind":"REWORK","_number":18,"created":"2017-07-05 05:54:01.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/18"}}},"commit":{"parents":[{"commit":"c7c7916bc101f1437871e5ca4b25ca53769335f7","subject":"Merge \"Switch from oslosphinx to openstackdocstheme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c7c7916bc101f1437871e5ca4b25ca53769335f7"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-29 08:27:01.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-05 05:53:41.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b09c556d233116145e9071f0eef85b03d3d662e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b09c556d233116145e9071f0eef85b03d3d662e3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c7c7916bc101f1437871e5ca4b25ca53769335f7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8f5e9aca2f635302ee0f3149316bfc9b34d29d55":{"kind":"REWORK","_number":19,"created":"2017-07-06 12:24:34.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/19"}}},"commit":{"parents":[{"commit":"8390829df224db63edca83da6c1c90ae305e3bf0","subject":"Merge \"Update volume-status waiter for new cinder attach\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8390829df224db63edca83da6c1c90ae305e3bf0"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-29 08:27:01.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-06 12:23:05.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8f5e9aca2f635302ee0f3149316bfc9b34d29d55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8f5e9aca2f635302ee0f3149316bfc9b34d29d55"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8390829df224db63edca83da6c1c90ae305e3bf0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b7affe4c0339083d3870ad498143d1b1fbdd1c1b":{"kind":"REWORK","_number":20,"created":"2017-07-11 02:47:57.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/474164/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/64/474164/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/64/474164/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/64/474164/20"}}},"commit":{"parents":[{"commit":"8390829df224db63edca83da6c1c90ae305e3bf0","subject":"Merge \"Update volume-status waiter for new cinder attach\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8390829df224db63edca83da6c1c90ae305e3bf0"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-06-29 08:27:01.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-11 02:47:04.000000000","tz":330},"subject":"Added script for doing tempest plugin sanity","message":"Added script for doing tempest plugin sanity\n\nIt will help to catch tempest plugin config and dependency issues\nwith tempest.\nsh ./tools/tools/tempest-plugin-sanity.sh or\ntox -e plugin-sanity-check will invoke the same.\n\nChange-Id: If42561bfc1d6dcffbb6537634d7769ef90fab21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b7affe4c0339083d3870ad498143d1b1fbdd1c1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b7affe4c0339083d3870ad498143d1b1fbdd1c1b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8390829df224db63edca83da6c1c90ae305e3bf0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
