)]}'
{"id":"openstack%2Ftempest~480429","triplet_id":"openstack%2Ftempest~master~Ia82d76d2ec45ba4967bc11e7524993d2bfd652c1","project":"openstack/tempest","branch":"master","topic":"ping_test","hashtags":[],"change_id":"Ia82d76d2ec45ba4967bc11e7524993d2bfd652c1","subject":"Added test_boot_volume_network_connectivity scenario test","status":"ABANDONED","created":"2017-07-05 08:37:53.000000000","updated":"2017-07-28 15:34:13.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b494ed13477f48c46e4454f86709724d75f0af5f","_number":480429,"virtual_id_number":480429,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":1,"date":"2017-07-06 07:36:05.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2017-07-13 14:16:52.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"date":"2017-07-06 09:23:34.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"date":"2017-07-07 08:44:53.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":20734,"name":"Prateek Arora","email":"1192prateek@gmail.com","username":"prateek_a"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"recommended":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":-1,"date":"2017-07-13 07:58:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":-1,"date":"2017-07-13 08:26:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2017-07-06 11:25:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":1,"date":"2017-07-07 13:35:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2017-07-10 05:22:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2017-07-07 10:48:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20734,"name":"Prateek Arora","email":"1192prateek@gmail.com","username":"prateek_a"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2017-07-10 14:06:37.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":20734,"name":"Prateek Arora","email":"1192prateek@gmail.com","username":"prateek_a"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":20734,"name":"Prateek Arora","email":"1192prateek@gmail.com","username":"prateek_a"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":20734,"name":"Prateek Arora","email":"1192prateek@gmail.com","username":"prateek_a"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-05 08:38:43.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2017-07-05 08:39:07.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2017-07-05 09:52:17.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2017-07-05 09:52:25.000000000","updated_by":{"_account_id":20734,"name":"Prateek Arora","email":"1192prateek@gmail.com","username":"prateek_a"},"reviewer":{"_account_id":20734,"name":"Prateek Arora","email":"1192prateek@gmail.com","username":"prateek_a"},"state":"REVIEWER"},{"updated":"2017-07-06 07:36:05.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-07-06 09:23:34.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-07-06 11:25:46.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2017-07-07 08:44:53.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2017-07-07 10:48:09.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2017-07-07 13:35:41.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-07-10 05:22:02.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2017-07-10 14:06:37.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2017-07-13 07:58:30.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2017-07-13 08:26:10.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"}],"messages":[{"id":"b16f704b5680f8e4ebc38ac65912b40aee9e0940","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-05 08:37:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bf6dd7dec5fce8066bdf97f7a26c51618d67f93","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2017-07-05 08:45:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4291f4e5a79f8765837eef7a89d2f7d11c2f9459","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2017-07-05 08:51:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c776b22ba4b25d843fcf38e42ed610500dcfe605","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-05 10:10:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/29/480429/1/check/dsvm-tempest-neutron-network/b42416b : SUCCESS in 1h 32m 03s","accounts_in_message":[],"_revision_number":1},{"id":"15d403c59923bb14a0190dfd75ef528a00d673b7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-05 10:44:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/29/480429/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/76071e1/ : SUCCESS in 1h 15m 31s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/29/480429/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/b06cb40/ : SUCCESS in 1h 17m 23s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/29/480429/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/85ee977/ : SUCCESS in 1h 11m 16s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/29/480429/1/check/gate-tempest-docs-ubuntu-xenial/995e024//doc/build/html/ : SUCCESS in 2m 02s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/29/480429/1/check/gate-tempest-pep8-ubuntu-xenial/9ded8cf/ : SUCCESS in 6m 53s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/29/480429/1/check/gate-tempest-python27-ubuntu-xenial/d467a2f/ : SUCCESS in 2m 12s\n- gate-tempest-python35 http://logs.openstack.org/29/480429/1/check/gate-tempest-python35/5a54d82/ : SUCCESS in 2m 02s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/29/480429/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/e884782/ : SUCCESS in 45m 12s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/29/480429/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/c6fe966/ : SUCCESS in 46m 04s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/29/480429/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/a81e6e7/ : SUCCESS in 1h 05m 31s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/29/480429/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/47c1126/ : SUCCESS in 47m 52s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/29/480429/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/bd1e0fc/ : SUCCESS in 58m 15s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/29/480429/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/c1fb034/ : SUCCESS in 47m 26s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/29/480429/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/bf9462f/ : SUCCESS in 44m 02s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/29/480429/1/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/c446a39/ : SUCCESS in 1h 45m 23s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/29/480429/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/0fd774d/ : SUCCESS in 2h 00m 38s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/29/480429/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/bc8593d/ : SUCCESS in 1h 22m 55s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/29/480429/1/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/3361e4f/ : SUCCESS in 1h 25m 09s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/29/480429/1/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/c78bcbf/ : SUCCESS in 2m 02s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/29/480429/1/check/tempest-coverage-ubuntu-xenial-nv/0df7cc1/ : SUCCESS in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f37358404b13aeeac6bb42fc378b3af4ca2ddbb4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-07-05 16:08:57.000000000","message":"Patch Set 1: Code-Review-1\n\nwhile the code is certainly good, I don\u0027t see any CI job actually running the test. Have we tried to run it in some jobs or CI?","accounts_in_message":[],"_revision_number":1},{"id":"4d281bf71cd0e1156697de1dd9d1abd6b8f84f24","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-05 16:31:13.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"0bbea6c48e33b0c1b9731ba241c243e6a92cdef5","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-05 16:36:26.000000000","message":"Patch Set 1:\n\n\u003e while the code is certainly good, I don\u0027t see any CI job actually\n \u003e running the test. Have we tried to run it in some jobs or CI?\n\nThis job is running this test gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv go to http://logs.openstack.org/29/480429/1/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/3361e4f/logs/testr_results.html.gz and click on tempest.scenario.test_volume_boot_pattern.TestVolumeBootPattern here we can find test_boot_volume_network_connectivity is passed.","accounts_in_message":[],"_revision_number":1},{"id":"30a6279487fb49658022d6acd6ff09fe304ddaed","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-07-05 16:36:37.000000000","message":"Patch Set 1: Code-Review+1\n\nnevermind, the test is running fine in http://logs.openstack.org/29/480429/1/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/3361e4f/logs/testr_results.html.gz","accounts_in_message":[],"_revision_number":1},{"id":"af079eb3e4cd9e9abd09178eba076e98d519da38","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-05 18:29:53.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- gate-tempest-dsvm-neutron-pg-full-ubuntu-xenial http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-neutron-pg-full-ubuntu-xenial/8f9340c/ : FAILURE in 58m 42s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/29/480429/1/experimental/gate-grenade-dsvm-neutron-ubuntu-xenial/c09348e/ : SUCCESS in 38m 25s\n- periodic-tempest-dsvm-all-master http://logs.openstack.org/29/480429/1/experimental/periodic-tempest-dsvm-all-master/e13a269/ : SUCCESS in 1h 34m 46s\n- gate-tempest-dsvm-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-multinode-full-ubuntu-xenial-nv/0dc07db/ : SUCCESS in 1h 53m 19s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/b2b233b/ : SUCCESS in 1h 16m 41s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-neutron-dvr-multinode-full-ubuntu-xenial-nv/4382add/ : SUCCESS in 1h 20m 06s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv/a548953/ : SUCCESS in 1h 27m 50s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/ab8583b/ : SUCCESS in 22m 53s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-cells-ubuntu-xenial/51c418f/ : SUCCESS in 42m 25s\n- gate-tempest-dsvm-full-test-accounts-ubuntu-xenial http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-full-test-accounts-ubuntu-xenial/a3b1e05/ : SUCCESS in 1h 27m 22s\n- gate-tempest-dsvm-neutron-full-test-accounts-ubuntu-xenial http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-neutron-full-test-accounts-ubuntu-xenial/afe6bb0/ : SUCCESS in 1h 38m 20s\n- gate-tempest-dsvm-identity-v3-test-accounts-ubuntu-xenial http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-identity-v3-test-accounts-ubuntu-xenial/6814ffb/ : FAILURE in 1h 13m 18s\n- gate-tempest-dsvm-neutron-full-non-admin-ubuntu-xenial http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-neutron-full-non-admin-ubuntu-xenial/b092fa0/ : SUCCESS in 1h 40m 30s\n- gate-tempest-dsvm-nova-v20-api-ubuntu-xenial http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-nova-v20-api-ubuntu-xenial/fd293b8/ : FAILURE in 39m 43s\n- gate-tempest-dsvm-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-lvm-multibackend-ubuntu-xenial-nv/78b7f06/ : SUCCESS in 1h 11m 59s (non-voting)\n- gate-tempest-dsvm-cinder-v1-ubuntu-xenial http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-cinder-v1-ubuntu-xenial/0f631e9/ : SUCCESS in 37m 09s\n- gate-tempest-dsvm-py35-full-devstack-plugin-ceph-ubuntu-xenial-nv http://logs.openstack.org/29/480429/1/experimental/gate-tempest-dsvm-py35-full-devstack-plugin-ceph-ubuntu-xenial-nv/b413acf/ : FAILURE in 17m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4024ec9ccaf75618b87c748b506d4c19353bcba9","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-07-06 02:39:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)\n\nI\u0027m actually unsure that there\u0027s a difference between \"boot from image\" and \"boot from volume\" regarding to floating IP operations. But I think some clouds have only \"boot from volume\" instances, so, this is a good test case for it.\n\nAnd I think we probably can merge this test into an existing scenario test like test_volume_boot_pattern() but, I\u0027m OK to have this separately because they are very different so far.","accounts_in_message":[],"_revision_number":1},{"id":"881ed73fedaafec8def92e9e657d8eca670c573d","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2017-07-06 03:19:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nsome nits","accounts_in_message":[],"_revision_number":1},{"id":"44d299ccccd500612b21319d08cd312b8f398faf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-06 04:49:26.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"7455f76347b6c5a3476b72e2e8a56cc99ce3be8b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-06 06:09:24.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3e85e2d060f7af376cb560dd3b2838c4fb597d4c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-06 06:09:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"80a11a87b65e2058090bde32a8fddca99855f06f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-06 07:36:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/29/480429/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/32cceb1/ : SUCCESS in 1h 17m 15s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/29/480429/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/05644a8/ : SUCCESS in 40m 47s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/29/480429/2/check/gate-tempest-dsvm-py35-ubuntu-xenial/7828b1c/ : SUCCESS in 1h 14m 46s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/29/480429/2/check/gate-tempest-docs-ubuntu-xenial/76c68f2//doc/build/html/ : SUCCESS in 2m 26s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/29/480429/2/check/gate-tempest-pep8-ubuntu-xenial/d9ea94f/ : SUCCESS in 3m 36s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/29/480429/2/check/gate-tempest-python27-ubuntu-xenial/8313968/ : SUCCESS in 1m 09s\n- gate-tempest-python35 http://logs.openstack.org/29/480429/2/check/gate-tempest-python35/258ff6f/ : SUCCESS in 1m 07s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/29/480429/2/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/75d52b2/ : SUCCESS in 46m 59s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/29/480429/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/b40b882/ : SUCCESS in 45m 32s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/29/480429/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/dd3c7ef/ : FAILURE in 1h 14m 07s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/29/480429/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/9f94871/ : SUCCESS in 52m 17s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/29/480429/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/1245df6/ : SUCCESS in 41m 36s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/29/480429/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/5808fe8/ : SUCCESS in 48m 29s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/29/480429/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/a421b64/ : SUCCESS in 29m 21s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/29/480429/2/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/50753c9/ : SUCCESS in 1h 25m 32s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/29/480429/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/04ed795/ : SUCCESS in 1h 19m 47s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/29/480429/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/6e6e688/ : SUCCESS in 1h 05m 27s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/29/480429/2/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/98dcb28/ : SUCCESS in 1h 09m 16s (non-voting)\n- gate-tempest-plugin-sanity-ubuntu-xenial-nv http://logs.openstack.org/29/480429/2/check/gate-tempest-plugin-sanity-ubuntu-xenial-nv/31793b1/ : SUCCESS in 6m 05s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/29/480429/2/check/tempest-coverage-ubuntu-xenial-nv/3e3c207/ : SUCCESS in 2m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8852f64b273d4edc1cdff2f23931e6ec0415ebbe","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-06 09:23:34.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/29/480429/2/check/dsvm-tempest-neutron-network/7fbcc43 : FAILURE in 2h 11m 17s","accounts_in_message":[],"_revision_number":2},{"id":"987eb6ddcabdcb9f1ed416911012786cecce6477","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2017-07-06 11:25:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"317df7b0f526506d6ceb8ff6ce62cbc312d1d159","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2017-07-07 08:44:53.000000000","message":"Patch Set 2:\n\nthe code itself is ok for me, and I am just hesitated whether to add\na testcase, or to add the connection check into an existing one","accounts_in_message":[],"_revision_number":2},{"id":"13ce32f9a99cc35ef2f7aec56030b1e6e2b37dfa","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2017-07-07 08:48:19.000000000","message":"Patch Set 2:\n\n@zhufl I\u0027m okay with this test, and also notice that this test in particular, boot a vm from a volume, and not from a image, which other tests don\u0027t do.","accounts_in_message":[],"_revision_number":2},{"id":"2ca57d082a5f41b6d2da562915622aa959ac8975","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2017-07-07 10:48:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"79db41db55f55ecda900ed47859bbc27c251aee5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-07-07 13:35:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1f0c49a30cc0e7f76744fd0505084f7507aa0557","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-07-10 05:22:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"729166e2aafb5eb506fd58d73cbb010e159d17e9","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2017-07-13 07:58:30.000000000","message":"Patch Set 2: Code-Review-1\n\nThe test_volume_boot_pattern already checks everything that is done by this test, so I would prefer not to add a new test.\nIf you need a more stripped down test for TripleO CI it can be maintained elsewhere - but adding the test here would end up in running the same check twice in the gate. Which is not bad per se, but since we have limited time and resources in the gate we should avoid it.","accounts_in_message":[],"_revision_number":2},{"id":"93b8724c7546c4c0132a78142880dbcdc716652b","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2017-07-13 08:00:16.000000000","message":"Patch Set 2:\n\n@Andrea, this tests takes 2 minutes to run, I don\u0027t think it will add so much...","accounts_in_message":[],"_revision_number":2},{"id":"1fd3b0673116d6214e8533c419dbd7540699cbb4","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2017-07-13 08:26:10.000000000","message":"Patch Set 2: Code-Review-1\n\nEvery new scenario has to cover something what otherwise not covered, I do not see what is the new thing here.","accounts_in_message":[],"_revision_number":2},{"id":"51efc30cc5d7b1db9101aeeeb215f76c8c2a7f74","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-13 14:16:52.000000000","message":"Patch Set 2:\n\n@andreaf, @afazekas, we wanted a boot from volume test which takes less time then current test_boot_volume_pattern. For this single test, we cannot put a tempest plugin for the same. What are other alternatives here?","accounts_in_message":[],"_revision_number":2},{"id":"b0632b2b23fe1cee570f997dc7026423cd9f2029","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2017-07-17 12:09:57.000000000","message":"Patch Set 2:\n\nYou can consider to run  \ntest_create_server_from_volume_snapshot together in parallel with  one of the test cases from test_network_basic_ops . Which expected to have similar time and coverage.\n\nBTW, test_create_server_from_volume_snapshot does not verifies the vm liveness (or something disk related not just by api) over ssh, do not be surprised it will be added in the future.\n\nYou can check the timing of this test, maybe this what you really want. \nFYI: Many backed service is able to create a snapshot in fraction of a second.","accounts_in_message":[],"_revision_number":2},{"id":"fa1545ccf74ddce75ae4e0ed1266e068048bf8ba","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-28 15:34:13.000000000","message":"Abandoned\n\nAbandoning as it is no longer needed.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1b474c53106f7a56fca85ce5688b3253718d3c8e","revisions":{"1182d991b56c741654f489a5bf4f9c0433b3b90a":{"kind":"REWORK","_number":1,"created":"2017-07-05 08:37:53.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/29/480429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/29/480429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/480429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/480429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/480429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/29/480429/1"}}},"commit":{"parents":[{"commit":"5db2ee8e37f2cfcac06354101edbf08ea2e7224c","subject":"Merge \"Fix check_service_client_function doc typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5db2ee8e37f2cfcac06354101edbf08ea2e7224c"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-05 08:22:41.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-05 08:37:38.000000000","tz":330},"subject":"Added test_boot_volume_network_connectivity scenario test","message":"Added test_boot_volume_network_connectivity scenario test\n\n* This scenario test will test the floating ip connectivity for\n  an instance booted from cinder volume\n* Below is the list of operation it does:\n - Create in Cinder some bootable volume importing a Glance image\n - Create keypair and security group\n - Boot an instance from the bootable volume\n - Attach floating IP to it and check network connectivity\n - Delete the instance.\n\nChange-Id: Ia82d76d2ec45ba4967bc11e7524993d2bfd652c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1182d991b56c741654f489a5bf4f9c0433b3b90a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1182d991b56c741654f489a5bf4f9c0433b3b90a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5db2ee8e37f2cfcac06354101edbf08ea2e7224c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1b474c53106f7a56fca85ce5688b3253718d3c8e":{"kind":"REWORK","_number":2,"created":"2017-07-06 06:09:26.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/29/480429/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/29/480429/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/480429/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/480429/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/480429/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/29/480429/2"}}},"commit":{"parents":[{"commit":"6f4adce6606ab210eb81ffed5c8e4d21a5db031f","subject":"Merge \"Boot server from encrypted volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6f4adce6606ab210eb81ffed5c8e4d21a5db031f"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-05 08:22:41.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-06 06:08:56.000000000","tz":330},"subject":"Added test_boot_volume_network_connectivity scenario test","message":"Added test_boot_volume_network_connectivity scenario test\n\n* This scenario test will test the floating ip connectivity for\n  an instance booted from cinder volume\n* Below is the list of operation it does:\n - Create in Cinder some bootable volume importing a Glance image\n - Create keypair and security group\n - Boot an instance from the bootable volume\n - Attach floating IP to it and check network connectivity\n - Delete the instance.\n\nChange-Id: Ia82d76d2ec45ba4967bc11e7524993d2bfd652c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1b474c53106f7a56fca85ce5688b3253718d3c8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1b474c53106f7a56fca85ce5688b3253718d3c8e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6f4adce6606ab210eb81ffed5c8e4d21a5db031f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
