)]}'
{"id":"openstack%2Ftempest~48219","triplet_id":"openstack%2Ftempest~master~I006c556facd457edfd1fa1cbaa95ccf28a89e236","project":"openstack/tempest","branch":"master","topic":"boot_from_snapshot","hashtags":[],"change_id":"I006c556facd457edfd1fa1cbaa95ccf28a89e236","subject":"Add test case for booting server from snapshot","status":"ABANDONED","created":"2013-09-25 11:15:46.000000000","updated":"2013-09-29 01:58:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3506dbffa6c7535bb3b9a2a29d50538b2b99b1eb","_number":48219,"virtual_id_number":48219,"owner":{"_account_id":8823,"name":"Jinhe Fang","email":"hzfangjh@corp.netease.com","username":"hzfangjh"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2013-09-25 11:48:28.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4468,"name":"Wangpan","email":"hzwangpan@corp.netease.com","username":"hzwangpan"},{"_account_id":6983,"name":"YangLei","email":"iamhappg@aliyun.com","username":"YangLei"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":1678,"name":"Tiantian Gao","email":"gtt116@gmail.com","username":"gtt116"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1678,"name":"Tiantian Gao","email":"gtt116@gmail.com","username":"gtt116"},"disliked":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4468,"name":"Wangpan","email":"hzwangpan@corp.netease.com","username":"hzwangpan"},{"value":1,"date":"2013-09-26 06:25:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6983,"name":"YangLei","email":"iamhappg@aliyun.com","username":"YangLei"},{"value":-1,"date":"2013-09-27 02:52:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"date":"2013-09-27 08:53:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":1,"date":"2013-09-26 03:05:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1678,"name":"Tiantian Gao","email":"gtt116@gmail.com","username":"gtt116"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"date":"2013-09-25 11:18:02.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2013-09-25 11:18:22.000000000","_account_id":4468,"name":"Wangpan","email":"hzwangpan@corp.netease.com","username":"hzwangpan"},{"_account_id":6983,"name":"YangLei","email":"iamhappg@aliyun.com","username":"YangLei"},{"date":"2013-09-27 02:52:13.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"date":"2013-09-26 15:27:12.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"date":"2013-09-25 11:18:08.000000000","_account_id":1678,"name":"Tiantian Gao","email":"gtt116@gmail.com","username":"gtt116"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4468,"name":"Wangpan","email":"hzwangpan@corp.netease.com","username":"hzwangpan"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6983,"name":"YangLei","email":"iamhappg@aliyun.com","username":"YangLei"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1678,"name":"Tiantian Gao","email":"gtt116@gmail.com","username":"gtt116"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1678,"name":"Tiantian Gao","email":"gtt116@gmail.com","username":"gtt116"},{"_account_id":4468,"name":"Wangpan","email":"hzwangpan@corp.netease.com","username":"hzwangpan"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":6983,"name":"YangLei","email":"iamhappg@aliyun.com","username":"YangLei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-09-25 11:18:02.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2013-09-25 11:18:08.000000000","updated_by":{"_account_id":1678,"name":"Tiantian Gao","email":"gtt116@gmail.com","username":"gtt116"},"reviewer":{"_account_id":1678,"name":"Tiantian Gao","email":"gtt116@gmail.com","username":"gtt116"},"state":"REVIEWER"},{"updated":"2013-09-25 11:18:22.000000000","updated_by":{"_account_id":4468,"name":"Wangpan","email":"hzwangpan@corp.netease.com","username":"hzwangpan"},"reviewer":{"_account_id":4468,"name":"Wangpan","email":"hzwangpan@corp.netease.com","username":"hzwangpan"},"state":"REVIEWER"},{"updated":"2013-09-26 06:25:25.000000000","updated_by":{"_account_id":6983,"name":"YangLei","email":"iamhappg@aliyun.com","username":"YangLei"},"reviewer":{"_account_id":6983,"name":"YangLei","email":"iamhappg@aliyun.com","username":"YangLei"},"state":"REVIEWER"},{"updated":"2013-09-26 15:27:12.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2013-09-27 02:52:13.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"}],"messages":[{"id":"a26b8e2aac264663eaa290e1dba7de4b50fbb085","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-09-25 11:48:28.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/19/48219/1/check/gate-tempest-pep8/2ac6e4f : SUCCESS in 1m 22s\n- gate-tempest-docs http://logs.openstack.org/19/48219/1/check/gate-tempest-docs/c3957fa : SUCCESS in 3m 24s\n- gate-tempest-devstack-vm-full http://logs.openstack.org/19/48219/1/check/gate-tempest-devstack-vm-full/b910ad7 : SUCCESS in 27m 33s\n- gate-tempest-devstack-vm-postgres-full http://logs.openstack.org/19/48219/1/check/gate-tempest-devstack-vm-postgres-full/e6618d2 : SUCCESS in 32m 36s\n- gate-tempest-devstack-vm-neutron http://logs.openstack.org/19/48219/1/check/gate-tempest-devstack-vm-neutron/fd4a8a6 : SUCCESS in 16m 59s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/19/48219/1/check/gate-tempest-devstack-vm-large-ops/458ba5a : SUCCESS in 11m 04s (non-voting)\n- gate-grenade-devstack-vm http://logs.openstack.org/19/48219/1/check/gate-grenade-devstack-vm/607f9b7 : SUCCESS in 23m 51s\n","accounts_in_message":[],"_revision_number":1},{"id":"7f364d6dde90b3ede4352c94d03df0faa4fc1524","author":{"_account_id":1678,"name":"Tiantian Gao","email":"gtt116@gmail.com","username":"gtt116"},"date":"2013-09-26 03:05:17.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\nMore tests always great.","accounts_in_message":[],"_revision_number":1},{"id":"574699b273006585d84ceca7144b1cfdcf4628c2","author":{"_account_id":6983,"name":"YangLei","email":"iamhappg@aliyun.com","username":"YangLei"},"date":"2013-09-26 06:25:25.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":1},{"id":"669114c1c661a868d08fa05bab6ac5cc41f46e4e","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2013-09-26 15:27:12.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"a5aa4581d8f2ae5eb41f1aa5c2fff2d68ff2b161","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2013-09-27 02:52:13.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\nSo there aren\u0027t any asserts in this test case you\u0027re just running api commands and saying that the test passes if they don\u0027t error out. \n\nYou should be more explicit in what you\u0027re testing. Check the response codes to make sure they are consistent. Also, check the state of the resources you\u0027re manipulating to see if things are in the expected state at given points. (For example, check glance to see if the image was created correctly)","accounts_in_message":[],"_revision_number":1},{"id":"b9e38657e4c1a6207f5484d5f8f73407c0375fbe","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2013-09-27 08:53:46.000000000","message":"Patch Set 1: No score\n\nActually, I also realized we also have a scenario test going trough these same steps. It uses the official client but leaving this as is probably doesn\u0027t add much value.\n\nSo despite my initial +2 , I think mtreinish\u0027s suggestion to add a few more checks against the actual image in glance is even more appropriate.","accounts_in_message":[],"_revision_number":1},{"id":"521051090576b559d6712c830b041283a7dcfb0c","author":{"_account_id":8823,"name":"Jinhe Fang","email":"hzfangjh@corp.netease.com","username":"hzfangjh"},"date":"2013-09-29 01:58:20.000000000","message":"Patch Set 1: Abandoned\n\nI realized there is a pretty good scenario test going trough these same steps. Abandon this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f911c1f76ec308f073d8e613c807662b52e3ab19","revisions":{"f911c1f76ec308f073d8e613c807662b52e3ab19":{"kind":"REWORK","_number":1,"created":"2013-09-25 11:15:46.000000000","uploader":{"_account_id":8823,"name":"Jinhe Fang","email":"hzfangjh@corp.netease.com","username":"hzfangjh"},"ref":"refs/changes/19/48219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/19/48219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/48219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/48219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/48219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/19/48219/1"}}},"commit":{"parents":[{"commit":"6da0db87265dd028f2aef507dfaf829203ae9a1a","subject":"Merge \"add negative volume tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6da0db87265dd028f2aef507dfaf829203ae9a1a"}]}],"author":{"name":"Jinhe Fang","email":"hzfangjh@corp.netease.com","date":"2013-09-25 10:58:58.000000000","tz":480},"committer":{"name":"Jinhe Fang","email":"hzfangjh@corp.netease.com","date":"2013-09-25 10:58:58.000000000","tz":480},"subject":"Add test case for booting server from snapshot","message":"Add test case for booting server from snapshot\n\nCreate a new image from a server, then boot another server from the new\nimage.\n\nChange-Id: I006c556facd457edfd1fa1cbaa95ccf28a89e236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f911c1f76ec308f073d8e613c807662b52e3ab19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f911c1f76ec308f073d8e613c807662b52e3ab19"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6da0db87265dd028f2aef507dfaf829203ae9a1a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
