)]}'
{"id":"openstack%2Ftempest~487884","triplet_id":"openstack%2Ftempest~master~I72ccf82debe36607b58fa75e52af535494999959","project":"openstack/tempest","branch":"master","topic":"bp/cinder-new-attach-apis","hashtags":[],"change_id":"I72ccf82debe36607b58fa75e52af535494999959","subject":"Add attachments checks to iscsi live migration test","status":"ABANDONED","created":"2017-07-27 14:56:27.000000000","updated":"2021-07-08 22:48:13.000000000","total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6d5d302d8b3a3c9ffd8455d67320769b46c4bc56","_number":487884,"virtual_id_number":487884,"owner":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2018-02-23 21:00:34.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"date":"2018-02-23 13:35:56.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"date":"2018-02-23 13:57:58.000000000","_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2018-02-23 15:32:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2018-02-23 16:30:06.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"date":"2018-02-23 13:08:11.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":24147,"name":"Mark Giles","email":"mark.giles@oracle.com","username":"mark-giles"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":-1,"date":"2021-07-08 17:12:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24147,"name":"Mark Giles","email":"mark.giles@oracle.com","username":"mark-giles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":24147,"name":"Mark Giles","email":"mark.giles@oracle.com","username":"mark-giles"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":24147,"name":"Mark Giles","email":"mark.giles@oracle.com","username":"mark-giles"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24147,"name":"Mark Giles","email":"mark.giles@oracle.com","username":"mark-giles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-27 14:57:44.000000000","updated_by":{"_account_id":24147,"name":"Mark Giles","email":"mark.giles@oracle.com","username":"mark-giles"},"reviewer":{"_account_id":24147,"name":"Mark Giles","email":"mark.giles@oracle.com","username":"mark-giles"},"state":"REVIEWER"},{"updated":"2017-07-27 20:13:39.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2017-07-27 20:13:57.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2017-07-28 06:24:03.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2017-09-04 08:10:23.000000000","updated_by":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"reviewer":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"state":"REVIEWER"},{"updated":"2017-09-07 07:13:37.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2017-09-12 21:45:34.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2017-09-26 17:28:51.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2017-09-26 20:05:19.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-12-06 21:51:58.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-02-23 13:08:11.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-23 13:35:56.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-23 15:32:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-23 16:30:06.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-23 21:00:34.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-08 17:12:12.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"82f0c4e2287f5e428b5ae6fae0fd32cf318f29e2","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-07-27 14:56:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d923070cb0bf41df8bf322e3e3894486def8f2cf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-27 16:37:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8000988/ : SUCCESS in 1h 23m 04s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/ad98d21/ : SUCCESS in 48m 16s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/169243f/ : SUCCESS in 1h 12m 23s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/1/check/gate-tempest-docs-ubuntu-xenial/57cbfe6//doc/build/html/ : SUCCESS in 1m 51s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/1/check/gate-tempest-pep8-ubuntu-xenial/bb4268c/ : FAILURE in 2m 43s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/1/check/gate-tempest-python27-ubuntu-xenial/7c8573b/ : SUCCESS in 3m 06s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/1/check/gate-tempest-python35/35183f0/ : SUCCESS in 2m 01s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/1b22e2f/ : SUCCESS in 42m 47s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/84/487884/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/571cfe6/ : SUCCESS in 28m 14s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/1f561db/ : SUCCESS in 1h 15m 45s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/c241a5f/ : SUCCESS in 44m 21s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/4aaf121/ : FAILURE in 1h 06m 18s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/8d99fe1/ : SUCCESS in 42m 21s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/c7b7382/ : SUCCESS in 37m 11s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/1/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/7896be2/ : SUCCESS in 1h 26m 57s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/ecf4178/ : SUCCESS in 1h 37m 44s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/c58f4e1/ : FAILURE in 1h 03m 00s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/1/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/ca34dc9/ : SUCCESS in 1h 02m 37s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/1/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/aae8911/ : FAILURE in 45m 37s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/1/check/tempest-coverage-ubuntu-xenial-nv/5fbb60e/ : SUCCESS in 3m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c83cd0d570898bb228c3da1e6e9d75a04c5184a9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-27 16:55:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/84/487884/1/check/dsvm-tempest-neutron-network/6e32f57 : SUCCESS in 1h 58m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7f366932ac54ed0f81b901ef306f9773be7da878","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-07-27 17:00:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"adcc10f9f6e220c75ae47a27695ebe2c5665588d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-27 18:44:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/17de3c9/ : SUCCESS in 1h 24m 55s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/026a3e7/ : SUCCESS in 55m 44s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/2/check/gate-tempest-dsvm-py35-ubuntu-xenial/b995b84/ : SUCCESS in 1h 34m 16s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/2/check/gate-tempest-docs-ubuntu-xenial/b0a9825//doc/build/html/ : SUCCESS in 2m 31s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/2/check/gate-tempest-pep8-ubuntu-xenial/4cde518/ : SUCCESS in 2m 53s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/2/check/gate-tempest-python27-ubuntu-xenial/a8d26c0/ : SUCCESS in 1m 43s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/2/check/gate-tempest-python35/0a9c5d3/ : SUCCESS in 1m 46s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/2/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/f1ed103/ : SUCCESS in 46m 28s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/84/487884/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/f858465/ : SUCCESS in 32m 13s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/a3bf99e/ : SUCCESS in 1h 09m 11s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/9aa938f/ : SUCCESS in 1h 00m 37s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/9b3dc88/ : SUCCESS in 55m 13s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/7066d35/ : SUCCESS in 1h 04m 08s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/7311cc4/ : SUCCESS in 40m 08s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/2/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/fb4e02b/ : SUCCESS in 1h 31m 33s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/f60f192/ : SUCCESS in 1h 19m 31s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/875104c/ : SUCCESS in 1h 12m 19s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/2/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/2e6c1cb/ : SUCCESS in 1h 07m 17s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/2/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/d4da34c/ : SUCCESS in 36m 55s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/2/check/tempest-coverage-ubuntu-xenial-nv/331f23d/ : SUCCESS in 2m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6888d9ab881e34cef83b00c2f8236f2b0ef4ddbf","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-27 18:57:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/84/487884/2/check/dsvm-tempest-neutron-network/4161163 : SUCCESS in 1h 55m 51s","accounts_in_message":[],"_revision_number":2},{"id":"242dae7dab79f2ca7c83b4841a06acb46f1808dc","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2017-07-28 06:24:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\na reno file is needed to add new interfaces in lib/services","accounts_in_message":[],"_revision_number":2},{"id":"6f7858618a87b679ec5f98ba32690e18f908882f","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-07-28 14:27:23.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1ee9b630a12fd057b0a34e7972c4eb9c5fe9fdcc","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-07-28 14:53:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cfcf9d78ac03768e65c2137e11bf7b6d4bb001a7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-28 16:26:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/1003d41/ : FAILURE in 1h 03m 23s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/3/check/gate-grenade-dsvm-neutron-ubuntu-xenial/e795f80/ : SUCCESS in 35m 31s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/3/check/gate-tempest-dsvm-py35-ubuntu-xenial/314d54d/ : FAILURE in 50m 21s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/3/check/gate-tempest-docs-ubuntu-xenial/4ca5b83//doc/build/html/ : SUCCESS in 4m 58s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/3/check/gate-tempest-pep8-ubuntu-xenial/e2da027/ : SUCCESS in 1m 34s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/3/check/gate-tempest-python27-ubuntu-xenial/88b770c/ : SUCCESS in 1m 17s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/3/check/gate-tempest-python35/9061044/ : SUCCESS in 1m 12s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/84/487884/3/check/gate-tempest-releasenotes/c3d4a53//releasenotes/build/html/ : SUCCESS in 2m 48s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/3/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/5f26423/ : SUCCESS in 1h 05m 54s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/84/487884/3/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/8af78c6/ : SUCCESS in 27m 18s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/1f6a84e/ : FAILURE in 1h 12m 40s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/3/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/605dce7/ : SUCCESS in 56m 15s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/3/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/27c85eb/ : FAILURE in 1h 05m 20s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/3/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/a96d93e/ : SUCCESS in 45m 25s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/3/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/881dfd9/ : SUCCESS in 27m 45s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/3/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/e32aa48/ : FAILURE in 1h 03m 01s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/6231913/ : SUCCESS in 1h 20m 42s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/208f9f2/ : SUCCESS in 1h 12m 39s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/3/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/6b43702/ : FAILURE in 1h 30m 28s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/3/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/00cf5e1/ : FAILURE in 46m 36s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/3/check/tempest-coverage-ubuntu-xenial-nv/31c3f31/ : SUCCESS in 1m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"833019a435dee553039dfd352fc468bdc814c665","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-28 16:47:32.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/84/487884/3/check/dsvm-tempest-neutron-network/2461643 : FAILURE in 1h 53m 14s","accounts_in_message":[],"_revision_number":3},{"id":"bb8b854c3509f063b38bb29c0f7ce4b9d0d0057b","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-07-28 18:35:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"49bd85401820e8e9ab8833206068f8bcc0e87ca5","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-07-31 14:37:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7444da646a4fa44df11f0121840138e05ea55b4b","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-07-31 17:34:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"63f98fab96857a539243d2ea23124482277e50ec","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-31 19:13:44.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/84/487884/4/check/dsvm-tempest-neutron-network/e9ddc84 : FAILURE in 1h 38m 04s","accounts_in_message":[],"_revision_number":4},{"id":"01ebf0da2ad04ded91a729b3a7ea6e0be8645b02","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-31 21:11:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/2673a61/ : FAILURE in 1h 24m 45s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/8bfeffd/ : SUCCESS in 53m 27s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-py35-ubuntu-xenial/419b4ab/ : FAILURE in 1h 11m 51s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/4/check/gate-tempest-docs-ubuntu-xenial/3a3de3f//doc/build/html/ : SUCCESS in 3m 00s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-pep8-ubuntu-xenial/74ba317/ : SUCCESS in 2m 07s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-python27-ubuntu-xenial/b655ea1/ : SUCCESS in 2m 28s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/4/check/gate-tempest-python35/d8f9046/ : SUCCESS in 2m 28s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/84/487884/4/check/gate-tempest-releasenotes/f1090ad//releasenotes/build/html/ : SUCCESS in 3m 43s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/b692df2/ : SUCCESS in 42m 02s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/92bb876/ : SUCCESS in 37m 31s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/e685479/ : FAILURE in 1h 05m 21s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/0cabd89/ : SUCCESS in 58m 38s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/945710b/ : FAILURE in 34m 12s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/2c305ef/ : FAILURE in 1h 06m 39s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/8642493/ : SUCCESS in 26m 13s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/0362b50/ : FAILURE in 1h 05m 24s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/0e599d2/ : SUCCESS in 1h 21m 23s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/1a53afc/ : SUCCESS in 1h 13m 01s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/9f8395f/ : FAILURE in 1h 39m 55s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/7aabcb7/ : SUCCESS in 35m 53s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/4/check/tempest-coverage-ubuntu-xenial-nv/526a100/ : SUCCESS in 1m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"479083ee0611a0fe42917dcdcb0d6877c8e99e29","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2017-08-01 13:15:50.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":4},{"id":"c3d2cb2036610097112f221b3687c02132252e03","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-08-01 15:13:16.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a69865286307e216cd9cdee95703b2065e77af19","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-01 16:48:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/534c3e1/ : FAILURE in 1h 07m 16s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/a2da729/ : SUCCESS in 45m 56s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-py35-ubuntu-xenial/8548aca/ : FAILURE in 1h 05m 07s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/4/check/gate-tempest-docs-ubuntu-xenial/0ced3ee//doc/build/html/ : SUCCESS in 7m 43s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-pep8-ubuntu-xenial/b936ba9/ : SUCCESS in 2m 25s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-python27-ubuntu-xenial/4cf1c8d/ : SUCCESS in 2m 50s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/4/check/gate-tempest-python35/57daf5c/ : SUCCESS in 1m 22s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/84/487884/4/check/gate-tempest-releasenotes/578bdf6//releasenotes/build/html/ : SUCCESS in 3m 18s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/4714f91/ : SUCCESS in 1h 08m 49s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/29739f5/ : SUCCESS in 38m 49s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/0397332/ : FAILURE in 1h 03m 39s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/819e73b/ : SUCCESS in 45m 55s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/6d7cb79/ : SUCCESS in 59m 23s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/194ba75/ : FAILURE in 1h 02m 59s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/35bd463/ : SUCCESS in 38m 36s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/f5bf56d/ : FAILURE in 1h 03m 20s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/4e8f628/ : SUCCESS in 1h 28m 13s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/a87f90a/ : SUCCESS in 1h 11m 04s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/fe91a5b/ : FAILURE in 1h 26m 30s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/677ba6b/ : FAILURE in 45m 55s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/4/check/tempest-coverage-ubuntu-xenial-nv/acadd6a/ : SUCCESS in 2m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a1551d4aff0220d729c3bce1bfa32f5908238434","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-08-04 13:32:29.000000000","message":"Patch Set 4:\n\nThe tempest tests using cinder v3.27 attach are going to fail until bug 1708188 is fixed.","accounts_in_message":[],"_revision_number":4},{"id":"6c2285d4593a3263467c9e2923e3e65eb749ef3d","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-08-08 18:51:21.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d2e63c62ede9acc6122d186ba2f47b96caf384fd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-08 22:39:47.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d7fb305/ : FAILURE in 1h 14m 26s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/c40c2e2/ : SUCCESS in 48m 43s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-py35-ubuntu-xenial/f289fcd/ : FAILURE in 1h 28m 01s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/4/check/gate-tempest-docs-ubuntu-xenial/a9a84b9//doc/build/html/ : SUCCESS in 18m 31s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-pep8-ubuntu-xenial/6948173/ : SUCCESS in 12m 46s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-python27-ubuntu-xenial/075a85f/ : SUCCESS in 2m 14s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/4/check/gate-tempest-python35/09a565c/ : SUCCESS in 2m 40s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/84/487884/4/check/gate-tempest-releasenotes/bac0a6b//releasenotes/build/html/ : SUCCESS in 3m 17s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/494dce2/ : SUCCESS in 1h 04m 33s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/4070cd5/ : FAILURE in 1h 03m 12s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/dc650ab/ : FAILURE in 1h 49m 54s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/1ff0910/ : SUCCESS in 56m 11s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/44fee93/ : FAILURE in 20m 52s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/b6b79bc/ : SUCCESS in 53m 17s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/6a17fe6/ : SUCCESS in 36m 34s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/1ba216d/ : FAILURE in 1h 58m 25s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/47a6f1a/ : SUCCESS in 1h 32m 07s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/1770ac5/ : FAILURE in 1h 06m 11s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/a50821f/ : FAILURE in 1h 34m 43s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/71062d4/ : FAILURE in 45m 38s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/4/check/tempest-coverage-ubuntu-xenial-nv/eb6597e/ : SUCCESS in 3m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"57a96e291e19e67d6af0af88cae87216b3792313","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-08-09 12:47:23.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"21faf1caf6bc9da15565f64eecd19f922abd1726","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-09 19:06:24.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/dc73a02/ : FAILURE in 1h 00m 40s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/7a524bf/ : SUCCESS in 54m 05s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-py35-ubuntu-xenial/d36f58b/ : FAILURE in 2h 54m 44s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/4/check/gate-tempest-docs-ubuntu-xenial/018a78a//doc/build/html/ : SUCCESS in 6m 13s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-pep8-ubuntu-xenial/4dcbbb8/ : SUCCESS in 3m 03s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-python27-ubuntu-xenial/b98761d/ : SUCCESS in 6m 08s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/4/check/gate-tempest-python35/2499251/ : SUCCESS in 2m 22s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/84/487884/4/check/gate-tempest-releasenotes/f928ad8//releasenotes/build/html/ : SUCCESS in 4m 28s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/284c5e0/ : SUCCESS in 55m 02s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/90ffb18/ : FAILURE in 1h 03m 01s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/d1120fb/ : FAILURE in 56m 47s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/6935621/ : FAILURE in 1h 06m 02s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/393936b/ : SUCCESS in 52m 34s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/a5e886e/ : SUCCESS in 1h 00m 34s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/4/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/a648ac6/ : SUCCESS in 40m 21s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/ccda3ed/ : FAILURE in 1h 14m 58s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/5545ff4/ : SUCCESS in 1h 46m 02s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/6838f0f/ : SUCCESS in 1h 52m 35s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/4/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/2add2d3/ : FAILURE in 1h 22m 09s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/4/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/76d63b1/ : FAILURE in 45m 34s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/4/check/tempest-coverage-ubuntu-xenial-nv/42c05b4/ : SUCCESS in 2m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cfdeda2251356e4cd53344d27945ad0b9adde771","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-08-10 15:24:18.000000000","message":"Patch Set 4:\n\nThis change will continue to fail until the release of the a cinder client with this fix: https://review.openstack.org/#/c/490040/","accounts_in_message":[],"_revision_number":4},{"id":"d86f3e89ec8d35157f20846ee10026f562fd6e2f","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-08-28 17:50:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"43aa798ddc34b160a65d6155e5060bdf91c8c1c5","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-08-28 17:51:01.000000000","message":"Patch Set 5:\n\npatch set 5 fixed a merge conflict","accounts_in_message":[],"_revision_number":5},{"id":"5f327e6db6c832cf9811d3bc2044760140c4e616","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-08-28 19:27:05.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/84/487884/5/check/dsvm-tempest-neutron-network/75c8801 : FAILURE in 1h 35m 07s","accounts_in_message":[],"_revision_number":5},{"id":"b6b52d863164813878d204c6463c0a8e09944aa5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-28 19:43:30.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"406ef1680a1d93e85e086f404c69f68cad8ab302","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-08-29 17:09:58.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"bc70d59d53b6ea764039f0e1ae68efb2f962c99e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-29 17:18:01.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"1cf7c76a24e0f479a613c9eb1ecf6103627c3a99","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-08-30 13:16:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"d044c7dd2c9add53823da5357bdb70f50b8960f2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-08-30 13:18:10.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"bd27218b29aaa03748a2adb8abe7dd3e7850b520","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 13:18:35.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"2ff20a8d66ee94fb84937721f75198d7cf5a63dd","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-08-30 14:39:41.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c6158f6bdfed5e14ecb17740b24d70323645abc9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 15:25:49.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"970a5f09b74c180c2ba243e9ad780a7b0a03c020","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-08-30 20:41:51.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a59a05640e615e933f87a730b8b18b891dec5032","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 22:19:21.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/71f7067/ : FAILURE in 1h 15m 25s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3139448/ : SUCCESS in 48m 36s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-py35-ubuntu-xenial/a0e123f/ : FAILURE in 50m 58s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/6/check/gate-tempest-docs-ubuntu-xenial/bc36bf4//doc/build/html/ : SUCCESS in 5m 12s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-pep8-ubuntu-xenial/c577882/ : SUCCESS in 3m 25s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-python27-ubuntu-xenial/85fb4c1/ : SUCCESS in 2m 26s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/6/check/gate-tempest-python35/6954376/ : SUCCESS in 2m 17s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/84/487884/6/check/gate-tempest-releasenotes/8c8f036//releasenotes/build/html/ : SUCCESS in 3m 49s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/c81f704/ : SUCCESS in 1h 03m 22s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/216687c/ : SUCCESS in 47m 38s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/571deaf/ : FAILURE in 1h 19m 29s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/6/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/0ce7a04/ : SUCCESS in 1h 02m 05s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/6/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/1695e09/ : SUCCESS in 59m 55s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/6/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/7ee6b88/ : SUCCESS in 58m 27s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/6/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/3397d40/ : SUCCESS in 39m 37s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/a348a09/ : FAILURE in 1h 02m 33s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/0af6281/ : SUCCESS in 1h 32m 38s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/85a1043/ : FAILURE in 1h 01m 19s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/6449698/ : FAILURE in 1h 29m 30s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/e77e397/ : FAILURE in 1h 17m 27s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/f4cd44e/ : FAILURE in 45m 56s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/6/check/tempest-coverage-ubuntu-xenial-nv/90f471a/ : SUCCESS in 3m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4d91dafd2835e2bda4a6d3bdc8e7ac9c9e67a721","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2017-09-04 08:10:23.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"18e34cc14b4df3d909c3fb4bffc4a1fdb8acbda7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-04 10:20:40.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/11b2fed/ : SUCCESS in 1h 35m 45s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-grenade-dsvm-neutron-ubuntu-xenial/5ee4fe9/ : SUCCESS in 51m 23s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-py35-ubuntu-xenial/068338d/ : SUCCESS in 1h 16m 18s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/6/check/gate-tempest-docs-ubuntu-xenial/01906eb//doc/build/html/ : SUCCESS in 4m 39s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-pep8-ubuntu-xenial/f3db754/ : SUCCESS in 2m 40s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-python27-ubuntu-xenial/d959a76/ : SUCCESS in 1m 50s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/6/check/gate-tempest-python35/70d87c2/ : SUCCESS in 2m 10s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/84/487884/6/check/gate-tempest-releasenotes/33f7fae//releasenotes/build/html/ : SUCCESS in 5m 14s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/6a9ab83/ : SUCCESS in 1h 03m 44s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/41db413/ : SUCCESS in 46m 44s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/91ee8a3/ : SUCCESS in 1h 18m 47s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/6/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/c1c942b/ : SUCCESS in 1h 02m 24s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/6/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/e819c4d/ : SUCCESS in 57m 58s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/6/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/ec4904f/ : SUCCESS in 59m 15s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/6/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/eee456c/ : SUCCESS in 40m 15s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/aebb32c/ : FAILURE in 2h 03m 51s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/7f420e9/ : SUCCESS in 1h 31m 08s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/922cddf/ : SUCCESS in 1h 27m 03s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/5d415fb/ : SUCCESS in 1h 25m 23s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/748c493/ : SUCCESS in 1h 36m 44s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/b369e55/ : FAILURE in 45m 29s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/6/check/tempest-coverage-ubuntu-xenial-nv/4b221d8/ : SUCCESS in 1m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"09c08e3e35fcff7e5bfff12c4bc000be7634eaa9","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-09-04 10:21:32.000000000","message":"Patch Set 6:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial: https://bugs.launchpad.net/bugs/1686542\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"1a2563bd7b8bf58b2257925fb0ce5bbe50e3aec1","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2017-09-04 23:07:32.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"46b83248000e20c0446178bc997ed5e8289a768c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-05 00:46:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/bfcba15/ : SUCCESS in 1h 33m 54s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-grenade-dsvm-neutron-ubuntu-xenial/c34ca3e/ : SUCCESS in 53m 40s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-py35-ubuntu-xenial/5378520/ : SUCCESS in 1h 16m 33s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/6/check/gate-tempest-docs-ubuntu-xenial/6e2f2c9//doc/build/html/ : SUCCESS in 1m 55s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-pep8-ubuntu-xenial/01629e2/ : SUCCESS in 3m 01s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-python27-ubuntu-xenial/a1dacad/ : SUCCESS in 1m 37s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/6/check/gate-tempest-python35/093ddeb/ : SUCCESS in 2m 14s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/84/487884/6/check/gate-tempest-releasenotes/7ebc7c4//releasenotes/build/html/ : SUCCESS in 4m 12s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/e516920/ : SUCCESS in 1h 00m 32s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/c25e9a2/ : SUCCESS in 46m 55s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/56dfda1/ : SUCCESS in 1h 08m 18s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/6/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/3db9dde/ : SUCCESS in 54m 02s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/6/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/38a9b95/ : SUCCESS in 51m 04s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/6/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/8dbc768/ : SUCCESS in 1h 02m 22s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/6/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/e82cafe/ : SUCCESS in 41m 16s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/2d0bf4f/ : SUCCESS in 1h 29m 09s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/5d7f3f9/ : SUCCESS in 1h 30m 34s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/94d96df/ : SUCCESS in 1h 26m 57s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/61eec44/ : SUCCESS in 1h 23m 52s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/84/487884/6/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/e26816c/ : SUCCESS in 1h 31m 27s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/6/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/233d450/ : FAILURE in 45m 20s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/6/check/tempest-coverage-ubuntu-xenial-nv/49a8afb/ : SUCCESS in 2m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6fed674f052bbb409de2232dbd2ca4a0719662c7","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-09-07 07:13:37.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nCould you add unit tests for the new client to ensure the stable interface of tempest.lib?","accounts_in_message":[],"_revision_number":6},{"id":"3728cd10f0f5935528c77e24b8e515dce6d266b6","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-09-07 13:57:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8974d8abacf5e20e2394a99d7fc8a441d91bd2e0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-09-07 16:15:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8e4d885e9cacc82b9101be4f605228b5b68455ea","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-07 16:28:18.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/84/487884/7/check/dsvm-tempest-neutron-network/4bcba3c : SUCCESS in 2h 13m 22s","accounts_in_message":[],"_revision_number":7},{"id":"a34e4ed413d96fce807375618eb1f10a09a8b2fa","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-09-07 16:46:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0243c193c4bf1290f59dd3fda1d11a421c4c03dd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-07 16:49:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d815571/ : SUCCESS in 1h 32m 45s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/7/check/gate-grenade-dsvm-neutron-ubuntu-xenial/9e2cead/ : SUCCESS in 59m 31s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/7/check/gate-tempest-dsvm-py35-ubuntu-xenial/9954646/ : SUCCESS in 1h 21m 11s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/7/check/gate-tempest-docs-ubuntu-xenial/579e5f2//doc/build/html/ : SUCCESS in 5m 01s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/7/check/gate-tempest-pep8-ubuntu-xenial/4c0b04a/ : SUCCESS in 3m 05s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/7/check/gate-tempest-python27-ubuntu-xenial/68b48b2/ : SUCCESS in 2m 28s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/7/check/gate-tempest-python35/3460adc/ : SUCCESS in 3m 26s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/84/487884/7/check/gate-tempest-releasenotes/3908f62//releasenotes/build/html/ : SUCCESS in 4m 24s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/7/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/586e56b/ : SUCCESS in 1h 02m 55s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/84/487884/7/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/df89e9f/ : SUCCESS in 42m 01s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/108e6be/ : SUCCESS in 1h 06m 09s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/7/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/6a71f8e/ : SUCCESS in 1h 02m 47s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/7/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/17df3be/ : SUCCESS in 59m 19s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/7/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/f1b35b0/ : SUCCESS in 1h 02m 42s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/7/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/d96d82e/ : SUCCESS in 36m 59s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/7/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/e011d20/ : SUCCESS in 1h 48m 24s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/b0afca5/ : SUCCESS in 2h 43m 32s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/ab4ae07/ : SUCCESS in 1h 43m 42s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/7/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/1a655af/ : SUCCESS in 1h 13m 17s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/84/487884/7/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/0dc90d9/ : SUCCESS in 2h 05m 56s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/7/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/1347427/ : FAILURE in 45m 47s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/7/check/tempest-coverage-ubuntu-xenial-nv/a0d7591/ : SUCCESS in 2m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"30d8793138715f5d7e421e5cad0b7aa9906b5407","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-09-07 19:42:51.000000000","message":"Patch Set 7:\n\nPassed here:\n\nhttp://logs.openstack.org/05/501805/2/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/069e82e/console.html#_2017-09-07_17_45_00_706868","accounts_in_message":[],"_revision_number":7},{"id":"2744e25d222f88908aab6a3b206be77f0edc720b","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2017-09-12 21:45:34.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"44c64524a5e10e218ea5d279c458737dc56195cb","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-09-21 15:52:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e26ea8b89cad8b1ca0c9f125bd65bf78bf157a01","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-09-22 15:44:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"65d66aa9478dc872b6a3f95fc7114e59112b066e","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-09-22 17:17:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"30f0260cfae86e66c6c0cb82c645dcae96586106","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 19:50:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/0477ab0/ : SUCCESS in 1h 34m 52s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-grenade-dsvm-neutron-ubuntu-xenial/69a8ce1/ : SUCCESS in 50m 47s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-py35-ubuntu-xenial/f786291/ : SUCCESS in 1h 22m 25s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/8/check/gate-tempest-docs-ubuntu-xenial/f2bb0b0//doc/build/html/ : SUCCESS in 4m 15s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-pep8-ubuntu-xenial/ce62bc8/ : SUCCESS in 2m 49s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-python27-ubuntu-xenial/d611608/ : SUCCESS in 2m 28s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/8/check/gate-tempest-python35/82f6a1a/ : SUCCESS in 2m 14s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/84/487884/8/check/gate-tempest-releasenotes/50c889f//releasenotes/build/html/ : SUCCESS in 4m 30s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/8394383/ : SUCCESS in 59m 15s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/52d9f59/ : SUCCESS in 1h 21m 55s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/8/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/61cf705/ : SUCCESS in 55m 45s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/8/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/e0db94f/ : SUCCESS in 57m 50s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/8/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/dde1648/ : SUCCESS in 58m 50s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/8/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/6b8e7ff/ : SUCCESS in 44m 07s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/6c85174/ : SUCCESS in 1h 23m 50s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-pike http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-pike/37082f6/ : SUCCESS in 1h 35m 23s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/7f75880/ : FAILURE in 1h 06m 01s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/4e0e796/ : SUCCESS in 1h 16m 21s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/b4db453/ : SUCCESS in 1h 39m 46s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/bc240eb/ : SUCCESS in 1h 56m 22s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/08c45f8/ : FAILURE in 45m 55s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/8/check/tempest-coverage-ubuntu-xenial-nv/d537c7e/ : SUCCESS in 2m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"02fa727903b916fbbd36eab0b3e9a11257c061d6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-22 19:52:06.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/84/487884/8/check/dsvm-tempest-neutron-network/4b82a19 : SUCCESS in 2h 33m 51s","accounts_in_message":[],"_revision_number":8},{"id":"74980bc2e879bf5a93f57540c6bb98c3641ca179","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-09-26 14:49:02.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"3ae94951630ee65254398b5c324f49dec318d936","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-26 16:46:57.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/0a9e49b/ : FAILURE in 44m 15s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-grenade-dsvm-neutron-ubuntu-xenial/8d6558c/ : SUCCESS in 1h 22m 35s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-py35-ubuntu-xenial/25062a8/ : SUCCESS in 1h 15m 59s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/8/check/gate-tempest-docs-ubuntu-xenial/3141023//doc/build/html/ : SUCCESS in 6m 46s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-pep8-ubuntu-xenial/256b799/ : SUCCESS in 3m 34s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-python27-ubuntu-xenial/479d9c0/ : SUCCESS in 3m 20s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/8/check/gate-tempest-python35/6be1bb1/ : SUCCESS in 2m 54s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/84/487884/8/check/gate-tempest-releasenotes/06066cd//releasenotes/build/html/ : SUCCESS in 4m 05s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/0db0747/ : SUCCESS in 1h 29m 38s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/d38388e/ : FAILURE in 1h 25m 55s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/8/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/304773d/ : FAILURE in 38m 13s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/8/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/180c65b/ : FAILURE in 39m 45s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/8/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/687d5fa/ : FAILURE in 42m 43s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/8/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/79c1a57/ : FAILURE in 1m 34s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/cce9cda/ : SUCCESS in 1h 26m 02s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-pike http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-pike/88a9153/ : SUCCESS in 1h 44m 29s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/b2d9dc6/ : SUCCESS in 1h 41m 42s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/9c4b3d1/ : SUCCESS in 1h 41m 52s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/b5817bf/ : FAILURE in 1h 38m 12s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/430073d/ : SUCCESS in 1h 33m 18s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/39d2f5f/ : FAILURE in 1m 20s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/8/check/tempest-coverage-ubuntu-xenial-nv/2600c10/ : SUCCESS in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7dc7796e714b4a0701ccbc0610e368671f79aa6f","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-09-26 17:28:51.000000000","message":"Patch Set 8:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial: https://bugs.launchpad.net/bugs/1298006 https://bugs.launchpad.net/bugs/1282876\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":8},{"id":"cfbd42908b31e7d1a8e380959d64be3efc6e08aa","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2017-09-26 17:33:30.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"b21c594f96d9ac786cab5bb25bd38ab2e7eea08b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-26 20:05:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c2e30c3/ : SUCCESS in 1h 17m 01s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-grenade-dsvm-neutron-ubuntu-xenial/d1769a3/ : SUCCESS in 54m 43s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-py35-ubuntu-xenial/8484344/ : SUCCESS in 1h 26m 36s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/84/487884/8/check/gate-tempest-docs-ubuntu-xenial/25578cc//doc/build/html/ : SUCCESS in 7m 39s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-pep8-ubuntu-xenial/76ee3ce/ : SUCCESS in 4m 33s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-python27-ubuntu-xenial/636f16f/ : SUCCESS in 4m 01s\n- gate-tempest-python35 http://logs.openstack.org/84/487884/8/check/gate-tempest-python35/5e70948/ : SUCCESS in 4m 02s\n- gate-tempest-releasenotes http://docs-draft.openstack.org/84/487884/8/check/gate-tempest-releasenotes/fab76bb//releasenotes/build/html/ : SUCCESS in 5m 12s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/b699db3/ : SUCCESS in 1h 07m 44s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/1f54bd8/ : FAILURE in 1h 06m 29s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/84/487884/8/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/cd04299/ : FAILURE in 39m 40s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/84/487884/8/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/32fca09/ : FAILURE in 42m 00s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/84/487884/8/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/852a097/ : FAILURE in 2m 32s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/84/487884/8/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/2013978/ : FAILURE in 28m 51s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/57516e4/ : SUCCESS in 1h 46m 00s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-pike http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-pike/4b0b6e6/ : SUCCESS in 1h 43m 45s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/3d8cf1f/ : SUCCESS in 1h 35m 35s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/f77da65/ : SUCCESS in 1h 20m 27s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/d60228c/ : SUCCESS in 1h 20m 17s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/84/487884/8/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/05a31bd/ : FAILURE in 2m 27s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/84/487884/8/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/50cec5c/ : FAILURE in 46m 41s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/84/487884/8/check/tempest-coverage-ubuntu-xenial-nv/27565cb/ : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ca5735c3d1fe4998809c3e04fe07c941856407d1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-12-06 21:51:58.000000000","message":"Patch Set 8:\n\nrecheck dep updates","accounts_in_message":[],"_revision_number":8},{"id":"fa02885d2158117f194394506906b5da069eabcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-07 01:00:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/84/487884/8/check/legacy-tempest-dsvm-py35/5e2cb96/ : SUCCESS in 1h 21m 18s\n- openstack-tox-pep8 http://logs.openstack.org/84/487884/8/check/openstack-tox-pep8/0c413b3/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/84/487884/8/check/openstack-tox-py27/713a77e/ : SUCCESS in 3m 03s\n- openstack-tox-py35 http://logs.openstack.org/84/487884/8/check/openstack-tox-py35/b36fe94/ : SUCCESS in 3m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/84/487884/8/check/build-openstack-sphinx-docs/1964160/html/ : SUCCESS in 6m 04s\n- build-openstack-releasenotes http://logs.openstack.org/84/487884/8/check/build-openstack-releasenotes/8b3b07c/html/ : SUCCESS in 4m 50s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/84/487884/8/check/legacy-grenade-dsvm-neutron-multinode/be721c7/ : SUCCESS in 1h 04m 37s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/84/487884/8/check/legacy-grenade-dsvm-neutron/b79e0fb/ : SUCCESS in 59m 05s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/84/487884/8/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/3174727/ : FAILURE in 58m 41s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/84/487884/8/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/2151d06/ : SUCCESS in 57m 56s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/84/487884/8/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/c9b65ec/ : SUCCESS in 53m 37s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/84/487884/8/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/198bd76/ : SUCCESS in 1h 09m 14s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/84/487884/8/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/d98d7d7/ : SUCCESS in 43m 50s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/84/487884/8/check/legacy-tempest-dsvm-neutron-dvr/bb84dd3/ : SUCCESS in 1h 13m 47s\n- legacy-tempest-dsvm-neutron-full-pike http://logs.openstack.org/84/487884/8/check/legacy-tempest-dsvm-neutron-full-pike/8ae1fd8/ : SUCCESS in 1h 24m 27s (non-voting)\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/84/487884/8/check/legacy-tempest-dsvm-neutron-full-ocata/7d0e6bb/ : SUCCESS in 1h 46m 47s (non-voting)\n- legacy-tempest-dsvm-neutron-full-newton http://logs.openstack.org/84/487884/8/check/legacy-tempest-dsvm-neutron-full-newton/cc82559/ : SUCCESS in 1h 13m 06s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend http://logs.openstack.org/84/487884/8/check/legacy-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend/0f54f3c/ : SUCCESS in 1h 13m 31s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/84/487884/8/check/legacy-tempest-dsvm-neutron-multinode-full/c8c3a3b/ : SUCCESS in 1h 32m 07s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/84/487884/8/check/openstack-tox-cover/28ae376/cover/ : SUCCESS in 3m 17s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/84/487884/8/check/legacy-tempest-dsvm-neutron-full/4629640/ : SUCCESS in 1h 18m 13s\n- tempest-full http://logs.openstack.org/84/487884/8/check/tempest-full/04bb810/ : SUCCESS in 1h 05m 05s (non-voting)\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/84/487884/8/check/tempest-tox-plugin-sanity-check/8150eb3/ : POST_FAILURE in 58m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"95073ae54451fd6a6989d02e6f1f9d5da14d50bd","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2018-02-23 11:07:27.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"66192830f5343c8b0ded589b2c24bac358c5997a","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2018-02-23 11:08:09.000000000","message":"Patch Set 9:\n\nRebased and fixed the merge issues.","accounts_in_message":[],"_revision_number":9},{"id":"97134cbfa1098e3495e703e485523545ad62bf8e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-02-23 12:06:37.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset021-master https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset021-master/853/ : FAILURE in 50m 26s","accounts_in_message":[],"_revision_number":9},{"id":"6fbb05f369b782f2e71fdd11166e1195882ecc7c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-02-23 13:08:11.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/84/487884/9/check/dsvm-tempest-neutron-network/e58467f : SUCCESS in 1h 59m 15s","accounts_in_message":[],"_revision_number":9},{"id":"f33de7ee8538399f2f94702cb296f3d0fe9b5185","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-02-23 13:35:56.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- OVS_HW_offload http://52.169.200.208/84/487884/9/check-sriov-tempest/OVS_HW_offload/70de8e8 : SUCCESS in 52m 06s (non-voting)\n- Tempest-Sriov http://52.169.200.208/84/487884/9/check-sriov-tempest/Tempest-Sriov/c8adec3 : SUCCESS in 1h 02m 41s (non-voting)\n\nTo re-run the job post \u0027recheck tempest-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":9},{"id":"971281179235b0a9dba39b231c7674ffd094f735","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-23 13:36:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/84/487884/9/check/tempest-full/8e9a29b/ : SUCCESS in 1h 18m 49s\n- neutron-grenade http://logs.openstack.org/84/487884/9/check/neutron-grenade/822f789/ : SUCCESS in 42m 02s\n- tempest-full-py3 http://logs.openstack.org/84/487884/9/check/tempest-full-py3/516d284/ : SUCCESS in 1h 34m 42s\n- openstack-tox-pep8 http://logs.openstack.org/84/487884/9/check/openstack-tox-pep8/9e9eabe/ : SUCCESS in 2m 50s\n- openstack-tox-py27 http://logs.openstack.org/84/487884/9/check/openstack-tox-py27/1a76203/ : SUCCESS in 3m 29s\n- openstack-tox-py35 http://logs.openstack.org/84/487884/9/check/openstack-tox-py35/1ba4b90/ : SUCCESS in 2m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/84/487884/9/check/build-openstack-sphinx-docs/4d2a6e4/html/ : SUCCESS in 13m 27s\n- build-openstack-releasenotes http://logs.openstack.org/84/487884/9/check/build-openstack-releasenotes/bcbe49b/html/ : SUCCESS in 4m 47s\n- neutron-grenade-multinode http://logs.openstack.org/84/487884/9/check/neutron-grenade-multinode/dd9c5e2/ : SUCCESS in 1h 09m 34s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/84/487884/9/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/04795f5/ : SUCCESS in 1h 09m 19s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/84/487884/9/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/b5b0925/ : TIMED_OUT in 1h 30m 49s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/84/487884/9/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/beca9f2/ : TIMED_OUT in 1h 30m 53s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/84/487884/9/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/97076bb/ : TIMED_OUT in 1h 30m 45s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/84/487884/9/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/d54e72c/ : TIMED_OUT in 1h 30m 46s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/84/487884/9/check/neutron-tempest-dvr/5f00f74/ : SUCCESS in 1h 33m 53s\n- legacy-tempest-dsvm-neutron-full-pike http://logs.openstack.org/84/487884/9/check/legacy-tempest-dsvm-neutron-full-pike/d2228e5/ : SUCCESS in 2h 18m 34s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/84/487884/9/check/legacy-tempest-dsvm-neutron-full-ocata/1e6faf2/ : SUCCESS in 1h 21m 13s\n- legacy-tempest-dsvm-neutron-full-newton http://logs.openstack.org/84/487884/9/check/legacy-tempest-dsvm-neutron-full-newton/2d43d35/ : SUCCESS in 1h 14m 27s\n- legacy-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend http://logs.openstack.org/84/487884/9/check/legacy-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend/377ee06/ : SUCCESS in 1h 20m 05s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/84/487884/9/check/neutron-tempest-multinode-full/de156e1/ : SUCCESS in 1h 37m 16s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/84/487884/9/check/openstack-tox-cover/6af31eb/cover/ : SUCCESS in 3m 47s (non-voting)\n- nova-multiattach http://logs.openstack.org/84/487884/9/check/nova-multiattach/544dd82/ : SUCCESS in 46m 27s\n- tempest-full-queens http://logs.openstack.org/84/487884/9/check/tempest-full-queens/2db4c2e/ : SUCCESS in 1h 20m 26s\n- tempest-full-queens-py3 http://logs.openstack.org/84/487884/9/check/tempest-full-queens-py3/7081e8a/ : SUCCESS in 1h 34m 09s\n- tempest-full-pike http://logs.openstack.org/84/487884/9/check/tempest-full-pike/fab2d3f/ : FAILURE in 1h 21m 37s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/84/487884/9/check/tempest-tox-plugin-sanity-check/a458239/ : FAILURE in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"25d2a665b6549f33a8c9cc18d5dfb4fa5fb28534","author":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"date":"2018-02-23 13:57:58.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"75835e3e9e000a926349b1c977e7efdcd4674cf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-23 15:32:59.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/84/487884/9/check/tempest-full/3cf3bdd/ : SUCCESS in 1h 20m 38s\n- neutron-grenade http://logs.openstack.org/84/487884/9/check/neutron-grenade/90923ee/ : SUCCESS in 41m 59s\n- tempest-full-py3 http://logs.openstack.org/84/487884/9/check/tempest-full-py3/97e6b80/ : SUCCESS in 1h 24m 34s\n- openstack-tox-pep8 http://logs.openstack.org/84/487884/9/check/openstack-tox-pep8/3b2042b/ : SUCCESS in 2m 27s\n- openstack-tox-py27 http://logs.openstack.org/84/487884/9/check/openstack-tox-py27/4b7701e/ : SUCCESS in 2m 37s\n- openstack-tox-py35 http://logs.openstack.org/84/487884/9/check/openstack-tox-py35/056397a/ : SUCCESS in 3m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/84/487884/9/check/build-openstack-sphinx-docs/a6503b2/html/ : SUCCESS in 13m 36s\n- build-openstack-releasenotes http://logs.openstack.org/84/487884/9/check/build-openstack-releasenotes/e50673f/html/ : SUCCESS in 4m 56s\n- neutron-grenade-multinode http://logs.openstack.org/84/487884/9/check/neutron-grenade-multinode/a123bd5/ : SUCCESS in 56m 41s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/84/487884/9/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/5833674/ : SUCCESS in 1h 00m 33s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/84/487884/9/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/d6ca282/ : TIMED_OUT in 1h 30m 54s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/84/487884/9/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/550f4f3/ : TIMED_OUT in 1h 30m 54s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/84/487884/9/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/742a96c/ : TIMED_OUT in 1h 30m 45s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/84/487884/9/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/03445b5/ : TIMED_OUT in 1h 31m 05s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/84/487884/9/check/neutron-tempest-dvr/64657fc/ : SUCCESS in 1h 29m 42s\n- legacy-tempest-dsvm-neutron-full-pike http://logs.openstack.org/84/487884/9/check/legacy-tempest-dsvm-neutron-full-pike/c6dfc64/ : SUCCESS in 1h 19m 42s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/84/487884/9/check/legacy-tempest-dsvm-neutron-full-ocata/591ea95/ : SUCCESS in 1h 24m 41s\n- legacy-tempest-dsvm-neutron-full-newton http://logs.openstack.org/84/487884/9/check/legacy-tempest-dsvm-neutron-full-newton/a2e1504/ : SUCCESS in 1h 21m 44s\n- legacy-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend http://logs.openstack.org/84/487884/9/check/legacy-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend/5819718/ : FAILURE in 1h 19m 46s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/84/487884/9/check/neutron-tempest-multinode-full/e46d0ab/ : SUCCESS in 1h 30m 15s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/84/487884/9/check/openstack-tox-cover/b386937/cover/ : SUCCESS in 3m 11s (non-voting)\n- nova-multiattach http://logs.openstack.org/84/487884/9/check/nova-multiattach/24ce760/ : SUCCESS in 1h 01m 32s\n- tempest-full-queens http://logs.openstack.org/84/487884/9/check/tempest-full-queens/ef81d1c/ : SUCCESS in 1h 22m 31s\n- tempest-full-queens-py3 http://logs.openstack.org/84/487884/9/check/tempest-full-queens-py3/31b1885/ : SUCCESS in 1h 22m 29s\n- tempest-full-pike http://logs.openstack.org/84/487884/9/check/tempest-full-pike/474310e/ : SUCCESS in 1h 26m 55s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/84/487884/9/check/tempest-tox-plugin-sanity-check/e7dba29/ : FAILURE in 28m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a0d1c5cff499c6f850ac7cde92bb8edf4575df61","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-02-23 16:30:06.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset021-master https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset021-master/859/ : FAILURE in 56m 11s","accounts_in_message":[],"_revision_number":9},{"id":"7606c49709daedba6cba5658ece3b99e4f00cf85","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-02-23 21:00:34.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/487884/9/EMC_XtremIO_FC/2048 : SUCCESS in 1h 36m 20s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/487884/9/EMC_XtremIO_iSCSI/2259 : FAILURE in 2h 48m 03s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":9},{"id":"ae403bb1eeaafed44f730a9f5074304637b6cbf7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-08 17:12:12.000000000","message":"Patch Set 9: Code-Review-1\n\nPlease rebase and update the patch if this is still needed. If no activity then we will abandon this review. Thanks.","accounts_in_message":[],"_revision_number":9},{"id":"2d393d4b6075ed64e5896821034cd97b9ac93c1a","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2021-07-08 19:47:53.000000000","message":"Patch Set 9:\n\nHi Ghanshyam,\n\nI think it\u0027s safe to abandon this change. It can still be picked up later.\n\nThanks,\nIldikó","accounts_in_message":[],"_revision_number":9},{"id":"26a12a702ff71c243069b21f36df3227fc9c0842","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-08 22:48:09.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 9:\n\u003e \n\u003e Hi Ghanshyam,\n\u003e \n\u003e I think it\u0027s safe to abandon this change. It can still be picked up later.\n\u003e \n\u003e Thanks,\n\u003e Ildikó\n\nThanks Ildiko for response. will do.","accounts_in_message":[],"_revision_number":9},{"id":"6d5d302d8b3a3c9ffd8455d67320769b46c4bc56","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-08 22:48:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"9349733212ddd9d608ef86d2c585c2d7fa8ccc0c","revisions":{"44ae57e533bccb26e3d18e704d5f286d02c07431":{"kind":"REWORK","_number":1,"created":"2017-07-27 14:56:27.000000000","uploader":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"ref":"refs/changes/84/487884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/84/487884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/84/487884/1"}}},"commit":{"parents":[{"commit":"0252bca0d926774eab86fd88590401897eefa41b","subject":"Merge \"Fix some nits in object storage clients release notes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0252bca0d926774eab86fd88590401897eefa41b"}]}],"author":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-27 14:53:41.000000000","tz":-240},"committer":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-27 14:53:41.000000000","tz":-240},"subject":"add attachments checks to iscsi live migration test","message":"add attachments checks to iscsi live migration test\n\nWith the new live migration flow if a delete attachment\nis omitted, you can acccumulate cinder attachment objects\nwithout being aware of it. After detach, these attachments\nare no longer associated with a volume or an instance\nso the only way to know if they are gone is to check the\ncinder db.\n\nThis adds a check after migration to make sure the original\nattachments have been deleted.\n\nA new attachments client is created to provide a way to\ncall cinder to get the attachment list.\n\nDepends-On: Ifc01dbf98545104c998ab96f65ff8623a6db0f28\nDepends-On: I0bfb11296430dfffe9b091ae7c3a793617bd9d0d\n\nPartially Implements: blueprint cinder-new-attach-apis\n\nChange-Id: I72ccf82debe36607b58fa75e52af535494999959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/44ae57e533bccb26e3d18e704d5f286d02c07431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/44ae57e533bccb26e3d18e704d5f286d02c07431"}]},"branch":"refs/heads/master"},"d14fe3a763d68a591a7976b3b8504cda9c8c317d":{"kind":"REWORK","_number":2,"created":"2017-07-27 17:00:53.000000000","uploader":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"ref":"refs/changes/84/487884/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/84/487884/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/84/487884/2"}}},"commit":{"parents":[{"commit":"0252bca0d926774eab86fd88590401897eefa41b","subject":"Merge \"Fix some nits in object storage clients release notes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0252bca0d926774eab86fd88590401897eefa41b"}]}],"author":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-27 14:53:41.000000000","tz":-240},"committer":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-27 16:56:46.000000000","tz":-240},"subject":"add attachments checks to iscsi live migration test","message":"add attachments checks to iscsi live migration test\n\nWith the new live migration flow if a delete attachment\nis omitted, you can acccumulate cinder attachment objects\nwithout being aware of it. After detach, these attachments\nare no longer associated with a volume or an instance\nso the only way to know if they are gone is to check the\ncinder db.\n\nThis adds a check after migration to make sure the original\nattachments have been deleted.\n\nA new attachments client is created to provide a way to\ncall cinder to get the attachment list.\n\nDepends-On: Ifc01dbf98545104c998ab96f65ff8623a6db0f28\nDepends-On: I0bfb11296430dfffe9b091ae7c3a793617bd9d0d\n\nPartially Implements: blueprint cinder-new-attach-apis\n\nChange-Id: I72ccf82debe36607b58fa75e52af535494999959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d14fe3a763d68a591a7976b3b8504cda9c8c317d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d14fe3a763d68a591a7976b3b8504cda9c8c317d"}]},"branch":"refs/heads/master"},"f901a27f8e88557e983e588a5a29a61d5094094f":{"kind":"REWORK","_number":3,"created":"2017-07-28 14:53:24.000000000","uploader":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"ref":"refs/changes/84/487884/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/84/487884/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/84/487884/3"}}},"commit":{"parents":[{"commit":"0252bca0d926774eab86fd88590401897eefa41b","subject":"Merge \"Fix some nits in object storage clients release notes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0252bca0d926774eab86fd88590401897eefa41b"}]}],"author":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-27 14:53:41.000000000","tz":-240},"committer":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-28 14:42:20.000000000","tz":-240},"subject":"add attachments checks to iscsi live migration test","message":"add attachments checks to iscsi live migration test\n\nWith the new live migration flow if a delete attachment\nis omitted, you can acccumulate cinder attachment objects\nwithout being aware of it. After detach, these attachments\nare no longer associated with a volume or an instance\nso the only way to know if they are gone is to check\nattachments in the cli.\n\nThis adds a check after migration to make sure the original\nattachments have been deleted.\n\nA new attachments client is created to provide a way to\ncall cinder to get the attachment list.\n\nDepends-On: Ifc01dbf98545104c998ab96f65ff8623a6db0f28\nDepends-On: I0bfb11296430dfffe9b091ae7c3a793617bd9d0d\n\nPartially Implements: blueprint cinder-new-attach-apis\n\nChange-Id: I72ccf82debe36607b58fa75e52af535494999959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f901a27f8e88557e983e588a5a29a61d5094094f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f901a27f8e88557e983e588a5a29a61d5094094f"}]},"branch":"refs/heads/master"},"2751ab4dbc03a0935f41fd682843a7ccb0440a59":{"kind":"REWORK","_number":4,"created":"2017-07-31 17:34:53.000000000","uploader":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"ref":"refs/changes/84/487884/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/84/487884/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/84/487884/4"}}},"commit":{"parents":[{"commit":"5043df2ba9e341c9944f152f16d969f94cb9f154","subject":"Merge \"TrivialFix: Correct reST field lists in docstrings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5043df2ba9e341c9944f152f16d969f94cb9f154"}]}],"author":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-27 14:53:41.000000000","tz":-240},"committer":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-31 15:20:11.000000000","tz":-240},"subject":"add attachments checks to iscsi live migration test","message":"add attachments checks to iscsi live migration test\n\nWith the new live migration flow if a delete attachment\nis omitted, you can acccumulate cinder attachment objects\nwithout being aware of it. After detach, these attachments\nare no longer associated with a volume or an instance\nso the only way to know if they are gone is to check\nattachments in the cli.\n\nThis adds a check after migration to make sure the original\nattachments have been deleted.\n\nA new attachments client is created to provide a way to\ncall cinder to get the attachment list.\n\nDepends-On: Ifc01dbf98545104c998ab96f65ff8623a6db0f28\nDepends-On: I0bfb11296430dfffe9b091ae7c3a793617bd9d0d\n\nPartially Implements: blueprint cinder-new-attach-apis\n\nChange-Id: I72ccf82debe36607b58fa75e52af535494999959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2751ab4dbc03a0935f41fd682843a7ccb0440a59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2751ab4dbc03a0935f41fd682843a7ccb0440a59"}]},"branch":"refs/heads/master"},"fe246dc772be77242a20c71e61c69c570f066720":{"kind":"REWORK","_number":5,"created":"2017-08-28 17:50:35.000000000","uploader":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"ref":"refs/changes/84/487884/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/84/487884/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/84/487884/5"}}},"commit":{"parents":[{"commit":"5e9499f3e8887da6a68836f3949bb7ce30ae8e6b","subject":"Merge \"Added unit tests for blacklist and whitelist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5e9499f3e8887da6a68836f3949bb7ce30ae8e6b"}]}],"author":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-27 14:53:41.000000000","tz":-240},"committer":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-08-28 17:44:21.000000000","tz":-240},"subject":"add attachments checks to iscsi live migration test","message":"add attachments checks to iscsi live migration test\n\nWith the new live migration flow if a delete attachment\nis omitted, you can acccumulate cinder attachment objects\nwithout being aware of it. After detach, these attachments\nare no longer associated with a volume or an instance\nso the only way to know if they are gone is to check\nattachments in the cli.\n\nThis adds a check after migration to make sure the original\nattachments have been deleted.\n\nA new attachments client is created to provide a way to\ncall cinder to get the attachment list.\n\nDepends-On: Ifc01dbf98545104c998ab96f65ff8623a6db0f28\nDepends-On: I0bfb11296430dfffe9b091ae7c3a793617bd9d0d\n\nPartially Implements: blueprint cinder-new-attach-apis\n\nChange-Id: I72ccf82debe36607b58fa75e52af535494999959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fe246dc772be77242a20c71e61c69c570f066720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fe246dc772be77242a20c71e61c69c570f066720"}]},"branch":"refs/heads/master"},"1f336f351b3f8795ebff9a962b49384c100607b8":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2017-08-30 13:16:47.000000000","uploader":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"ref":"refs/changes/84/487884/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/84/487884/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/84/487884/6"}}},"commit":{"parents":[{"commit":"4c8b6f9ab2caaeeb915ebd4bd447726360d63b3e","subject":"Merge \"Add detail specific fields check in list_backups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4c8b6f9ab2caaeeb915ebd4bd447726360d63b3e"}]}],"author":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-27 14:53:41.000000000","tz":-240},"committer":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-08-30 13:08:27.000000000","tz":-240},"subject":"add attachments checks to iscsi live migration test","message":"add attachments checks to iscsi live migration test\n\nWith the new live migration flow if a delete attachment\nis omitted, you can acccumulate cinder attachment objects\nwithout being aware of it. After detach, these attachments\nare no longer associated with a volume or an instance\nso the only way to know if they are gone is to check\nattachments in the cli.\n\nThis adds a check after migration to make sure the original\nattachments have been deleted.\n\nA new attachments client is created to provide a way to\ncall cinder to get the attachment list.\n\nDepends-On: Ifc01dbf98545104c998ab96f65ff8623a6db0f28\nDepends-On: I0bfb11296430dfffe9b091ae7c3a793617bd9d0d\n\nPartially Implements: blueprint cinder-new-attach-apis\n\nChange-Id: I72ccf82debe36607b58fa75e52af535494999959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1f336f351b3f8795ebff9a962b49384c100607b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1f336f351b3f8795ebff9a962b49384c100607b8"}]},"branch":"refs/heads/master"},"a21da7449172ebce1660c2c5474ed7b415981da6":{"kind":"REWORK","_number":7,"created":"2017-09-07 13:57:20.000000000","uploader":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"ref":"refs/changes/84/487884/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/84/487884/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/84/487884/7"}}},"commit":{"parents":[{"commit":"4c8b6f9ab2caaeeb915ebd4bd447726360d63b3e","subject":"Merge \"Add detail specific fields check in list_backups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4c8b6f9ab2caaeeb915ebd4bd447726360d63b3e"}]}],"author":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-27 14:53:41.000000000","tz":-240},"committer":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-09-07 13:54:16.000000000","tz":-240},"subject":"Add attachments checks to iscsi live migration test","message":"Add attachments checks to iscsi live migration test\n\nWith the new live migration flow if a delete attachment\nis omitted, you can acccumulate cinder attachment objects\nwithout being aware of it. After detach, these attachments\nare no longer associated with a volume or an instance\nso the only way to know if they are gone is to check\nattachments in the cli.\n\nThis adds a check after migration to make sure the original\nattachments have been deleted.\n\nA new attachments client is created to provide a way to\ncall cinder to get the attachment list.\n\nDepends-On: Ifc01dbf98545104c998ab96f65ff8623a6db0f28\nDepends-On: I0bfb11296430dfffe9b091ae7c3a793617bd9d0d\n\nPartially Implements: blueprint cinder-new-attach-apis\n\nChange-Id: I72ccf82debe36607b58fa75e52af535494999959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a21da7449172ebce1660c2c5474ed7b415981da6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a21da7449172ebce1660c2c5474ed7b415981da6"}]},"branch":"refs/heads/master"},"be5a960f471abc03b13bf11296b3c88eabcb9b97":{"kind":"REWORK","_number":8,"created":"2017-09-22 17:17:10.000000000","uploader":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"ref":"refs/changes/84/487884/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/84/487884/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/84/487884/8"}}},"commit":{"parents":[{"commit":"12c58abe82312f29d9c3b55dd1a6c9ac1a4d6b9f","subject":"Merge \"Move identity_utils to common.identity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/12c58abe82312f29d9c3b55dd1a6c9ac1a4d6b9f"}]}],"author":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-27 14:53:41.000000000","tz":-240},"committer":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-09-22 17:05:25.000000000","tz":-240},"subject":"Add attachments checks to iscsi live migration test","message":"Add attachments checks to iscsi live migration test\n\nWith the new live migration flow if a delete attachment\nis omitted, you can acccumulate cinder attachment objects\nwithout being aware of it. After detach, these attachments\nare no longer associated with a volume or an instance\nso the only way to know if they are gone is to check\nattachments in the cli.\n\nThis adds a check after migration to make sure the original\nattachments have been deleted.\n\nThis test can also be run in a grenade environment where the\nlive migration is from a pike to queens host, in which case\nthe old migration flow is used. The test handles both cases.\n\nA new attachments client is created to provide a way to\ncall cinder to get the attachment list.\n\nDepends-On: Ifc01dbf98545104c998ab96f65ff8623a6db0f28\nDepends-On: I0bfb11296430dfffe9b091ae7c3a793617bd9d0d\n\nPartially Implements: blueprint cinder-new-attach-apis\n\nChange-Id: I72ccf82debe36607b58fa75e52af535494999959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/be5a960f471abc03b13bf11296b3c88eabcb9b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/be5a960f471abc03b13bf11296b3c88eabcb9b97"}]},"branch":"refs/heads/master"},"9349733212ddd9d608ef86d2c585c2d7fa8ccc0c":{"kind":"REWORK","_number":9,"created":"2018-02-23 11:07:27.000000000","uploader":{"_account_id":17386,"name":"Steve Noyes","email":"steve.noyes@oracle.com","username":"ssn1ssn"},"ref":"refs/changes/84/487884/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/84/487884/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/487884/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/84/487884/9"}}},"commit":{"parents":[{"commit":"beb7716bb4204bdc43758cd0b38e475088b0f190","subject":"Merge \"Remove leftover call to _create_stestrepository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/beb7716bb4204bdc43758cd0b38e475088b0f190"}]}],"author":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2017-07-27 14:53:41.000000000","tz":-240},"committer":{"name":"Steve Noyes","email":"steve.noyes@oracle.com","date":"2018-02-23 10:40:20.000000000","tz":-300},"subject":"Add attachments checks to iscsi live migration test","message":"Add attachments checks to iscsi live migration test\n\nWith the new live migration flow if a delete attachment\nis omitted, you can acccumulate cinder attachment objects\nwithout being aware of it. After detach, these attachments\nare no longer associated with a volume or an instance\nso the only way to know if they are gone is to check\nattachments in the cli.\n\nThis adds a check after migration to make sure the original\nattachments have been deleted.\n\nThis test can also be run in a grenade environment where the\nlive migration is from a pike to queens host, in which case\nthe old migration flow is used. The test handles both cases.\n\nA new attachments client is created to provide a way to\ncall cinder to get the attachment list.\n\nPartially Implements: blueprint cinder-new-attach-apis\n\nChange-Id: I72ccf82debe36607b58fa75e52af535494999959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9349733212ddd9d608ef86d2c585c2d7fa8ccc0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9349733212ddd9d608ef86d2c585c2d7fa8ccc0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
