)]}'
{"id":"openstack%2Ftempest~491376","triplet_id":"openstack%2Ftempest~master~I0af99cfb2bedcca8875c864a232f8770c67deb24","project":"openstack/tempest","branch":"master","topic":"unused_import","hashtags":[],"change_id":"I0af99cfb2bedcca8875c864a232f8770c67deb24","subject":"TrivialFix: Remove the unused import code","status":"ABANDONED","created":"2017-08-07 07:20:32.000000000","updated":"2017-08-08 02:45:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"29fc1ff61b88f4af419bde0ee425acaff4b6626f","_number":491376,"virtual_id_number":491376,"owner":{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":-1,"date":"2017-08-07 08:08:02.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":12024,"name":"Tong Liu","email":"tongl@vmware.com","username":"tong101"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":12024,"name":"Tong Liu","email":"tongl@vmware.com","username":"tong101"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-1,"date":"2017-08-07 18:39:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":-1,"date":"2017-08-07 19:34:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12024,"name":"Tong Liu","email":"tongl@vmware.com","username":"tong101"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":12024,"name":"Tong Liu","email":"tongl@vmware.com","username":"tong101"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":12024,"name":"Tong Liu","email":"tongl@vmware.com","username":"tong101"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":12024,"name":"Tong Liu","email":"tongl@vmware.com","username":"tong101"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-07 08:08:02.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-08-07 18:39:46.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2017-08-07 19:34:09.000000000","updated_by":{"_account_id":12024,"name":"Tong Liu","email":"tongl@vmware.com","username":"tong101"},"reviewer":{"_account_id":12024,"name":"Tong Liu","email":"tongl@vmware.com","username":"tong101"},"state":"REVIEWER"}],"messages":[{"id":"5cb854888b626b2bab533b14016fd18d26b99dd3","author":{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"},"date":"2017-08-07 07:20:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61da11aff549ffb79482e8478242cfdb82585583","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-07 08:08:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/76/491376/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c16132d/ : FAILURE in 29m 33s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/76/491376/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/7a35989/ : FAILURE in 26m 19s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/76/491376/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/6c44381/ : FAILURE in 23m 05s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/76/491376/1/check/gate-tempest-docs-ubuntu-xenial/f23843d//doc/build/html/ : SUCCESS in 11m 55s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/76/491376/1/check/gate-tempest-pep8-ubuntu-xenial/7e2c9cc/ : SUCCESS in 4m 03s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/76/491376/1/check/gate-tempest-python27-ubuntu-xenial/f8ad2bb/ : FAILURE in 2m 23s\n- gate-tempest-python35 http://logs.openstack.org/76/491376/1/check/gate-tempest-python35/dba4d32/ : FAILURE in 2m 25s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/76/491376/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/9595fb6/ : FAILURE in 33m 42s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/76/491376/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/09123c6/ : FAILURE in 30m 57s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/76/491376/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/4736c96/ : FAILURE in 24m 22s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/76/491376/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/165bdd3/ : FAILURE in 45m 40s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/76/491376/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/6921b78/ : FAILURE in 38m 18s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/76/491376/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/fd31db6/ : FAILURE in 40m 42s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/76/491376/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/c3a5e31/ : FAILURE in 27m 36s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/76/491376/1/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/e1dab92/ : FAILURE in 24m 36s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/76/491376/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/c198c71/ : FAILURE in 26m 33s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/76/491376/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/8e472c3/ : FAILURE in 22m 43s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/76/491376/1/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/1d88c12/ : FAILURE in 32m 29s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/76/491376/1/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/d3394dc/ : FAILURE in 46m 00s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/76/491376/1/check/tempest-coverage-ubuntu-xenial-nv/6c7b835/ : FAILURE in 2m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84cdbf6a6a4272de5af1e8cb622d43c26f02dc29","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2017-08-07 18:39:46.000000000","message":"Patch Set 1: Code-Review-1\n\nWe cannot remove the code as the test failure\n\nhttp://logs.openstack.org/76/491376/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c16132d/logs/devstacklog.txt.gz#_2017-08-07_07_49_39_949","accounts_in_message":[],"_revision_number":1},{"id":"4d26ccaa94dc419191615cb09cd5e9b7dd2f6583","author":{"_account_id":12024,"name":"Tong Liu","email":"tongl@vmware.com","username":"tong101"},"date":"2017-08-07 19:34:09.000000000","message":"Patch Set 1: Code-Review-1\n\nIt breaks a lot of tests and I think we cannot remove them.","accounts_in_message":[],"_revision_number":1},{"id":"6f59a066ce648b8e96c95dc736e337c96a0527aa","author":{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"},"date":"2017-08-08 02:45:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"504cc03c1b78ee92e4a5c7b0c794f9b725e04213","revisions":{"504cc03c1b78ee92e4a5c7b0c794f9b725e04213":{"kind":"REWORK","_number":1,"created":"2017-08-07 07:20:32.000000000","uploader":{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"},"ref":"refs/changes/76/491376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/76/491376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/76/491376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/76/491376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/76/491376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/76/491376/1"}}},"commit":{"parents":[{"commit":"c465bfe00183c18ae83e28dadfc2560f4ff1394e","subject":"Merge \"Remove _list_assertions from test_roles.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c465bfe00183c18ae83e28dadfc2560f4ff1394e"}]}],"author":{"name":"yfzhao","email":"dsware@126.com","date":"2017-08-07 07:18:58.000000000","tz":480},"committer":{"name":"yfzhao","email":"dsware@126.com","date":"2017-08-07 07:18:58.000000000","tz":480},"subject":"TrivialFix: Remove the unused import code","message":"TrivialFix: Remove the unused import code\n\nRemove the unused import code\n\nChange-Id: I0af99cfb2bedcca8875c864a232f8770c67deb24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/504cc03c1b78ee92e4a5c7b0c794f9b725e04213"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/504cc03c1b78ee92e4a5c7b0c794f9b725e04213"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c465bfe00183c18ae83e28dadfc2560f4ff1394e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
