)]}'
{"id":"openstack%2Ftempest~49511","triplet_id":"openstack%2Ftempest~master~Id6c29f26f19f69abab558d5f94d8cbf6190ed43b","project":"openstack/tempest","branch":"master","topic":"nose-skip","hashtags":[],"change_id":"Id6c29f26f19f69abab558d5f94d8cbf6190ed43b","subject":"TEMPEST_PY26_NOSE_COMPAT in py26 jobs","status":"MERGED","created":"2013-10-03 10:53:19.000000000","updated":"2013-10-03 20:40:14.000000000","submitted":"2013-10-03 20:40:14.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1002a2e3edc5f8cff8819b749d80dbd042761a5a","_number":49511,"virtual_id_number":49511,"owner":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-10-03 20:40:14.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-10-03 20:08:15.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":2,"date":"2013-10-03 17:40:26.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-10-03 20:08:15.000000000","_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"date":"2013-10-03 17:36:34.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"date":"2013-10-03 17:40:26.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-10-03 12:00:11.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2013-10-03 17:40:26.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2013-10-03 20:08:15.000000000","updated_by":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"reviewer":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"state":"REVIEWER"},{"updated":"2013-10-03 20:40:14.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ed04328ab7f25836bfabc16d12aca0a9c0f68c21","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-03 11:32:20.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/11/49511/1/check/gate-tempest-pep8/c9c8103 : SUCCESS in 1m 48s\n- gate-tempest-docs http://logs.openstack.org/11/49511/1/check/gate-tempest-docs/2564658 : SUCCESS in 3m 33s\n- check-tempest-devstack-vm-full http://logs.openstack.org/11/49511/1/check/check-tempest-devstack-vm-full/b4603c2 : SUCCESS in 30m 48s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/11/49511/1/check/check-tempest-devstack-vm-postgres-full/0864023 : SUCCESS in 38m 38s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/11/49511/1/check/check-tempest-devstack-vm-neutron/fc4ad71 : SUCCESS in 19m 37s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/11/49511/1/check/gate-tempest-devstack-vm-large-ops/15c28f7 : SUCCESS in 12m 08s\n- check-grenade-devstack-vm http://logs.openstack.org/11/49511/1/check/check-grenade-devstack-vm/2dfd43b : SUCCESS in 25m 01s\n","accounts_in_message":[],"_revision_number":1},{"id":"1141e9a6781564db909ad82c03322a43a1a81eba","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-10-03 12:00:11.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\nthat being said, we really need to discuss dropping py26 in icehouse, as py26 tempest is rapidly becoming a very weird fork of real tempest.","accounts_in_message":[],"_revision_number":1},{"id":"d00338c52bc5795ae1637d85352b86d165265ab3","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2013-10-03 14:06:31.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\nYou should also add this to run_tests.sh.","accounts_in_message":[],"_revision_number":1},{"id":"f075ad081e935b2278e6c7a6d2f209441c69f11a","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-10-03 17:36:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2dfd5bf1d09f323cfb8975a2b79c3e2049ff3a5d","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2013-10-03 17:40:26.000000000","message":"Patch Set 2: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"f44fe75371ac395f113955096843be5b0e611472","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-03 18:21:26.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/11/49511/2/check/gate-tempest-pep8/755ceca : SUCCESS in 1m 31s\n- gate-tempest-docs http://logs.openstack.org/11/49511/2/check/gate-tempest-docs/a890152 : SUCCESS in 4m 01s\n- check-tempest-devstack-vm-full http://logs.openstack.org/11/49511/2/check/check-tempest-devstack-vm-full/82046b9 : SUCCESS in 39m 38s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/11/49511/2/check/check-tempest-devstack-vm-postgres-full/26d4b39 : SUCCESS in 41m 38s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/11/49511/2/check/check-tempest-devstack-vm-neutron/28e310f : SUCCESS in 25m 00s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/11/49511/2/check/gate-tempest-devstack-vm-large-ops/8a490c0 : SUCCESS in 11m 53s\n- check-grenade-devstack-vm http://logs.openstack.org/11/49511/2/check/check-grenade-devstack-vm/46b85c3 : SUCCESS in 44m 38s\n","accounts_in_message":[],"_revision_number":2},{"id":"cb3a122080a7dee22f167bd1be6ae73b79e5a09f","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2013-10-03 20:08:15.000000000","message":"Patch Set 2: Approved\n\n","accounts_in_message":[],"_revision_number":2},{"id":"a28c66ef2b630aaded3319fe8b9d014fb908f7ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-03 20:08:16.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"6f076ce637b1329a44af1df6b8ec903e13baa90f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-03 20:40:14.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2},{"id":"fe1fe91a9abcdd943ac8dab51da9f4300790cc75","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-03 20:40:14.000000000","message":"Patch Set 2: Verified\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/11/49511/2/gate/gate-tempest-pep8/c2428af : SUCCESS in 1m 28s\n- gate-tempest-docs http://logs.openstack.org/11/49511/2/gate/gate-tempest-docs/1a297a1 : SUCCESS in 4m 09s\n- gate-tempest-devstack-vm-full http://logs.openstack.org/11/49511/2/gate/gate-tempest-devstack-vm-full/c65da06 : SUCCESS in 31m 43s\n- gate-tempest-devstack-vm-postgres-full http://logs.openstack.org/11/49511/2/gate/gate-tempest-devstack-vm-postgres-full/a6ca75c : SUCCESS in 30m 47s\n- gate-tempest-devstack-vm-neutron http://logs.openstack.org/11/49511/2/gate/gate-tempest-devstack-vm-neutron/0caf0e7 : SUCCESS in 17m 49s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/11/49511/2/gate/gate-tempest-devstack-vm-large-ops/e4e1c73 : SUCCESS in 10m 25s\n- gate-grenade-devstack-vm http://logs.openstack.org/11/49511/2/gate/gate-grenade-devstack-vm/8bfa845 : SUCCESS in 24m 39s\n","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea88d9f516ded089e02709cac8fec9ed04ced3aa","revisions":{"0f3dbd871fa67601cc4203647275e2e5789678ae":{"kind":"REWORK","_number":1,"created":"2013-10-03 10:53:19.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/11/49511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/11/49511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/49511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/49511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/49511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/11/49511/1"}}},"commit":{"parents":[{"commit":"5573270d9b35fc8b2336eff9cb608e51472013e1","subject":"Merge \"Unskip test_tokens and update expected status to 404 from 401\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5573270d9b35fc8b2336eff9cb608e51472013e1"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2013-10-03 10:51:44.000000000","tz":120},"committer":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2013-10-03 10:51:44.000000000","tz":120},"subject":"TEMPEST_PY26_NOSE_COMPAT in py26 jobs","message":"TEMPEST_PY26_NOSE_COMPAT in py26 jobs\n\nUse py26 skip exception compatibility code path in py26 jobs.\n\nChange-Id: Id6c29f26f19f69abab558d5f94d8cbf6190ed43b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0f3dbd871fa67601cc4203647275e2e5789678ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0f3dbd871fa67601cc4203647275e2e5789678ae"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5573270d9b35fc8b2336eff9cb608e51472013e1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ea88d9f516ded089e02709cac8fec9ed04ced3aa":{"kind":"REWORK","_number":2,"created":"2013-10-03 17:36:34.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/11/49511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/11/49511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/49511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/49511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/49511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/11/49511/2"}}},"commit":{"parents":[{"commit":"4d91977ccd4273e302330c6aa35dae32b9bb877c","subject":"Merge \"add server delete/unpause tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4d91977ccd4273e302330c6aa35dae32b9bb877c"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2013-10-03 10:51:44.000000000","tz":120},"committer":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2013-10-03 17:36:15.000000000","tz":120},"subject":"TEMPEST_PY26_NOSE_COMPAT in py26 jobs","message":"TEMPEST_PY26_NOSE_COMPAT in py26 jobs\n\nUse py26 skip exception compatibility code path in py26 jobs.\n\nChange-Id: Id6c29f26f19f69abab558d5f94d8cbf6190ed43b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ea88d9f516ded089e02709cac8fec9ed04ced3aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ea88d9f516ded089e02709cac8fec9ed04ced3aa"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4d91977ccd4273e302330c6aa35dae32b9bb877c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
